)]}'
{"id":"openstack%2Frequirements~733536","triplet_id":"openstack%2Frequirements~master~I6bdc02cc690e37c2e9870f96f8a47da8db19fcad","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6bdc02cc690e37c2e9870f96f8a47da8db19fcad","subject":"update constraint for oslo.concurrency to new release 4.1.0","status":"MERGED","created":"2020-06-04 10:48:16.000000000","updated":"2020-06-04 17:06:38.000000000","submitted":"2020-06-04 17:06:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733536-1591290398574-8cb36f4b","meta_rev_id":"8d1abf35e8aafb6da3d51256d6eef7e74f326359","_number":733536,"virtual_id_number":733536,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 17:06:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-04 13:07:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-04 14:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-04 14:31:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 13:07:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-04 14:31:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-04 17:06:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53c42d7865cd188d471af12d4834084a1c81997d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-04 10:48:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ddb342d4e1f7c1c113b0c96671bd082ff7e63a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:25:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3941fce5fdac473aa2324337ea56bd20 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c17bd4500584a0ca01708f27e2a9449 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee5c88f8a2d7435dae5fbe5d7e05b246 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5988b34b34f44809b99ea75749517d0 : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66d450d867c24a31b0838682c62ce9f6 : SUCCESS in 6m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df305173f0a64d54812e9ac070b67102 : SUCCESS in 3m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7e3dd91e510045b39aa0afe8df9b0249 : SUCCESS in 12m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b31602eb75264506a1ea192ed4512bb8 : SUCCESS in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b4d2fa6509e140e7aeaf76c4074e20c5 : SUCCESS in 12m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ced26c7a8f544a61a3cbfddff4bbd2e7 : SUCCESS in 5m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f271abe0f109425aaa376bfa3ac80c73 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/10ba50b971834e03b09df3498df0976a : SUCCESS in 9m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c9e01a9b04ab4a07a9eb570f12559a50 : SUCCESS in 33m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/45dbcbc06836479f9621d9a5d4d354d9 : SUCCESS in 23m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/733a8409bbda46b3a255456cc4586cb1 : SUCCESS in 16m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4892f1154280405d8e2332556ac402a0 : SUCCESS in 16m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/baae16fc441741ca9a5bbb2eebf0bc0e : SUCCESS in 8m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/916110cd68f24ff7ae1ee36438ded957 : SUCCESS in 10m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b0f579463dff4b6a9db8d9922fc681ba : SUCCESS in 12m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fecd6961b9954c098b676f5ba4f58023 : SUCCESS in 1h 01m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c091da9aead49388cfb4dc6ac4d14ed : SUCCESS in 1h 13m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92fe3493da3a4f8d831b4f363faaeaba : SUCCESS in 1h 14m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/3ac37316976245ce99989da8db3bd1a7 : SUCCESS in 1h 10m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0afb63bd8b9576db062f6936c8f7139e3e796514","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 13:07:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"beeb175884dd78a4d088776f290d007fcdb197ba","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-04 14:31:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"efe839baacdfbe9de12213993eb5bbc806803a80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:31:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86132af8d1974e87f2148b9d9de89be4231bdc6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:06:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e58cf5f73dfcf3f9e8ca1b278fd76f7374d5073f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:06:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8f3c52d0501415bbcaa53a19dacadd1 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37a760ebbdbd496e9c3da191d23509e4 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5399d119cddb42998759485734b93470 : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07f28e1ae29d44b18d5043a1ed8c2458 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82b2a7f7fde8483a9be5405077659f1a : SUCCESS in 6m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44362724907b41af8c49ff210b89083f : SUCCESS in 5m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/237d68d96f264a0aad483b4ecd429ec7 : SUCCESS in 10m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/29aac11df2d64a1c94ea5c3275348c24 : SUCCESS in 11m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/77d58c0f81c442fb938354374110e0db : SUCCESS in 10m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5d9df7136f824c32a9138ba817e7abb0 : SUCCESS in 6m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5994cc2d8ff5480c8de9a59b4487fddc : SUCCESS in 13m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cdfd42436dc54baa8bfdfe517bd08799 : SUCCESS in 10m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8faccfe0dbf54f06aab72e1773dec664 : SUCCESS in 17m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d798409eaf304c498dfc354817c5fa98 : SUCCESS in 31m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b7c70409ebdb451ab86d13876a09fca9 : SUCCESS in 15m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1551ad6016cb4a0d9e96ef65be97f915 : SUCCESS in 15m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1458cc21793449c78b01d6bbdeb2a8f8 : SUCCESS in 7m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c37f73e61e844b5fb4b2f3bfd6637d74 : SUCCESS in 9m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3a1f39fe636743d7b5d6afa24746962d : SUCCESS in 12m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f684364acbed468fb422dd5fa02a6fb6 : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9b54de8edc149b6a70fb57ecf12c0ba : SUCCESS in 1h 01m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f27929fba9d404bbed5bfa1d571ee2e : SUCCESS in 1h 15m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/283abc3ca50248218f9b5795af75bf16 : SUCCESS in 1h 25m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/afce51894eea446fa189fecd3f59bbd8 : SUCCESS in 1h 06m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dbb44cba7b4ea0c77027d32a5a555123908ddd1d","revisions":{"dbb44cba7b4ea0c77027d32a5a555123908ddd1d":{"kind":"REWORK","_number":1,"created":"2020-06-04 10:48:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/733536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/733536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/733536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/733536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/733536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/733536/1"}}},"commit":{"parents":[{"commit":"aeddfbeac59589aa4fe39c5e74b306a02678536e","subject":"Merge \"update constraint for ironic-lib to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeddfbeac59589aa4fe39c5e74b306a02678536e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:48:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:48:11.000000000","tz":0},"subject":"update constraint for oslo.concurrency to new release 4.1.0","message":"update constraint for oslo.concurrency to new release 4.1.0\n\nChange-Id: I6bdc02cc690e37c2e9870f96f8a47da8db19fcad\nmeta:version: 4.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dbb44cba7b4ea0c77027d32a5a555123908ddd1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dbb44cba7b4ea0c77027d32a5a555123908ddd1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
