)]}'
{"id":"openstack%2Frequirements~733537","triplet_id":"openstack%2Frequirements~master~I054397bae3a6bf863c6bbc4e8b357c09c864a739","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I054397bae3a6bf863c6bbc4e8b357c09c864a739","subject":"update constraint for osprofiler to new release 3.2.0","status":"MERGED","created":"2020-06-04 10:48:32.000000000","updated":"2020-06-04 17:06:36.000000000","submitted":"2020-06-04 17:06:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733537-1591290396604-5027401d","meta_rev_id":"1a25533b7e0a7bb65e2d2149f2e0a9c9223229a5","_number":733537,"virtual_id_number":733537,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 17:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-04 12:21:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-04 14:31:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-04 14:31:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 12:21:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-04 14:31:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-04 17:06:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da70a140966a0d2c5e227daacddf956dc3ba67ee","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-04 10:48:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"835c17923885b64f44c9d37d339b4bb94f0da49a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:18:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d0f2fc762a242ab92002fa35ae06539 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebe0f46301334a5185e64f57d737ee33 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4add9d486c54554a03e74284d67bd21 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab360ec1df34433c9600af876bebc1fe : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3389f8e184f44d91bc5e76207455194e : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/21df6122fe8040ab9d5ff815674ad187 : SUCCESS in 6m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/67f8e8b8f8c84a1b89a498676cd45bb7 : SUCCESS in 10m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b694c5191c8d485a938808423517a2ae : SUCCESS in 11m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cfb07ad23c4c427587bbd74c7e745ee0 : SUCCESS in 9m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cc1e6fe60072443cbca86631a5ff0a96 : SUCCESS in 6m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c58771a1d047400c853934b75596e5ce : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c88302deb5b4c75a9cfce2002456bc3 : SUCCESS in 8m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7f35ae08fb3142a0b4107a00adc700d9 : SUCCESS in 15m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/633b32882d224606a028e82af595cb68 : SUCCESS in 25m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6e29a89d7f3f4b5d9e88d9831cbe10d9 : SUCCESS in 15m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/21bfeaffd736418ca207aab06b098dc1 : SUCCESS in 23m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5d9f91ea89ad4adb8cf6bc96d469accd : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/73ae39f35c224875ba0ade219850b925 : SUCCESS in 6m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ea7541f25f704e3482c6fec397d51c96 : SUCCESS in 13m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c3ddf15e6ae45c481d7d9d78536751f : SUCCESS in 1h 03m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6953933b96064a0e867a58af873b4e33 : SUCCESS in 1h 05m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/317fe0b840ac469ea8fef9d452f67060 : SUCCESS in 54m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/fdda31de3747418dbfadfd95b58a2a21 : SUCCESS in 42m 59s","accounts_in_message":[],"_revision_number":1},{"id":"041c083c275493b9e6fb838248a80f35f93f4c3c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 12:21:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e57db8b71aff9975a2720d71bc5be2c85fbf6786","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-04 14:31:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e304411b758b4e465a9eb12c306adacc65a8293d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:31:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"94780acfbbed2d54382b26edf24d4c2a9e3e7d8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:06:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b935a6d90ef5fa2219a39c4513362da31361e59f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:06:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12a004a207154b84bd2174779eb93a3c : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/380ed8cf997644889c57a96710ab566f : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a639c8ed6ec8484aaa8989cbae62d39d : SUCCESS in 9m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d73610d04b5f477eb81aeef287a10dbd : SUCCESS in 9m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a69e229ac0f4d97a86889cab78577ff : SUCCESS in 6m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4a15509fde44d1ca91bfe8df8c3c354 : SUCCESS in 5m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5c48deb05fd347fb9ba9876a8138715f : SUCCESS in 9m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d00e8b4c910349af8c0ebfe58c92569c : SUCCESS in 11m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8aadb280f4b94d2ebf86bbd8fdc8d0f7 : SUCCESS in 12m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/054af33c8aea4223923b6303872baaaf : SUCCESS in 7m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/381653f0a3cd4606865359ccb2951fef : SUCCESS in 14m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/711284ea019b43b1b65d5a75e7913881 : SUCCESS in 10m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b624add2ff2c4d15bcdec55070d52c92 : SUCCESS in 15m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/729ddfef668a4aeab349df419cf9d889 : SUCCESS in 25m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a284761c4b1840b7a25e774ae6d8f66c : SUCCESS in 15m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bd33e05229cf46dfad15362e03f7bcce : SUCCESS in 19m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dea4ef424d5f4c0b804bae7d7068a5ca : SUCCESS in 7m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc31f466277d46a095bd677c8f97f35b : SUCCESS in 9m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/aad7062d3dd54b04ac9655ddffc52f46 : SUCCESS in 11m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9c2fcb77017647a095c5ed17c7a0657c : SUCCESS in 15m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9f3821321744d5c89e5e0476291cd5a : SUCCESS in 1h 05m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7152ffaa01524460b327f0a882e9b07c : SUCCESS in 1h 09m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1ac210297b04e11b3feb69aa127eb08 : SUCCESS in 1h 45m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/dbe9ce48e5924749980d4e423c3c3c2d : SUCCESS in 1h 20m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b616607716acf925de6de514af7d139fa44b8589","revisions":{"b616607716acf925de6de514af7d139fa44b8589":{"kind":"REWORK","_number":1,"created":"2020-06-04 10:48:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/733537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/733537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/733537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/733537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/733537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/733537/1"}}},"commit":{"parents":[{"commit":"aeddfbeac59589aa4fe39c5e74b306a02678536e","subject":"Merge \"update constraint for ironic-lib to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeddfbeac59589aa4fe39c5e74b306a02678536e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:48:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:48:31.000000000","tz":0},"subject":"update constraint for osprofiler to new release 3.2.0","message":"update constraint for osprofiler to new release 3.2.0\n\nChange-Id: I054397bae3a6bf863c6bbc4e8b357c09c864a739\nmeta:version: 3.2.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b616607716acf925de6de514af7d139fa44b8589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b616607716acf925de6de514af7d139fa44b8589"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
