)]}'
{"id":"openstack%2Frequirements~733541","triplet_id":"openstack%2Frequirements~master~I83cfb9578cc0ca023a6a8ee3158ccfb3f2566db6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I83cfb9578cc0ca023a6a8ee3158ccfb3f2566db6","subject":"update constraint for debtcollector to new release 2.1.0","status":"MERGED","created":"2020-06-04 10:49:08.000000000","updated":"2020-06-04 17:06:40.000000000","submitted":"2020-06-04 17:06:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733541-1591290400737-ce7c6379","meta_rev_id":"ed85c787187d3fd736d7514089c79c5324c9cccb","_number":733541,"virtual_id_number":733541,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 17:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-04 13:07:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-04 14:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-04 14:31:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 13:07:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-04 14:31:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-04 17:06:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"735ecae5f9ba5efd0677a6a3e65b79580b10e4af","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-04 10:49:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74b32c1b611e7517f0bce884474fde7f23734a2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:35:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9589219a3ea45a6b4da7820a0e30b7d : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee222ed29754befb560833ffedb54bb : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f9bda14ab53403a93634a1462878dcd : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27473d5cd3bb4c4cb8f37053906daebd : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/68fb08a1b576477884a2c64b256f0043 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c3fd115536f47e08aa4e4572cc345d9 : SUCCESS in 4m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/106c1dab721e4c81b66fc4a3a3b3f67a : SUCCESS in 9m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/44ab11dda2694bbb86148e6130885f45 : SUCCESS in 11m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4f1a27754b574b2d937eac5d5408e25f : SUCCESS in 10m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/18c1471f34834cafaadcee5e17528763 : SUCCESS in 6m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0b06b91b3cef41d4b20b58a89bc8725e : SUCCESS in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/948fdb7dcd1c4cf2bbdab08f658657b8 : SUCCESS in 16m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f2dfab5d5aeb43d282eda3ea1d257380 : SUCCESS in 16m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e5094ae07f6942c594850bb8e9a3feb5 : SUCCESS in 25m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9fbc66cc5b2e4a44ae37249b717723fc : SUCCESS in 15m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82004736eca3482e81016d55bd1ed000 : SUCCESS in 19m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/80cc1e41825a4578bf8393b745a25536 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eb40a5c924d54bd9a5441a502d91dc64 : SUCCESS in 7m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/94d7852a14c3483fa114389aa26e8080 : SUCCESS in 16m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b7608c466e14c21a81de4aa2bf458f7 : SUCCESS in 54m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a5e159b946f46bf8c458c952ae64917 : SUCCESS in 49m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4feebfbcae9549259e747ed1e244e102 : SUCCESS in 1h 14m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/e72b8afe37034c79a5e6c2bb57da2472 : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8524c9c415094e7d2ff9d628a3a1d277bb570293","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 13:07:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f294e378c9d5a5cd9d262ef79c8a4a117cfbe7c4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-04 14:31:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c8c2810d9520cf2d06ba505530d357dee7b73fda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:31:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d47a4edeb3619d60b493b703127f8a10fa431d79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:06:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d98e369644b8262f9232e020718c6fe2209aaa9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:06:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb188e6daa364dbf9ca88daaab0c3e4d : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9d40c15092b4fe2a93f95c479d63753 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1e308b3ab994c9e88b22a85912a718d : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cae3a6aa9ef54bac8a203bfc9be25c0d : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/321dc58313a943f38c73cd42af0fc43b : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb548cc794714b2889c937eb8b4638d2 : SUCCESS in 7m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/96de587145ca4968b3f957087c61fb7d : SUCCESS in 10m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a978b1c027664907bed649bdd86091a5 : SUCCESS in 14m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/231658f2701c494daa3da39d93e2b4ea : SUCCESS in 11m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7046f419850740a88f90a2f4bfd1655b : SUCCESS in 8m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/29fd2345a28d49be959072e93aa084c3 : SUCCESS in 13m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7fd9c3d2ce74425f9375d7cdbce153dd : SUCCESS in 11m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f594ca40b8e949fb9d2e60209d8ef3cc : SUCCESS in 25m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d31558410bd742d2af3183e33d318e7f : SUCCESS in 27m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0c534d9997474275b3ae9122c7f0dd06 : SUCCESS in 17m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/81ee42ad99494f358aee27d67853d8f7 : SUCCESS in 20m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e35bbf2b3d3344ca82d92ec1623ffac0 : SUCCESS in 7m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8f73f64ff1a44682a7201c9bc9336f80 : SUCCESS in 11m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/eb5d11404f2b40c8bc0b781dbddc76f3 : SUCCESS in 10m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aa886017c7f24f4583c570c7f3ad7dfa : SUCCESS in 16m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d58eca9164f4801abe1b598b2d57441 : SUCCESS in 1h 01m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/11829667249145b8ba85cad998124cca : SUCCESS in 1h 15m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18409e8f610e4a598fb4c271b01c8a34 : SUCCESS in 1h 28m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/2953f0e39f5442d4a89353937d778d2b : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4afeca577de421a7221eb7a90ee3e1183c8d6ef2","revisions":{"4afeca577de421a7221eb7a90ee3e1183c8d6ef2":{"kind":"REWORK","_number":1,"created":"2020-06-04 10:49:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/733541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/733541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/733541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/733541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/733541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/733541/1"}}},"commit":{"parents":[{"commit":"aeddfbeac59589aa4fe39c5e74b306a02678536e","subject":"Merge \"update constraint for ironic-lib to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeddfbeac59589aa4fe39c5e74b306a02678536e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:49:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:49:06.000000000","tz":0},"subject":"update constraint for debtcollector to new release 2.1.0","message":"update constraint for debtcollector to new release 2.1.0\n\nChange-Id: I83cfb9578cc0ca023a6a8ee3158ccfb3f2566db6\nmeta:version: 2.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4afeca577de421a7221eb7a90ee3e1183c8d6ef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4afeca577de421a7221eb7a90ee3e1183c8d6ef2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
