)]}'
{"id":"openstack%2Frequirements~733543","triplet_id":"openstack%2Frequirements~master~I4a75888ccee8382d5a13161709dc85ab83abf0ad","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4a75888ccee8382d5a13161709dc85ab83abf0ad","subject":"update constraint for tooz to new release 2.5.0","status":"MERGED","created":"2020-06-04 10:51:56.000000000","updated":"2020-06-04 17:10:50.000000000","submitted":"2020-06-04 17:06:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733543-1591290406422-d5a688fa","meta_rev_id":"86b2f8329d6d6ea202b3907cd54f6c4ac6f8fb58","_number":733543,"virtual_id_number":733543,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 17:06:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-04 13:07:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-04 14:31:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-04 14:31:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 13:07:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-04 14:31:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-04 17:06:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c07062b59612b53cd2ca3fe80c30fc67011568e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-04 10:51:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"454db79d400f03aed006f10e97bc316b39c64897","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:49:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b986d5d72a184854ac49d82074b5d0ad : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22545be0bf374844aa2eac3ee4938479 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2b95d783bc04d7f983ae453c6912d1d : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41ef2d2d7bb248d99246599fb6434d59 : SUCCESS in 6m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc6dcc119d0b4e7cb94aa771ccf4400e : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/47f77c0af9144327a377ce00df585268 : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/35548928e4b2465aa0a3de5b31d40648 : SUCCESS in 9m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/09ef52cc838b4dab8f5fa5d5a0123ff1 : SUCCESS in 10m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a6090d54abb24d8da91892058f152a2c : SUCCESS in 8m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/df33950092aa482fa201675feac8f141 : SUCCESS in 9m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/02637056619740e9abb1b227ae01f8f3 : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/85abfb66e0054eeb8dfc96bd09d5f484 : SUCCESS in 9m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/588b90be37cd4b0a8632077245328cd0 : SUCCESS in 17m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/27b1ed2582f746c388e5e55964b72bc9 : SUCCESS in 28m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4bcefb4373fe4e8f856bc4e3d374412f : SUCCESS in 16m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9a5177739ced48c681111ab3b5c5d6f6 : SUCCESS in 18m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e5448876e7cf4f98b38306a7d4e326f6 : SUCCESS in 6m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af153f45d0f84ce59794ec5685edd4b3 : SUCCESS in 9m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9d3b47ea90b84ad39dcd6afaaa97c99d : SUCCESS in 17m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fb66a546fd54589afae756723eca652 : SUCCESS in 57m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/032d9a873214411bbf80e670372d5467 : SUCCESS in 1h 16m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a243bb1cda14f83960f35a1f1246a61 : SUCCESS in 1h 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/905197e9a77442db86c9c6e9ab7986cc : SUCCESS in 1h 05m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a00717be52687bdf568d43cd9b16e4263aafca31","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 13:07:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a3f964dd8b506aa2b64b070015719101e6e1bfff","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-04 14:31:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a844287b10401057547f26e9b502f48a8153d75b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:32:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9c3807a3aed1accc686d5104ced3385e09f4f68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:06:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e460c4e95dd4e6d961d2fd5f143aee2 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a2d5fb1cbf49e9abb5f3652c990364 : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/397d104c00e44037a11a65315427ec9b : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e987698a83a34970a3d2a66ae38bdb96 : SUCCESS in 8m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09af7c7cae5a46cc9fe0168f5f39b78c : SUCCESS in 6m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/034a28103fb2432fb12d5987b939b529 : SUCCESS in 6m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9f47db0d5c3a4e1e9c0d7003f194c730 : SUCCESS in 11m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/17534fa575844afe86b395cc85ec82f4 : SUCCESS in 10m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/abd5a3859cf24c30a917a32702258c6b : SUCCESS in 11m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/84d68ada13b249f9998745e28196fdc2 : SUCCESS in 9m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1feb6268442c49b2932715d053b1f97b : SUCCESS in 14m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b36d06000be4d57b0ee5d4e1de8def8 : SUCCESS in 9m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/19a7f8d82cdb4432936b7e2ed1b30a31 : SUCCESS in 17m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fb214151d52b4ef0aa6d18901b0fb4e0 : SUCCESS in 29m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0720358f681a4f3cbeecd0e6f84caf11 : SUCCESS in 17m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/70527c1ca9084bd2b78e763d661a8abb : SUCCESS in 18m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e74f9009aa47439a9543b5c37a9baebf : SUCCESS in 8m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3da764573c2e4a42a041c40040aa285a : SUCCESS in 9m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bcd8e5463dae48a8879c130a4fe4f270 : SUCCESS in 16m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aac0c8a115ec465697160472cf6ad4d6 : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ecb4b673f774e999c4f8d7038e1cf9a : SUCCESS in 58m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/44a0671cb96e445bba16840c730c81f9 : SUCCESS in 1h 15m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/152490a88e1b4d70b98acad983612503 : SUCCESS in 1h 29m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/02e4a09a39234281bcdd9386aef138ef : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c2122a7f0d70368a7a1e70c802f77004506ffa68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:06:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a33f081eea6e73740b5fbfa8347ad0974f8a3ce0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:10:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3367a07ea2ca48e9887f6bd2a2420bbe : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7e95d9407973d88f2769f63fce095fc7e1cbf46","revisions":{"d7e95d9407973d88f2769f63fce095fc7e1cbf46":{"kind":"REWORK","_number":1,"created":"2020-06-04 10:51:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/733543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/733543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/733543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/733543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/733543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/733543/1"}}},"commit":{"parents":[{"commit":"aeddfbeac59589aa4fe39c5e74b306a02678536e","subject":"Merge \"update constraint for ironic-lib to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeddfbeac59589aa4fe39c5e74b306a02678536e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:51:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:51:51.000000000","tz":0},"subject":"update constraint for tooz to new release 2.5.0","message":"update constraint for tooz to new release 2.5.0\n\nChange-Id: I4a75888ccee8382d5a13161709dc85ab83abf0ad\nmeta:version: 2.5.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7e95d9407973d88f2769f63fce095fc7e1cbf46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7e95d9407973d88f2769f63fce095fc7e1cbf46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
