)]}'
{"id":"openstack%2Frequirements~733545","triplet_id":"openstack%2Frequirements~master~I145a7222d88097651df8bf7fe275b0dce470cd5f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I145a7222d88097651df8bf7fe275b0dce470cd5f","subject":"update constraint for oslo.cache to new release 2.4.1","status":"MERGED","created":"2020-06-04 10:52:19.000000000","updated":"2020-06-04 17:17:33.000000000","submitted":"2020-06-04 17:17:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733545-1591291053409-0bd3b327","meta_rev_id":"537b456a7486fe362ec02fb3223288be009c6226","_number":733545,"virtual_id_number":733545,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 17:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-04 13:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-04 14:31:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-04 14:31:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 13:06:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-04 14:31:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-04 17:17:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"099604eff6001b4dcbd5e57e426bfd591a13869a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-04 10:52:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5844636b1c2af73b7068ccc2e61f004ee596e14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:00:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd0aa259b8844914986956c39f210064 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aef6d3354b540e8aa429cc104214399 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc0c728366364e068504a4314103eb8a : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a485ce55be34e00bc5083990c748cae : SUCCESS in 7m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/57801ea0da414dd7a91b806c2e65be2a : SUCCESS in 5m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b9b610c077d474ca59c7ca8969154f7 : SUCCESS in 6m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c9d281085e974032a5f9c0550df6a20a : SUCCESS in 9m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fe544b304afd4f2ba55e425428624c02 : SUCCESS in 13m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c7e60bca02d240309392f0b4a049ab47 : SUCCESS in 12m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8a0f1958a14e4db9b93173ac2ec16094 : SUCCESS in 6m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b7f116aad8ba417aad7a10e47bc57fa3 : SUCCESS in 14m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9a2e0ff493ca49059ea1db0099cfcbc4 : SUCCESS in 6m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4987fb30d1384ae0862eca41ecaca1f3 : SUCCESS in 15m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/66f78d5235454f2f8881e48c0c05be88 : SUCCESS in 31m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/db7b9cc4267c4a88b0a3b1f07f92cb3c : SUCCESS in 15m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/68233eca308147dd8339aae048235e00 : SUCCESS in 21m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/04f30f45cbca406091cc3945de71bbe2 : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e1119864a7a745b994bc456b60a2760e : SUCCESS in 7m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dcded4a5c26c4912980427464b55265f : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5be3d7ff29c46e59898d1689b3eca6d : SUCCESS in 59m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a683074f851940cba338495b04e43865 : SUCCESS in 1h 19m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6472c447d58946399c83907cb4b965ea : SUCCESS in 1h 18m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/7ad1c9e21fcb46399b9a1a95fd3f5c44 : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0f4c05353d82f8dd844ccc647155349166c4db6e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 13:06:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"87fb4ec5723f02f236c1465aba0da7b7a688475c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-04 14:31:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dd4c9cfe43ef8a961bcd5c8fbbc135614e4d9613","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:32:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8e87e8dd7d27a80a13b1fff3aa06c104f27e0ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:17:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"82267e75e10d7939cd6c49209e246238da854be6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:17:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34eca4859e8d4ef7a968180ff8358f90 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4eb81088fb048da93427fab8cf38523 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0720ce4d10f14b7b9d8029a3464f107d : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cdb533f34644f42a4e2432b979d5e55 : SUCCESS in 7m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c29e8ef71a8f42669598c787c861bb02 : SUCCESS in 6m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/224ecf745f794144a80bf5e2b0f81ead : SUCCESS in 7m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/809fe8b427424c82a8c9c4ec893aea60 : SUCCESS in 11m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3717dd26c60f4beea4e90e79b0592a2b : SUCCESS in 14m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/df0eda48ee2a4bc991a893424bc0ea72 : SUCCESS in 13m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/de587d397f0542ebbb16176dff5a0880 : SUCCESS in 8m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e5879dc9250b43189d36243de9d0edf1 : SUCCESS in 12m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0831b07539b34344a6fce650f53b28c4 : SUCCESS in 11m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3f3987cd5944424eb3409dc5ae85571a : SUCCESS in 16m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f926663cdce240b4a9a081b8cc746f78 : SUCCESS in 30m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/32bc2410089c42f1871b58f206b85159 : SUCCESS in 15m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9d75e57e0324a79a11fba8eee1495a8 : SUCCESS in 21m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0047d1a5f49d47399704713bf85d20be : SUCCESS in 6m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83ce98da90444022ad841d48c41f68e0 : SUCCESS in 8m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0d704dbe51a04f5dbcbca3fb6719906e : SUCCESS in 14m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f298575468fa41fc83131d22f83628c8 : SUCCESS in 14m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e499800c330841fc99f8632d5c1f696b : SUCCESS in 59m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/72f09f2b1f8743af969dea7baf812fa7 : SUCCESS in 1h 16m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/278a459bccab47e3aa61cfccdddf302f : SUCCESS in 1h 26m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/0669481df6ea4d62a75ca7983a4c2027 : SUCCESS in 1h 07m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22a121d9381877cf68c134fb9bdf9ceaa2b6d831","revisions":{"22a121d9381877cf68c134fb9bdf9ceaa2b6d831":{"kind":"REWORK","_number":1,"created":"2020-06-04 10:52:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/733545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/733545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/733545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/733545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/733545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/733545/1"}}},"commit":{"parents":[{"commit":"aeddfbeac59589aa4fe39c5e74b306a02678536e","subject":"Merge \"update constraint for ironic-lib to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeddfbeac59589aa4fe39c5e74b306a02678536e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:52:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:52:18.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 2.4.1","message":"update constraint for oslo.cache to new release 2.4.1\n\nChange-Id: I145a7222d88097651df8bf7fe275b0dce470cd5f\nmeta:version: 2.4.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22a121d9381877cf68c134fb9bdf9ceaa2b6d831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22a121d9381877cf68c134fb9bdf9ceaa2b6d831"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
