)]}'
{"id":"openstack%2Frequirements~733552","triplet_id":"openstack%2Frequirements~master~I33c30feec29e46563accd21b03ac71c4ef2d3232","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I33c30feec29e46563accd21b03ac71c4ef2d3232","subject":"update constraint for oslo.reports to new release 2.1.0","status":"MERGED","created":"2020-06-04 10:56:23.000000000","updated":"2020-06-04 17:22:13.000000000","submitted":"2020-06-04 17:17:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733552-1591291058438-91d87e12","meta_rev_id":"e8469bfc0d1b09f99a6162c907f33a10d33c81bc","_number":733552,"virtual_id_number":733552,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 17:17:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-04 13:30:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-04 14:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-04 14:31:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 13:30:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-04 14:31:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-04 17:17:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ba364d7db697c6b95b944d7bf329b719c1b9495","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-04 10:56:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"582a9caae40e9d7ebd4e0ab65dff64a24c7ebb49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:14:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2e824d900fe4a658fe12ca3188342df : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7311d7e831e9400aa9c72a7d52c76181 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c4273c2f1204bf2984c1d70d40798fd : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8eae51c71b4840b49ae98a85d411e9dd : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7831b2688f944fa8453103edd8a2c14 : SUCCESS in 6m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/94ff32bd0437495bb2f1e3cada25f9d1 : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/44bfb8cfcd594d169ab8fc8546a0953b : SUCCESS in 10m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3c9c0cee81264b90a26fad16ff77a088 : SUCCESS in 10m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b514f6e5ec1c4cfc9e0d3e9ab598c2d2 : SUCCESS in 9m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4ab437ffc15a448784bd4731ce1341bf : SUCCESS in 5m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d26e6ba03c1446a2aca2e0c36ee33134 : SUCCESS in 12m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/881157ae078f4ecb82808f8d76a6f5c8 : SUCCESS in 8m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2b02a55950664658933566fadfe75aa0 : SUCCESS in 18m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/edc56c6517774e909d0d2cd4cabba5b3 : SUCCESS in 24m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c0697c664db94334912bdc076db6f646 : SUCCESS in 31m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e0c3f5e9834e4fd5878cc5683209f8d7 : SUCCESS in 19m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/542678056ace41d7bf8d6052483665ae : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/01061ca15cec47d6b50f14a7ce9c4824 : SUCCESS in 8m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2ad688a63d654189a826ffab423a4358 : SUCCESS in 14m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d6c1aed29ba4737b3d52abe3a90ef12 : SUCCESS in 1h 11m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cbb3813987ad41709d160baad20b78e4 : SUCCESS in 1h 25m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49dac15f75bf47858b808921106a208b : SUCCESS in 1h 32m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/5c0c16d81e0b465db955740090294256 : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e7bee3b216a5bf8d0264f0d4a9e5f02cbd1d2df8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 13:30:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0280c8e84e30da4486df284bb7b4114081e37048","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-04 14:31:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d2a085988d4aab75909fb55ea5addaa53dfa9be7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:33:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7b9a81447e51121a845105c2415a2e197a93cfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:17:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3d4acabc524f37aaa1737aa448391f548efd8b9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:17:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e530f920a7e4643a5904f3c636b4563 : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a30e9537fb3b4c94b00d14990c561498 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37f0737ec04d42afb8346cabe2307a27 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77773892e9064ce9909b3b682a62ae5c : SUCCESS in 9m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b13b3fc4a1b441a4b156c9ff43ee7497 : SUCCESS in 6m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/823d02e30fc146eab459250e7f3bf48b : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f57989a8a73b4b839479767526c5ebc5 : SUCCESS in 9m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a2b4d146e8d144d89f08b3f1589acf09 : SUCCESS in 10m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/00dabbe24c4d464b805a7c31a9176e42 : SUCCESS in 10m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4ce77a81b21544a2af10e7d238c1a87f : SUCCESS in 7m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/48e0e9ac93eb4c22b54538acf21f0b83 : SUCCESS in 15m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd45cf1f7c1248a3a4ce3981f6ba2b4c : SUCCESS in 11m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b91e3fc9aa844d53ae4321591ade8570 : SUCCESS in 24m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/20ab77f6c6bc4f408b652c7fa802d9a2 : SUCCESS in 27m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/690ed815e1f14ed9a0e65f8c752cf788 : SUCCESS in 15m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ffa5b8eca69a4701a6b77bb653bcc813 : SUCCESS in 15m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3557e666e89b4776bb5e52bce4be0779 : SUCCESS in 7m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4bacac7d7ce74e88a72030b5a92f7199 : SUCCESS in 12m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9f43045cf28d4352aeceab8914f5aba6 : SUCCESS in 14m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/364445e4d0154b35bba5319d6c115af5 : SUCCESS in 14m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa5f970a1f60421591f09122d805fa98 : SUCCESS in 56m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b154a86a00a419081ccec8b8038ee64 : SUCCESS in 47m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/606d578c1ea14505815b201c8522b19f : SUCCESS in 59m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/7063090d0a7548abafca90caba3be0e1 : SUCCESS in 43m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2ed9dd044518054094d0198fa9d176e48a160511","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:22:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaf5264641f94038b5225a0c836ceda4 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7c4ac7f81f5890b2f98219fd2beda418c4e5ae6","revisions":{"e7c4ac7f81f5890b2f98219fd2beda418c4e5ae6":{"kind":"REWORK","_number":1,"created":"2020-06-04 10:56:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/733552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/733552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/733552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/733552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/733552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/733552/1"}}},"commit":{"parents":[{"commit":"aeddfbeac59589aa4fe39c5e74b306a02678536e","subject":"Merge \"update constraint for ironic-lib to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeddfbeac59589aa4fe39c5e74b306a02678536e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:56:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:56:21.000000000","tz":0},"subject":"update constraint for oslo.reports to new release 2.1.0","message":"update constraint for oslo.reports to new release 2.1.0\n\nChange-Id: I33c30feec29e46563accd21b03ac71c4ef2d3232\nmeta:version: 2.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7c4ac7f81f5890b2f98219fd2beda418c4e5ae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7c4ac7f81f5890b2f98219fd2beda418c4e5ae6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
