)]}'
{"id":"openstack%2Frequirements~733553","triplet_id":"openstack%2Frequirements~master~I4d9744e092c809b7ca94738fa31f285210bca5ff","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4d9744e092c809b7ca94738fa31f285210bca5ff","subject":"update constraint for oslo.serialization to new release 3.2.0","status":"MERGED","created":"2020-06-04 10:56:58.000000000","updated":"2020-06-04 17:17:36.000000000","submitted":"2020-06-04 17:17:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733553-1591291056684-8cb36f4b","meta_rev_id":"fa4c23160b0fb69a28aeb710a0a48d8e863752ec","_number":733553,"virtual_id_number":733553,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 17:17:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-04 13:30:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-04 14:31:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-04 14:31:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 13:30:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-04 14:31:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-04 17:17:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fec3ae9e811029120c0269f249d5ba1f084aaaf1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-04 10:56:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60bf95f9e2a36abc8be5a2cfa09a400a1801e8fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:28:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8008769c46b41d8945bf3b8bad29974 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eefff6fce49f421286280afee1e4fdd3 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e27fa44bcc634bc58468bb4f10e03111 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c3d8bb1c26a47fd8ffffaae8b659543 : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca98b2153f404cb39ab70fb5b7422754 : SUCCESS in 7m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9757d906918e41df87fb10ad2495dcc5 : SUCCESS in 4m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4e065bfeb9e048318a2b967eed11fb28 : SUCCESS in 9m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e400458eb02e4a0688204da7134ee07a : SUCCESS in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3045132b93564e33bd2bbaabb80dfce6 : SUCCESS in 8m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2ead8ccf72db4a16a22183b90e4d58fb : SUCCESS in 5m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7e72fbece9d146459477712dfe8d12e0 : SUCCESS in 12m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5fc6b05821d549379afe6cfafb6259ae : SUCCESS in 13m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8e1a26510f094d5595d412c629325cc3 : SUCCESS in 15m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9605587d25664ffcb5b69ae2b194656c : SUCCESS in 27m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/12c05cab5d31440cbd9d846ecd95226f : SUCCESS in 15m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e64e7f2adb84834a724689e4e9b57ca : SUCCESS in 19m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8071d94af3764adca4baf1c538acd5de : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/108fd846cc19449cab990bc89f0139d1 : SUCCESS in 6m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5d44aa908b524c1db758c8ee0edd1a8b : SUCCESS in 13m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0575b0636bac468daf4c30d4c7d2eebf : SUCCESS in 54m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f414651682ad405584780edea6867df1 : SUCCESS in 1h 04m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e74736f155324ad9bf9087d34e653902 : SUCCESS in 1h 37m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/df1244af36e44289a16aac4568d3e141 : SUCCESS in 1h 13m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e8dd1ea8f60a6cbe7a7ec82532703d7099f6857f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 13:30:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6dfce238fa60fcb8a4039347e9cfbb9602241ae3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-04 14:31:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4ed141ac06275db0bd3ea8928eeb6121d1243dcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:32:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"639ad5431eda432b53e1c9904a358caccdaeacb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:17:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6ea3529cc09274963d6e05b5cd255f9e622c47e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:17:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ccd2ab95f8c48c2a40c369903e578dd : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff53b5d3c4344f80876b400218f67162 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b61e9ea5a27145c9b52307064b626846 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0a6273ae4844b17ba12d6397f92742e : SUCCESS in 6m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a6023f5f3554872a57c5547b0f1ff0c : SUCCESS in 7m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77a2c43ed106403ea544d8a9092ce6d2 : SUCCESS in 6m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5254b86614e54ffa86f7b90ec52d148c : SUCCESS in 11m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/04ee8e9036a7419fbe43df7af685296c : SUCCESS in 13m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cd82d3b21d2c455ab389821c68d6e19e : SUCCESS in 14m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/829095ae092c4105bc2af6ffd2fe8762 : SUCCESS in 8m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e8da46c2bebb470ca108b3be90869bb3 : SUCCESS in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/96abec6661cb430f9c11bf2200fe0330 : SUCCESS in 12m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/da49e5d8d2814a66a811e030b83725f4 : SUCCESS in 26m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/008b2799434f4c33a1fdd39bbdf013ba : SUCCESS in 31m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/56eb0b98115a43129f067e3226f951e7 : SUCCESS in 19m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7e7eef157c3245d9b8b01a1426bded17 : SUCCESS in 16m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f1661df6de8547058a881f8959229c0d : SUCCESS in 7m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/088543cf35d24264a1428ba79cddc53a : SUCCESS in 10m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/af32bf3c434345dca26539f132e8cd53 : SUCCESS in 15m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/884377f539d042d1b7cc74ce8e9890d8 : SUCCESS in 15m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a68526bf1c5407ca38072191278711c : SUCCESS in 55m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/10b2a199ba9a46d5be3039a2f8b93572 : SUCCESS in 1h 12m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55a4b417da3f40dfa70db06f94f90a86 : SUCCESS in 1h 33m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/0bcb4eb7cc4246d9aad90067797bf3d1 : SUCCESS in 1h 14m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b3a1a32f1f7e91b04bbacbfc5d6adee1ce60f0c","revisions":{"0b3a1a32f1f7e91b04bbacbfc5d6adee1ce60f0c":{"kind":"REWORK","_number":1,"created":"2020-06-04 10:56:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/733553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/733553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/733553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/733553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/733553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/733553/1"}}},"commit":{"parents":[{"commit":"aeddfbeac59589aa4fe39c5e74b306a02678536e","subject":"Merge \"update constraint for ironic-lib to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeddfbeac59589aa4fe39c5e74b306a02678536e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:56:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:56:57.000000000","tz":0},"subject":"update constraint for oslo.serialization to new release 3.2.0","message":"update constraint for oslo.serialization to new release 3.2.0\n\nChange-Id: I4d9744e092c809b7ca94738fa31f285210bca5ff\nmeta:version: 3.2.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b3a1a32f1f7e91b04bbacbfc5d6adee1ce60f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b3a1a32f1f7e91b04bbacbfc5d6adee1ce60f0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
