)]}'
{"id":"openstack%2Frequirements~733869","triplet_id":"openstack%2Frequirements~master~If7906a967b8dc3e81c6f2ad14d18ca0f39dee295","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"If7906a967b8dc3e81c6f2ad14d18ca0f39dee295","subject":"manual update via generate-constraints","status":"MERGED","created":"2020-06-05 15:32:49.000000000","updated":"2020-06-05 21:28:40.000000000","submitted":"2020-06-05 21:27:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733869-1591392428460-0e537f8a","meta_rev_id":"c621d72f031cb5ccf6fa7c24f9bf8edcf7ae6e6e","_number":733869,"virtual_id_number":733869,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-05 18:42:09.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 21:27:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-05 20:18:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-05 20:18:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 20:18:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-05 21:27:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c51153df9b53e6136cdc5be6b941f3405dbbc1b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-05 15:32:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80ba87fac9bcf248c59aa4a3ecd9359fa1289ca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 17:11:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f9fccf27afe446c834db23a12339d19 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4649f0ad22e404d9d0e0614370796c0 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1df8142557f4fc49121a65fda2c0620 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91752066b78742aca126d74fce1c7354 : SUCCESS in 6m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7846e5f0f27e4c21b6fdbbb158ae8d6b : SUCCESS in 5m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b7428194c0342deb2469e13f6f5fec2 : SUCCESS in 7m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/389a01f4598e472babf7c5c6ef21ff51 : SUCCESS in 13m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ef114a3055044e43adece7f8f920955f : SUCCESS in 7m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ec60545fe42641069b0b6abc974964ad : SUCCESS in 8m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/288f8b19d3fa4c1eb53513e9b45d48ba : SUCCESS in 8m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6f34432e6dae477da702b867e7997bf6 : SUCCESS in 12m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/513760b28bb94efbb599b5818b324316 : SUCCESS in 9m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3e49e051ee7b41eb874e36c65dddb162 : SUCCESS in 14m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/80f2addfb2dd4d029b6c6d777de6c0c4 : SUCCESS in 28m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b4fdb9ee4b4d476aa5892a420e49de70 : SUCCESS in 15m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/23eb3350dd90435cbda812be99e7b6dc : SUCCESS in 20m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3ba6ed5013b84cada16e6be8846c19e4 : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b29cc330bf4f470caf7562eb0b00545f : SUCCESS in 8m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/38833bbc2d944978a96a801fceca0372 : FAILURE in 14m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/731cc742b1984b4393b6f758d7a7a8ff : SUCCESS in 51m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9863c517621b453faefce89925f1f356 : SUCCESS in 1h 22m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26cb364dd8a0483e9fddaaf47be10001 : SUCCESS in 1h 30m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/2e7501f895a24064acf8bfc1e65ec679 : SUCCESS in 1h 16m 01s","accounts_in_message":[],"_revision_number":1},{"id":"66312e627c681bcf5c4f4d69935b4ffff81c2b09","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-05 18:42:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"af9ee9189187bfb29f8f12ae3b453afcc385de97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 20:15:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0cf42747ae042519ea7eaf2538d34c5 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce25bb6eae2d46e594a50cf3a2ab6269 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bc370309db548e380ad3d99a13152fe : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86090e762dac4d24a5655707b06a6e32 : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff6d5e5c7d73434ea30b43d4f3dc49ed : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5bb86b4273ae4e3aa80c8edc1d9df7b7 : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f46a3401e5c549a7a3fad743769497a9 : SUCCESS in 8m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/016a480a9d904219aa17c73f2888da83 : SUCCESS in 11m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8577ad5ef66749a2911263aacb4b84b4 : SUCCESS in 9m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9b7618601cfe45cb87aa36e850ea7114 : SUCCESS in 4m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cc25f1a391fa46a5b61296483d5ef16c : SUCCESS in 10m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0fda17d8a3034c5fb3235e2a29212c3a : SUCCESS in 7m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ceb39c786f8647e78e4dc4699b79b493 : SUCCESS in 15m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fb18cdfd766b4d5aad35299810408f23 : SUCCESS in 24m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/70de1c883cad4c1681cda57cc81dbcfe : SUCCESS in 13m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2e9512bc443142ae8c71e8f6af66dc87 : SUCCESS in 20m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/50bd568058774764820fbf9fd64c87c0 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b9ee78efb7145d99c6788c7c8022544 : SUCCESS in 6m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/44587d37610148cabc0a8c4966ffb9e8 : SUCCESS in 10m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f41145490b6949bbb68de4baf78c70f6 : SUCCESS in 47m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5bbd6135a4d40d6a736b189a8a3eec7 : SUCCESS in 1h 03m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1524184f454b450f98a7fbcd606297dd : SUCCESS in 1h 28m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/4400f5c34f3643f2a4ac4a34b464d0c7 : SUCCESS in 1h 03m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b7e2c4626acad39bbd9b9388faa3556c58588e77","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-05 20:18:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9dfc023e434780dab39935972aac6424c8fb6c7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 20:18:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7268de9b8274986edc285c23ee55c992ff0123bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 21:27:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec173f3061f4fad80aa7545236b1b81 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4424a06f513b454a973769521d9f487a : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78417ccdced24618b16864489f463217 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f18b641e4154bc7a0c476b2eaec15ca : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05a5b606c64c4b03b3141e46feaebfc0 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/467cfe0ddf4846dca0b8346102943a79 : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2fb17cecc0d4486c928deb52b2fddb7c : SUCCESS in 9m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0f0dff35edd84c33a47b009af5141616 : SUCCESS in 10m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f9168b78f4b84cfca95efd9440cab86c : SUCCESS in 9m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bd89e97ef76248d9961464e61bbbc8df : SUCCESS in 5m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dceea96a5fa147a1ac3394da3c5f0469 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/db6ebcb5e265408390c0865bf0b74f85 : SUCCESS in 9m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7f6589848866462480cbd1d4043ed0c6 : SUCCESS in 12m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fbb28655de044530a8a7dc05ba96db79 : SUCCESS in 31m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/32c4c1c0107e4a9f9736963dda97c9e3 : SUCCESS in 14m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/45190af2c0bb426fa0353f253dc21a14 : SUCCESS in 13m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7db04db2211f470ca901842387b4476a : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce90536b3da94ce7a94aa777cc0773c6 : SUCCESS in 6m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a5dabfd42add4e519d5469927c46e26a : SUCCESS in 12m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/35d712655b8d4eba9244b07120495015 : SUCCESS in 9m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13b4e610a3dd44d09c75e3669774bae0 : SUCCESS in 49m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ab3a1717070408794f6eb47cf7e169a : SUCCESS in 1h 08m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/241c0960b9ee4264835e48f19de13a3c : SUCCESS in 58m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/0d74cbe7e54c47cc9c64261f803b2d63 : SUCCESS in 45m 40s","accounts_in_message":[],"_revision_number":1},{"id":"86592fc2d52fd69c36ba745ede19106a9ce0cffb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 21:27:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c621d72f031cb5ccf6fa7c24f9bf8edcf7ae6e6e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 21:28:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69d2b6864ac44d999157d2c3e25b45eb : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b14f43f19d779a952845b9c4ecb76f1a2e5fab1","revisions":{"2b14f43f19d779a952845b9c4ecb76f1a2e5fab1":{"kind":"REWORK","_number":1,"created":"2020-06-05 15:32:49.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/69/733869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/733869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/733869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/733869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/733869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/733869/1"}}},"commit":{"parents":[{"commit":"11654db00411e757794a662affba3792c009fda8","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/11654db00411e757794a662affba3792c009fda8"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-06-05 15:32:20.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-06-05 15:32:20.000000000","tz":-300},"subject":"manual update via generate-constraints","message":"manual update via generate-constraints\n\nChange-Id: If7906a967b8dc3e81c6f2ad14d18ca0f39dee295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b14f43f19d779a952845b9c4ecb76f1a2e5fab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b14f43f19d779a952845b9c4ecb76f1a2e5fab1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
