)]}'
{"id":"openstack%2Frequirements~733918","triplet_id":"openstack%2Frequirements~master~I0ff9e781a7470e3432129408114048b4dbdf2478","project":"openstack/requirements","branch":"master","topic":"oslo.log-blacklist","hashtags":[],"change_id":"I0ff9e781a7470e3432129408114048b4dbdf2478","subject":"Blacklist some versions of oslo.log","status":"MERGED","created":"2020-06-05 21:00:24.000000000","updated":"2020-06-09 20:51:14.000000000","submitted":"2020-06-09 20:49:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733918-1591735767508-eaf98a30","meta_rev_id":"4000811aaadc749f4ca91aa4fda304eaeb5d534c","_number":733918,"virtual_id_number":733918,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-06-05 21:17:51.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-09 20:49:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2020-06-05 21:03:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-06-06 00:26:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-09 19:04:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-09 19:04:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 21:03:34.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-06-06 00:26:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-09 19:04:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-09 20:49:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c13e604a1ec7b337e611335baa2cdc2370f87e3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-05 21:00:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8f98d0d4875ba9341bb1ab25dfa6561810e79d7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-05 21:01:09.000000000","message":"Patch Set 1:\n\nI\u0027m assuming this is the correct thing to do about these versions, added Ben for sanity check.","accounts_in_message":[],"_revision_number":1},{"id":"db39d4d71346079d7e47f8bf0c286d3d42aa47b6","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-06-05 21:03:34.000000000","message":"Patch Set 1: Code-Review+1\n\nYeah, this is good to do. My only reservation is that there will be a new stein release at some point to fix that branch too, but we can always blacklist that as a followup.","accounts_in_message":[],"_revision_number":1},{"id":"09064fe8bbcccdd93fc834d8a6aa243626d1c521","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-06-05 21:17:51.000000000","message":"Patch Set 1:\n\nBen - thanks for looking.  I couldn\u0027t find a recent stein release in the deliverables dir, figured it was maybe in EOL-mode?","accounts_in_message":[],"_revision_number":1},{"id":"a067a92ee9584db53537bb3c905db0adb8118164","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 22:41:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3da34f55dbc4621af69f761bbb42040 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7722fded05b041a785012ab1ac58eb7e : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e922fbdeb966433284e9edbe520a6bdd : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b13005e27b7d4aa498a79ef472b5f07e : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa17b8e0ba1a41c6a0bff1b74419fd53 : SUCCESS in 3m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bee4d23406104cc599df50ef156669e2 : SUCCESS in 47m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc60de1ec65441218a9d9a7f1711db8e : SUCCESS in 1h 02m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83824bd5a5e34e44bb670bbca7401046 : SUCCESS in 1h 35m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/c4192419e5334bf986cc0f27beba8cf6 : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e7a64319515e77d201f77326951224ec06574654","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-06 00:26:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"83077dd43b9be48adcaed1ba53319e12191edbd3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-09 19:04:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b3527334885b9f9299d254c8ceeb2558452be77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 19:04:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c51923cb0664a519648ca41715f8376381d8d8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 20:49:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f7854334bcf8fe8e30ab8666905ed151a216375a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 20:49:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b3236ac9a5496e9d2c587ad90de055 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f61cc16b62c4e9a8e93cbdecf6b79b3 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edf1d045c0574457ae971594fd89eb5b : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9e5b83b3b19419d9072725392b817f0 : SUCCESS in 7m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4dfe69a047cc4eb8adff047ee61c2932 : SUCCESS in 6m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7aa7ea7d14da49dab0e9c05656547eab : SUCCESS in 56m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/773940a493d34d158c392b9309a373eb : SUCCESS in 1h 11m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aab56ca78ff94b04ac957f2ed5cd38e6 : SUCCESS in 1h 14m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/57608ea88dcb4e9bb60d0f87a8d296c8 : SUCCESS in 1h 07m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1d6236eea94b556cf31fefa08f89e0048c48fe5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 20:51:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bab3434cff0d427da41104d19cb2a23c : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f835dc2832e04f6b4e820cdb485b5a18389dbe48","revisions":{"f835dc2832e04f6b4e820cdb485b5a18389dbe48":{"kind":"REWORK","_number":1,"created":"2020-06-05 21:00:24.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/18/733918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/733918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/733918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/733918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/733918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/733918/1"}}},"commit":{"parents":[{"commit":"11654db00411e757794a662affba3792c009fda8","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/11654db00411e757794a662affba3792c009fda8"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-06-05 18:20:30.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-06-05 20:58:43.000000000","tz":-240},"subject":"Blacklist some versions of oslo.log","message":"Blacklist some versions of oslo.log\n\nVersions 3.44.2, 4.1.2 and 4.2.0 of oslo.log were not\nbackwards-compatible and caused failures in the neutron repo.\nSince versions 3.44.3, 4.1.3 and 4.2.1 have been released, go\nahead and remove the bad versions from global-requirements.txt\n\nDepends-on: https://review.opendev.org/#/c/733878/\n\nChange-Id: I0ff9e781a7470e3432129408114048b4dbdf2478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f835dc2832e04f6b4e820cdb485b5a18389dbe48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f835dc2832e04f6b4e820cdb485b5a18389dbe48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
