)]}'
{"id":"openstack%2Frequirements~734110","triplet_id":"openstack%2Frequirements~master~Id8801ead9923bf11d6646659cf4e699af215ae14","project":"openstack/requirements","branch":"master","topic":"remove-mox","hashtags":[],"change_id":"Id8801ead9923bf11d6646659cf4e699af215ae14","subject":"Remove mox library","status":"MERGED","created":"2020-06-08 14:46:51.000000000","updated":"2020-06-09 20:57:25.000000000","submitted":"2020-06-09 20:55:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734110-1591736156529-44efa8ae","meta_rev_id":"0275799a4567aef236ded731bb9e3cfb722bf849","_number":734110,"virtual_id_number":734110,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-09 20:55:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-08 19:47:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-06-09 19:18:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 19:18:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-09 20:55:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"639ef383b1344b126ffe99074b10abc3344579b7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-08 14:46:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d0ec56df541bb302874f3d42e19c6228fce1b68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 16:15:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11cdc6d0f4f6404e80cee2de1f9458a0 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36c7bbadab8c414db7f06bdfec14558b : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61a983a9dd4f4ca9b4eaebc866d288e9 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d136845cbc6413895e05c0d6cdbd3ba : SUCCESS in 7m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3b42148be8143098d5f13ab17f6f265 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/187a18e4d36d4dcf9440899b9821cee4 : SUCCESS in 4m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/13b8b4cbb16a4eedbcae21440c8d287b : SUCCESS in 12m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bd2d1618d6094f0b830d7a175c419d61 : SUCCESS in 11m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/54bcd44092cf41f3a99b7f89c5168e89 : SUCCESS in 10m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4c81e35b688348ca89ed2edc17b2f18b : SUCCESS in 8m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bb875f676922486d9f8613ce7ae3e7de : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/94894f868ea54bd8b99800c044534985 : SUCCESS in 8m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/28a8d0745de54233affa9ebababf1845 : SUCCESS in 26m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6abb5da786084b96af2ff118a0cbd8c3 : SUCCESS in 24m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c0cb0926ab01485993beef2de62cb78e : SUCCESS in 13m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f63e29c085d4adeaf75b179d4cbde37 : SUCCESS in 22m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/36d2aa6c09a94ae69b110794b7e41649 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8e7ce716bcf846c2a75cb566134425b6 : SUCCESS in 8m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cce1812e4b714d3190a503216bd830af : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e59a7a30365648bb992be9f97563eed5 : SUCCESS in 50m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6bca5268ff2f4d4fb9832b58e74577cc : SUCCESS in 1h 09m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74ada60a578d4309b482fc31f96a6246 : SUCCESS in 1h 22m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/777a144a359248348c86fabcb24c0b0c : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":1},{"id":"77ec5941354999aea409ea4efd6be7247bb4ca5f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-08 19:47:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ebc8d0127bd3118eba17e5652c28b814bd12209a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-09 19:18:25.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b54a5e70335c02edb25f025cb717fdefc34469c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 19:18:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89605fe8ab5957c1f85f293039ee5565c2662eb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 20:55:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"28ff6fdd0a788821b0d7ee1e6bb0ada26dada9af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 20:55:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ecef4294166440da57a566732607d5a : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aaa1a1233864949ac63301c54bb92e1 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74af40f6205343c780e2b4990cb293dd : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5b7a5fe9cad447d8df173ed56ac4cd7 : SUCCESS in 10m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54c9a206819249f09c54d130ba3cce21 : SUCCESS in 7m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5b32659e2ef9439fb36a5a4349b7bfdc : SUCCESS in 7m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dcca7627aa5840b3869b018b131615b3 : SUCCESS in 12m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/17a5e6c3c854451790853f371b2fbbd8 : SUCCESS in 12m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3c2cff65874c4d95b4b4946d09cba4c9 : SUCCESS in 16m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e460fe271e844207b5afc9cdc96a256e : SUCCESS in 6m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/16fe6276074442b7a288e30bd4486f7b : SUCCESS in 15m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c1c9786ac844d75bae65d29302cccb9 : SUCCESS in 9m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5de9c43714f840d68b02ff020b6f3c92 : SUCCESS in 16m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/49e5b0f6378b42cc98022ce5a58b55ce : SUCCESS in 38m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9db8bd16d6c74c0aa397f511f4e34b15 : SUCCESS in 17m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/07f4f1d6e5d44a4ba31179f0142eedb1 : SUCCESS in 19m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4f33b079bcbb41dfa4e68d42647a697d : SUCCESS in 6m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12b9fa20dab44fcaab140207d73b7dd8 : SUCCESS in 8m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/98ac113fe66543ddb87840ff04dd3ccd : SUCCESS in 12m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5fc86bc7d1624d28bb051cdd88d675ae : SUCCESS in 12m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8c17e456fde4872ba374c2062ac7981 : SUCCESS in 1h 00m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c42c098349342bfb0916b455fb4a8a9 : SUCCESS in 1h 14m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ceb50f7c4a8846f88f518586af29bfa6 : SUCCESS in 1h 26m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/6ec72bc6019a44b4a6f24adb6ab39679 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b7d00e83f30a213abe373664f4ba23cc9453ab3f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 20:57:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44297dd2999a4ea0a057f3cc1f181724 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2567bf17d3b492d97f63dc81e18edc220c0b7d31","revisions":{"2567bf17d3b492d97f63dc81e18edc220c0b7d31":{"kind":"REWORK","_number":1,"created":"2020-06-08 14:46:51.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/10/734110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/734110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/734110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/734110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/734110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/734110/1"}}},"commit":{"parents":[{"commit":"2b14f43f19d779a952845b9c4ecb76f1a2e5fab1","subject":"manual update via generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b14f43f19d779a952845b9c4ecb76f1a2e5fab1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-08 14:46:12.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-08 14:46:12.000000000","tz":-300},"subject":"Remove mox library","message":"Remove mox library\n\nmox is a py2-only lib that is no longer being used. This cleans it up\nfrom global requirements and upper constraints.\n\nChange-Id: Id8801ead9923bf11d6646659cf4e699af215ae14\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2567bf17d3b492d97f63dc81e18edc220c0b7d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2567bf17d3b492d97f63dc81e18edc220c0b7d31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
