)]}'
{"id":"openstack%2Frequirements~734124","triplet_id":"openstack%2Frequirements~master~I7d2cc431d9fd0a80e29c176da6ba612b1262dfda","project":"openstack/requirements","branch":"master","topic":"ujson","hashtags":[],"change_id":"I7d2cc431d9fd0a80e29c176da6ba612b1262dfda","subject":"Remove ujson package","status":"MERGED","created":"2020-06-08 15:26:24.000000000","updated":"2020-06-09 23:55:19.000000000","submitted":"2020-06-09 23:52:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734124-1591746754365-44efa8ae","meta_rev_id":"83c1660efa80851f5e071ae76f5336689a830bd5","_number":734124,"virtual_id_number":734124,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-06-09 20:56:38.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-09 23:52:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-08 19:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-06-09 19:18:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 19:18:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-09 23:52:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2378c147373fd48eb52b43a11ca279f78e4bc3ef","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-08 15:26:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1985ba30f81e0dea4be9f15c415118e537f051d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 17:01:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50a3d6957ab04785b2536dc912bec103 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bade877b973841988d31dbb0f22eab03 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a177d3e9f5d84c76b75b0b57558b0a21 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78226e66161a4f4fa922cb3ee47e1633 : SUCCESS in 7m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05773ce9c9f84a3291bab19addd0f854 : SUCCESS in 7m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/369c90a9c9894f7ba30d4d519be8a016 : SUCCESS in 10m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3b36e2715dd64782883fc2653f55d205 : SUCCESS in 18m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d0455c60cc7741fb9f2301d9c17d447b : SUCCESS in 18m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c6e14663d7e54832a53533f8d9bd42e2 : SUCCESS in 17m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/63450044d86745f2a3a693e9ba968ee4 : SUCCESS in 9m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a821d8815f494e168483e9ae1b5f3cbf : SUCCESS in 13m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d769017bbe3e4304a69132b72cccbd88 : SUCCESS in 11m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/143063c3849d457face30f0204050715 : SUCCESS in 16m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ed4f07ef62294cb9948806a89b24f237 : SUCCESS in 26m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d0f5fa73840049eca13edf17b478f738 : SUCCESS in 15m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/df62d6d45c1a48ff94a7f1da2ac30e9c : SUCCESS in 20m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e92cfb9577b04537bf1295117cb5f615 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b5ec6410c494bacafab7c2e801bebd8 : SUCCESS in 8m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c82e2c76b28a48d08d51db30551596cf : SUCCESS in 17m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88edda46bb22465ea3f9eaf9a55ed02f : SUCCESS in 1h 03m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1aaaaecedad7431ea8bf920624804572 : SUCCESS in 1h 07m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33e0a1323b1b4ae88a923c5570925b7d : SUCCESS in 1h 23m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/3c3e828ed2b74ceeb2131a188ebdedc2 : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":1},{"id":"08e487605255824810aefdb95cd3b7cd2ae9dc1f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-08 19:47:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c4d81443dae4afe2b33bc499f27edec057b6d7f4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-09 19:18:29.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"866f5e53143317e94328767eaf321ade523941c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 19:18:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3a6d3ece20b8ab71a1de1693f4e5214c50bb684","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 20:55:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba6d6e49cf1244f19e04b2b290b79126 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b8d8528f04d43afa6f6e0401b89308a : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8ac97380ff4406aa7ace98a9eee0350 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db4869f7fcd9451f8e5fd56a192d6a56 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd6ff372c83c4af49a38c2e835f70ee3 : SUCCESS in 6m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aa07ec3d1c5144f182994ffb812f5519 : SUCCESS in 5m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b4cf3db942744b9587f8bcd10845f907 : SUCCESS in 11m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/14bc1d7be0b640a3a5b80c7c3233e009 : SUCCESS in 11m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3676d0b0644948e3a03e0c9a8abae735 : SUCCESS in 9m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0a1cd33bac994381b7836e05a9b238c5 : SUCCESS in 8m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/566c7a0781074f8a94bd2cb6d6da516f : SUCCESS in 13m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b0e3ff7f9444c1d97f4e3be2192ffc3 : SUCCESS in 7m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c73e5f0c81e24d38bf51f13783a29d95 : SUCCESS in 13m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/69e7994889194e1f9ce3e07815262721 : SUCCESS in 28m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ec6a2df210424db1b76542e06879f5b5 : SUCCESS in 14m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/633fdc2120c146a4816ecc016622da80 : SUCCESS in 19m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cf53dcdf1d3242f9a6f877586a8acc15 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/73325405728a4ed59c000733bb98beaa : SUCCESS in 8m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/57fea4ba587c45778640ac3fb4042dee : SUCCESS in 14m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f8b9f1db79d54fd8bf27869334722eca : SUCCESS in 13m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eba4c0585db649f2887dc3818ad6831e : SUCCESS in 37m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3998ec61d91e4e1babc326d41e258202 : SUCCESS in 49m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ae3dc0cc29f4ad3a23c00b1d53fc51a : FAILURE in 57m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/04a672bfcc064d49ba3df9db1be86d4b : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":1},{"id":"710cf0285d118bcb2065e694343c866c4bd02008","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-09 20:56:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0c8b0452d5881c3a43bf7103011d8b1bbaae8831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 22:27:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42677c9d40a7408fb5dd2e1ce70c67e8 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e96dd936e134eb1ab50458180cd509e : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdb1f5fa5db44da88d9d1ad15c681ebf : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29d1db60ffab448ba5ae4ac595957eca : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/083412759fc64cef9818e61750e4176f : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bfe514b84f2f453ab874780766d63431 : SUCCESS in 4m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/23fd7058006a4a399dd4e96b8abff37b : SUCCESS in 10m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/628095485fb1464582d7235981beb3a9 : SUCCESS in 10m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0cbaceb557b64f739c74e6e52e4027e3 : SUCCESS in 10m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a2c5f543477c4636ad29379a2672bac6 : SUCCESS in 5m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c35d236bee9741bca380f2f3ae3f7cf2 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d07dc14f9b19467e8ed1d8be0c0af266 : SUCCESS in 8m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cc7f2b6faf164dc4829788a152daa462 : SUCCESS in 13m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7e210d2b25784f9185f3e56c1ca14da7 : SUCCESS in 27m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/80fe0a503ba944dc9fb0670f0cb3b063 : SUCCESS in 13m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/928d85c7b2ea4dacac9a8c7a53355035 : SUCCESS in 19m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1d38606621b74e74bd413d6f0d0bd62d : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fbe989520b6142909b972a494f17074c : SUCCESS in 8m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a0e4bc01a6a1445dae9167fb8c7577b3 : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d03af4d0cfc4e4989d0aac35c6f8b39 : SUCCESS in 48m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cee04766eaeb496f80751882b18978e9 : SUCCESS in 1h 23m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76b131313c134bf7a638f66215bd92cf : SUCCESS in 1h 18m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/c6114175de6f4b5e828a61aba350f2cd : SUCCESS in 1h 09m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6f6bde270fa429da3621c89b9604e2d6fb65211c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 22:28:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e51969622d329d9e4e7e150363d6d9b960089444","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 23:52:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06926180844b4e64bf942648ab7ce45d : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03736fc6987048cf9afd99ee92b8d04c : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/220d249594704f64a6b4292a47f1610c : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/378e906d7a114777908497523f6dde13 : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c3688bcfd304c80b0b28fa9413d4b07 : SUCCESS in 6m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4322d8462dc54234a62870294d364851 : SUCCESS in 5m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9513dea8b7b84126ad57e942740aea99 : SUCCESS in 10m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d1f6840da7214c97853564b8c789d5f2 : SUCCESS in 12m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aaa91435eced4c239977c17130a8ab00 : SUCCESS in 12m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/32b756e3a62043eeac3c8d723e43d0d9 : SUCCESS in 7m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7bfe7d77f6d74928acafb918242fa5a7 : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eb7d68ea56ba40bfb469c41eee782ce5 : SUCCESS in 7m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0cbe6244a2b44e82b752374aeaf698c1 : SUCCESS in 13m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c6416c677d5c4ee1889b5c3fb820a8a6 : SUCCESS in 26m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/335ccc9ac2bf4a3cb5a6ffcbb441555f : SUCCESS in 12m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/27cc9f706e4e40f39deb1cf8c0568bbb : SUCCESS in 17m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/03242c8726464efb81183d3dbb4b4f4a : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d2d25111a14455487b7db0988bd9374 : SUCCESS in 6m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0ae6b54926db47c894a639d35b613541 : SUCCESS in 9m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0903c82ed3c54e97880d249a82b419e4 : SUCCESS in 10m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a7f26d6e6c242a19f0cb57c09394efe : SUCCESS in 53m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/03a781e584a841e78f9caac7db86b938 : SUCCESS in 1h 14m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7736936b53a421fadebdb5a69ab7bd9 : SUCCESS in 1h 19m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/e46a4cb14c0f42c593d5f4173ddb5f14 : SUCCESS in 1h 01m 29s","accounts_in_message":[],"_revision_number":1},{"id":"8904657da4f572493d7331cbb6be733bab31adad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 23:52:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"61f689d7b39f34a9c2b3d123656088456e2c23ee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 23:55:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c897b1ccb624129838305c02583e172 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"463049460ba0e7942fce07cc121deef584b5bdec","revisions":{"463049460ba0e7942fce07cc121deef584b5bdec":{"kind":"REWORK","_number":1,"created":"2020-06-08 15:26:24.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/24/734124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/734124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/734124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/734124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/734124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/734124/1"}}},"commit":{"parents":[{"commit":"2b14f43f19d779a952845b9c4ecb76f1a2e5fab1","subject":"manual update via generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b14f43f19d779a952845b9c4ecb76f1a2e5fab1"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-08 15:26:06.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-08 15:26:06.000000000","tz":-300},"subject":"Remove ujson package","message":"Remove ujson package\n\nThis isn\u0027t used by any openstack/ projects anymore.\n\nChange-Id: I7d2cc431d9fd0a80e29c176da6ba612b1262dfda\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/463049460ba0e7942fce07cc121deef584b5bdec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/463049460ba0e7942fce07cc121deef584b5bdec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
