)]}'
{"id":"openstack%2Frequirements~734425","triplet_id":"openstack%2Frequirements~master~I03d4e2cb87246fe867092dd4d95b2983b824b57a","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I03d4e2cb87246fe867092dd4d95b2983b824b57a","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-06-09 06:25:49.000000000","updated":"2020-06-10 20:52:25.000000000","submitted":"2020-06-10 20:50:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734425-1591822201258-00e32a87","meta_rev_id":"a6b4b82ccc3eeb188bafaa5ebc488239afad2ac3","_number":734425,"virtual_id_number":734425,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 20:50:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-10 18:58:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-10 19:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-10 19:08:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 18:58:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-10 19:08:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-10 20:50:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9c8d6acc1b52770bef9fe4acb097532baaf53eb","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-09 06:25:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ace2b25ff9b2c6ff246c0add59bc296d92fe444b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 09:00:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b67b474e4a64c919c5aa1ee14d1dbdc : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05f91c82b1654c0ca655cf1e1dc72bec : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b922431ba31c4d47b930f6bb069aeb88 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39e8b7ad6c404ea8aa6e62c74361184c : SUCCESS in 8m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf06baed84204e8db21ff9e614e9100d : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c066323724a544e1be52f58d66685d1c : SUCCESS in 5m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fcef8aaedca2458faf367ffcd3ceacc8 : FAILURE in 11m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/73736e5c551845f6942a0c96851a5ec9 : FAILURE in 13m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7dc01e111fed4f6daf6588b8b224fd41 : SUCCESS in 13m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bb54c56762234c1da4c720a153e25581 : SUCCESS in 8m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/95eaf466b1a246a098490f62118ac7b1 : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb6c14938db54e54b192edb9a10e2aeb : SUCCESS in 11m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/937d2311c02c42a69216042fb233b063 : SUCCESS in 19m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/874cb7f4917748c993614479794bea65 : SUCCESS in 19m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3561d947628e451b8868c93a4b29ab1e : FAILURE in 14m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/72ebe1ee7ffe46cd9c689fa2201af9bb : SUCCESS in 20m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/46b80892107446979737a889e73cef55 : SUCCESS in 8m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/798d88ca33304e77bef5d5ea3308921e : SUCCESS in 9m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/105e42a57f334069a6aff812f4c88a8d : SUCCESS in 13m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73f98381976347d8ada4b71721331607 : SUCCESS in 56m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2334d60b9ebf4f23add11214bab66ddb : FAILURE in 1h 23m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec28e0787ac445bd94233fcbab150d80 : SUCCESS in 1h 35m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/3011c6574a5d46f5900ed47bfa4572a8 : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6eb11495618d5b85d970536fb3c1a84f7b2be3a6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-09 22:53:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"73df171e12c1d00968a08fb71ae383ceb57392a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 00:07:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4666e82693dc4a18bc839ef08ecb94f8 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9505874e644e4f4fb96a8abc3488bcd9 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afcbdf23bd8241a3bd759bef983441c8 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7502c343e0644f2db7fbf7289b425b70 : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/72661560712b49268861c895aeb1f3e6 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/37e524d83e0240e8930acce5534f494b : SUCCESS in 4m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4c5f3f14ac4746648110adfdb894cc98 : FAILURE in 7m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c4837e27ebcb416bb685484af2ea7fd7 : FAILURE in 10m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/162cb3f322024d23b3e38b0a7211903e : SUCCESS in 10m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/284931891ff64bbf866a8a512cf6f245 : SUCCESS in 4m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/841d5e48016e4350896e1c4cbfae55b7 : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/91e500a256564a0196c0ba8409e3c2f5 : SUCCESS in 8m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0032a956166342128ea8935c3a0d98e1 : SUCCESS in 20m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9f217f1fbd2243eebd41288ae3c45ecd : SUCCESS in 18m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5005fee81a53434697223183310a5650 : FAILURE in 17m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bbb6abfc7abc4058852ffcdcc5e302c7 : SUCCESS in 17m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0efcce03d43e4519b61eb184059fb3d3 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/02173fdfe50a45628d91e4ffa44e336c : SUCCESS in 6m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/be46c81fea7c46b3bd9da842509b5aad : SUCCESS in 13m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f331fa8c50ed4612b9d5135bfc438ddf : SUCCESS in 1h 02m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ab95246b0434992a92664c0e024bfb2 : FAILURE in 52m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8bc659fa8ba8490981139971c225b011 : SUCCESS in 1h 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/ffea55dbbefa4381bbea74ee8455604c : SUCCESS in 1h 11m 27s","accounts_in_message":[],"_revision_number":2},{"id":"9106c2d219bce5a8492f5698a76f40d6403c014e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 00:54:46.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/474e2eb8e045475c8c164c9a5a0393cc : FAILURE in 1h 58m 52s","accounts_in_message":[],"_revision_number":2},{"id":"8c8d26711b53822cd98173cb68c5b4a023b0ecf0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-10 03:44:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"25729534f938efd940be8dac755b75cb9eabee38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 05:10:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e2db8d75a94af9873367c9ce1ad67e : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de2a5d0452dd48ad83459cb648ebfc6b : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b355a29e4a064894944f77b1a37a5794 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db80b26cdfed490a80031545e4159975 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd04c545fe82405296a0b45dd7e43386 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/02c7d08313bf469d89ead7426821c9d8 : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b95dcda8da5646d1aa23a52ec34898cf : FAILURE in 8m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6c7069ab05a046408f881bfbe7b892ce : FAILURE in 9m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a56024be3532427c919deb7952e93d26 : SUCCESS in 10m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/21f9f60e977145acabfdb47ae3debbbb : SUCCESS in 4m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5a45c3d8fe9645beaabc34d3ef5e5458 : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0be1a68d3e4e43f9b34783c4dee235d2 : SUCCESS in 7m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/725ebf7eace042fa9f00ee371afd5e52 : SUCCESS in 21m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/66246c01795e466993acf60693346fb6 : SUCCESS in 23m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/82e3e687cc3c4fd3b3ccffcf1fd95f56 : FAILURE in 13m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e5761bdd804491a8d3c8c68a02b1e0c : SUCCESS in 17m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6602735ea97648a697a141757347253b : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/41969a4551d4441db8652c018435d3e3 : SUCCESS in 6m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5fe10a4e63634bb1a9e2fda3180ee562 : SUCCESS in 9m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3241e4cc2ef4198a3a65c1193d9043c : SUCCESS in 50m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cfd7812ede04d099a53a64af793ae16 : SUCCESS in 1h 24m 42s","accounts_in_message":[],"_revision_number":3},{"id":"a1637aa841c4bc0cd4b329e669a6d766e5421aed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 05:53:11.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4b155cc21015479d8fca59763a96f5f8 : TIMED_OUT in 2h 00m 57s","accounts_in_message":[],"_revision_number":3},{"id":"48ade9d010432afe7ad77fa9df93b4e6a1fb63b5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-10 15:59:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d6a83df5735e0d6fb3f68871a67fc844fcd01e77","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-10 15:59:43.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":4},{"id":"c6cd8793ca91b9e90e9c9fa2634949d010418ec2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-10 17:50:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bfc9869983b6a8a449429297405be1bbb35ddc1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 17:55:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/978a0c6f1872400981f1d658a8ef5830 : SUCCESS in 1h 52m 16s","accounts_in_message":[],"_revision_number":4},{"id":"7ba52bc6d19ce8006e0c19806d87cb722703be0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 18:07:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e975091d6d1149c7b44df2a8167df5cd : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/317743896f4a4b0db4363307ece5d80e : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f1bb226fb2e44a4a819b9612a635918 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/200c9163919f443883b337522c1f7144 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cacc5ad691c54b8198085005d9a1a488 : SUCCESS in 5m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a3f8eaa864054fb19b44e8539fe1185c : SUCCESS in 5m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5956d506645b4e5a84275d2bc4103ba9 : SUCCESS in 11m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2a598abe09f24d09857385814987002e : SUCCESS in 10m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ca654c5f7f544c7aa1f410a5e152af33 : SUCCESS in 9m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/40bfc640363148ca8ef448bf7a3b657c : SUCCESS in 7m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a52fe0dfe13f4c42a684f25dfa3bfbf3 : SUCCESS in 12m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5c8fc21e0d664a3ba8106f96362f571a : SUCCESS in 8m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d28390680f424dc79efa06d217342ba2 : SUCCESS in 14m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/78f172670ef34b39b3fbb595755840d9 : SUCCESS in 29m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/602c5c20dde14184bb5b503ac33030f1 : SUCCESS in 13m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/191084cbd351426682be76376b1d83a7 : SUCCESS in 18m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3c86226ea6f8457e9e7e1c75206533f5 : SUCCESS in 8m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9de34840191a476e9304b582e21841aa : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4add683b6ed14ce69c413022736e5128 : SUCCESS in 11m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7de4fe1df8a84c9b948faa31d26cb9ac : SUCCESS in 1h 08m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4903833944d49a9ab4ecc1161bfcc0f : SUCCESS in 1h 18m 25s","accounts_in_message":[],"_revision_number":4},{"id":"b4fcd2ac3d9cfb29f98d9db7bfa21477829bfeb7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-10 18:58:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"faaef3b5b2d66005e50202a118aefcff44c92398","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-10 19:08:47.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0c849a7e4175cbe8cd879dbf36f89c35b6b3af88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 19:09:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e599cf9d46cc0d6f23b5e62e3f562f9f40d335d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 20:50:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"714c1115ac5813d07cb708face116ae32f440726","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 20:50:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44076935890744e0b41db4569aad776b : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41cfe75216c34b9b97da870de869bfce : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e867b5f2553440d8ad7dc02543800830 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4688ee520c34b7887e4380369654446 : SUCCESS in 8m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c744ff62569d4822a802758471967184 : SUCCESS in 7m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/480c78d74f7c43a79ec10b743220e9ea : SUCCESS in 7m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9aee1174893e4fb4a9d4a2a88e84e901 : SUCCESS in 12m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/075dafd9e8ad43b5927bc2fbb902170c : SUCCESS in 11m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/78612689a8db4fa0830bdfb187f133d9 : SUCCESS in 9m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b3f837aae2f54addb06a0a3df37d7459 : SUCCESS in 8m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bda7bb9cc0f3427e96376821eef4f3ca : SUCCESS in 12m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc21b373d324493fb78c7909547b92dc : SUCCESS in 12m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9564ecc59be146f8a364fb32c06d6e71 : SUCCESS in 37m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/09425df46c8c4a6f993e735efadddd77 : SUCCESS in 25m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d8bbf0a6367e4b1bad2a1017d6f82804 : SUCCESS in 15m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d5b60fe20e37401d90529786c07ab93c : SUCCESS in 20m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d4d4b10b27e8437d867b915d5f9349cd : SUCCESS in 7m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f8caafa32ec438d95ba103da679f284 : SUCCESS in 8m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/74d36cf1b83e4f1c952588a66982a842 : SUCCESS in 12m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bd5ec25c044446a2b95655f8864e0c9b : SUCCESS in 15m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5c28d3f674f47928a881647a2135b1d : SUCCESS in 52m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b18f87b834d47cd8a1e5c87e72f9f24 : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":4},{"id":"a6b4b82ccc3eeb188bafaa5ebc488239afad2ac3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 20:52:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eedf9b29be29435380fbbe9c208a8a49 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"78205a1ee9369719f13171433d05757fd0170d42","revisions":{"a0352511a43b437de268dd00e48a8ee2fc976122":{"kind":"REWORK","_number":1,"created":"2020-06-09 06:25:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/25/734425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/25/734425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/734425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/734425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/734425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/25/734425/1"}}},"commit":{"parents":[{"commit":"a1911d25a572794d60294587e716fc09b66ce7d9","subject":"Remove shade from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1911d25a572794d60294587e716fc09b66ce7d9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-09 06:25:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-09 06:25:47.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I03d4e2cb87246fe867092dd4d95b2983b824b57a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0352511a43b437de268dd00e48a8ee2fc976122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0352511a43b437de268dd00e48a8ee2fc976122"}]},"branch":"refs/heads/master"},"7b8cec9a5525a1804e4f6b3d040409dfbfd9b28d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-09 22:53:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/25/734425/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/25/734425/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/734425/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/734425/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/734425/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/25/734425/2"}}},"commit":{"parents":[{"commit":"7b639f3a252348d1982e87d0e919cee97b6e0329","subject":"Merge \"CI: add requirements-tox-py3x-check-uc job on AArch64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b639f3a252348d1982e87d0e919cee97b6e0329"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-09 06:25:47.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-09 22:53:03.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I03d4e2cb87246fe867092dd4d95b2983b824b57a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b8cec9a5525a1804e4f6b3d040409dfbfd9b28d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7b8cec9a5525a1804e4f6b3d040409dfbfd9b28d"}]},"branch":"refs/heads/master"},"530f316432870f8e7eb0b9bcd310e99d09d62d25":{"kind":"REWORK","_number":3,"created":"2020-06-10 03:44:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/25/734425/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/25/734425/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/734425/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/734425/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/734425/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/25/734425/3"}}},"commit":{"parents":[{"commit":"1b6e378b4644eae14d130ffc83df99d384a89f42","subject":"Merge \"Remove grenade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b6e378b4644eae14d130ffc83df99d384a89f42"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-10 03:44:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-10 03:44:51.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I03d4e2cb87246fe867092dd4d95b2983b824b57a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/530f316432870f8e7eb0b9bcd310e99d09d62d25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/530f316432870f8e7eb0b9bcd310e99d09d62d25"}]},"branch":"refs/heads/master"},"78205a1ee9369719f13171433d05757fd0170d42":{"kind":"REWORK","_number":4,"created":"2020-06-10 15:59:35.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/25/734425/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/25/734425/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/734425/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/734425/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/734425/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/25/734425/4"}}},"commit":{"parents":[{"commit":"1b6e378b4644eae14d130ffc83df99d384a89f42","subject":"Merge \"Remove grenade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b6e378b4644eae14d130ffc83df99d384a89f42"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-10 03:44:51.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-06-10 15:59:18.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back mock, msgpack and docutils\n\nChange-Id: I03d4e2cb87246fe867092dd4d95b2983b824b57a\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78205a1ee9369719f13171433d05757fd0170d42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78205a1ee9369719f13171433d05757fd0170d42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
