)]}'
{"id":"openstack%2Frequirements~734463","triplet_id":"openstack%2Frequirements~master~Ie7ba23f845bc5de2c22da21c54becf6c5a0be0a5","project":"openstack/requirements","branch":"master","topic":"cross-ironic","hashtags":[],"change_id":"Ie7ba23f845bc5de2c22da21c54becf6c5a0be0a5","subject":"Add an ironic cross-job with py38","status":"MERGED","created":"2020-06-09 08:57:44.000000000","updated":"2020-06-25 00:35:15.000000000","submitted":"2020-06-25 00:31:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734463-1593045104206-48842415","meta_rev_id":"17f796dfec9000e4a9b254f05932ee5739e8da9a","_number":734463,"virtual_id_number":734463,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 00:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-24 20:21:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-23 00:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-24 20:21:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 15:53:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-23 00:59:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-24 20:21:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-25 00:31:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6c90393ed7e2ec328fa6b7b95bb7d7a3b20249b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-09 08:57:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"968615e84b26b4aeb36b14087383afc6a5bfadc2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-09 12:15:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5f4206597a235060f46f73b580a185838139fa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 12:20:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a63d2efc0154251b5b886771d0d16c8 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f29f1f32669644b19b2f51a494516dc4 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc4cf3d7e4194ddba410a9940f035256 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/668adca014c94148abe9f53a8ca93ca4 : SUCCESS in 8m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/280d4201c22741858a12e8440cb3cb79 : SUCCESS in 4m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9fc4c2947aed4a6ba6a586c7d923b378 : SUCCESS in 10m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e46b66a878a64c38889308dc4826bde6 : SUCCESS in 7m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/116b23a91b48462db10cb267fa327cc6 : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/231e5c9bffc743fcaf98382ce8e47d77 : SUCCESS in 8m 30s\n- cross-ironic-py36 https://zuul.opendev.org/t/openstack/build/b8f4ec07ac3f4134a80b79638ffb5a87 : SUCCESS in 12m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/68f3fe2e705a48aca1c85f7d82222eff : SUCCESS in 17m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/691e54fa73d448999c48abd8efac1b79 : SUCCESS in 30m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3baa810c822e4af8b476185be9c24879 : SUCCESS in 18m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54424240ce01454fb2391407e36a7d37 : SUCCESS in 21m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/85359e23fc1c4a5f8364427a455bf098 : SUCCESS in 8m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/78807442b2c046f7bacc6cb9ce35dc65 : SUCCESS in 8m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cf31b734d1964d5889c56c75482a3450 : SUCCESS in 14m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc757119863a4cd9a684de6840a5489f : SUCCESS in 58m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ae0026ccab746bea166b67476284fc6 : SUCCESS in 1h 12m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8e27f24bc764cce82364eff6a93e137 : SUCCESS in 1h 26m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/fc31b72a4c384de5b69b6f2704f0cfe0 : SUCCESS in 1h 17m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1517f9acdd1bde534c08458d5939bc91251b969f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-09 16:30:43.000000000","message":"Patch Set 1:\n\nis there a py38 version?  I\u0027d like to move to those when we can.","accounts_in_message":[],"_revision_number":1},{"id":"3c3d74b398b492eebf576e21806ca1681b26b170","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-10 08:40:50.000000000","message":"Patch Set 1:\n\n\u003e is there a py38 version?  I\u0027d like to move to those when we can.\n\nYes, there is, I don\u0027t mind having Python 3.8 here, just let me know and I will change it to py38","accounts_in_message":[],"_revision_number":1},{"id":"3d2e2bb348358c3bf2ee80b7cfb7284f8a70dc93","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-10 15:57:28.000000000","message":"Patch Set 1:\n\nplease do (change this to py38)","accounts_in_message":[],"_revision_number":1},{"id":"c514bb82ad5f2a3ae1834618004c9c96406c3299","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-11 07:34:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5a878e78c54e10fddf1a6417123a93d838e6da88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 07:35:09.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/requirements on branch master.  The error was:\n\n  Job cross-ironic-py36 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python3-victoria-jobs\n      check:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-bindep:\n              voting: false\n          - requirements-tox-py36-check-uc\n          - requirements-tox-py38-check-uc\n          - requirements-tox-validate-projects:\n              voting: false\n          - cross-cinder-py36\n          - cross-glance-py36\n          - cross-horizon-py36\n          - cross-horizon-npm\n          - cross-ironic-py36\n          - cross-keystone-py36\n          - cross-neutron-py36\n          - cross-nova-py36\n          - cross-nova-functional\n          - cross-octavia-py36\n          - cross-osc-tox-docs\n          - cross-swift-py36\n          - openstack-tox-py36\n          - openstacksdk-functional-devstack\n          - tempest-full-py3:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      check-arm64:\n        jobs:\n          - requirements-tox-py38-check-uc-aarch64:\n              files:\n                - bindep.txt\n                - upper-constraints.txt\n          - build-wheel-cache-centos-8-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal-arm64:\n              files:\n                - bindep.txt\n      gate:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-py36-check-uc\n          - requirements-tox-py38-check-uc\n          - cross-cinder-py36\n          - cross-glance-py36\n          - cross-horizon-py36\n          - cross-horizon-npm\n          - cross-ironic-py36\n          - cross-keystone-py36\n          - cross-neutron-py36\n          - cross-nova-py36:\n              files: ^upper-constraints.txt$\n          - cross-nova-functional\n          - cross-octavia-py36\n          - cross-osc-tox-docs\n          - cross-swift-py27\n          - cross-swift-py36\n          - openstack-tox-py36\n          - openstacksdk-functional-devstack\n          - tempest-full-py3:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      experimental:\n        jobs:\n          - tempest-pg-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - tempest-full-py3-opensuse15:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n\n  in \"openstack/requirements/.zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"8f6724979b4147bf7e80ad8adf3e09d47ef06aae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 07:35:17.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/requirements on branch master.  The error was:\n\n  Job cross-ironic-py36 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python3-victoria-jobs\n      check:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-bindep:\n              voting: false\n          - requirements-tox-py36-check-uc\n          - requirements-tox-py38-check-uc\n          - requirements-tox-validate-projects:\n              voting: false\n          - cross-cinder-py36\n          - cross-glance-py36\n          - cross-horizon-py36\n          - cross-horizon-npm\n          - cross-ironic-py36\n          - cross-keystone-py36\n          - cross-neutron-py36\n          - cross-nova-py36\n          - cross-nova-functional\n          - cross-octavia-py36\n          - cross-osc-tox-docs\n          - cross-swift-py36\n          - openstack-tox-py36\n          - openstacksdk-functional-devstack\n          - tempest-full-py3:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      check-arm64:\n        jobs:\n          - requirements-tox-py38-check-uc-aarch64:\n              files:\n                - bindep.txt\n                - upper-constraints.txt\n          - build-wheel-cache-centos-8-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal-arm64:\n              files:\n                - bindep.txt\n      gate:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-py36-check-uc\n          - requirements-tox-py38-check-uc\n          - cross-cinder-py36\n          - cross-glance-py36\n          - cross-horizon-py36\n          - cross-horizon-npm\n          - cross-ironic-py36\n          - cross-keystone-py36\n          - cross-neutron-py36\n          - cross-nova-py36:\n              files: ^upper-constraints.txt$\n          - cross-nova-functional\n          - cross-octavia-py36\n          - cross-osc-tox-docs\n          - cross-swift-py27\n          - cross-swift-py36\n          - openstack-tox-py36\n          - openstacksdk-functional-devstack\n          - tempest-full-py3:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      experimental:\n        jobs:\n          - tempest-pg-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n          - tempest-full-py3-opensuse15:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n\n  in \"openstack/requirements/.zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"40fd03066ead97af8d973cfcd9864518431bc2e3","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-11 07:36:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5083efe2c664403749347bd1da6ad3fc99e671af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 09:04:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85cf5ae923bb4b0ab133146fff325019 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7091a6bac1774583800aee0e30819b00 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3e3a64ee98541b0b21bbaf177919a5d : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/257b217c7b244e95bab52409d510fa6d : SUCCESS in 9m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2eb6b28e727344d4a300dbd99c9d9a09 : SUCCESS in 4m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bc105cb8c7ca4771811112de442f2f78 : SUCCESS in 13m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d5dc270647044635872ec2c67dabcecb : SUCCESS in 5m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/57192d360b494195a91aad55fbfaeaee : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b1fbcaa57c2e4c0aa1c62d178cb5f376 : SUCCESS in 10m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/99517bd0b35b41c9ae15c26f9f711124 : FAILURE in 5m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f83f2042b7cf4fb5813fab114ba1515f : SUCCESS in 48m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3a49548b6f9149469d99dd191e19bc63 : SUCCESS in 23m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a438988f1d52453ab8299b36c836d895 : SUCCESS in 22m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d8d2bb26077849319f71e04e28d32043 : SUCCESS in 16m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7f93607f8fa745428dc1d9ed328f8869 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/681df2955b4246aba884982a84df35ff : SUCCESS in 7m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6c0ea5f58a364daa8862d1719c565019 : SUCCESS in 14m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7366550573c47399cd0877ffeb9a886 : SUCCESS in 59m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ba1368417e54f529dddee720cf9ee7f : SUCCESS in 1h 27m 00s","accounts_in_message":[],"_revision_number":3},{"id":"4c403eb7b7fb5db558dc480b5cc14cd5422f82ad","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-11 11:40:27.000000000","message":"Patch Set 3:\n\nyeah... little thing I haven\u0027t considered, ubuntu bionic doesn\u0027t have Python 3.8 installed by default :/","accounts_in_message":[],"_revision_number":3},{"id":"f53a6e8629e7e97929165c45010b89a815ce3e4e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-11 12:11:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"984bbf26fd426315b433ca418e610151b5fcf972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 13:45:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78f953fed1cb4f0ea07b9261f55c78e1 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfff04a1a946405098cf6fcbb00670b6 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/629d50cf295446e9963c3f2462ea98c0 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5087f4c9c9545e1ba92d05c49c35889 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3feb10942ab24f0ba754a13f69777750 : SUCCESS in 4m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/53978594e9dc4ac4be740f98c9be04e3 : SUCCESS in 10m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/89cf1a65883c4676a8849fd259886d34 : SUCCESS in 5m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3d907700c07d4e259e0533f7d847ad45 : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af088cfef914450ca81e341682b4c84d : SUCCESS in 8m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1508434e04e246daac2337dde2f8f798 : RETRY_LIMIT in 3m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8d15a23566224fc0afc39ef0dc101fc9 : SUCCESS in 13m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ffc34a33d81c43768b28e1b012a1c15c : SUCCESS in 26m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/973d7b3a6ec444eda5b73092152b574d : SUCCESS in 12m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb6b55f40a0a44019becd72fcb9747a8 : SUCCESS in 20m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dd096a4bbce24e94a389f5043042608f : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d8f6e5d11d442929e12493d72616780 : SUCCESS in 7m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f676097924e745bf8f81edf2b2c18420 : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/821919f501644e2aab122edf3d962a59 : SUCCESS in 1h 01m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0e5998bbf314928af493ead681de0a5 : SUCCESS in 1h 27m 31s","accounts_in_message":[],"_revision_number":4},{"id":"a464729ed079b1677525e670d31b794a685d2f0c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-11 15:18:21.000000000","message":"Patch Set 4:\n\nlooks like a syntax error in a zuul script, I submitted a patch to fix that https://review.opendev.org/735174","accounts_in_message":[],"_revision_number":4},{"id":"58969d483d3abc1a270c5d4ea9bf623c42f91b1d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-16 15:04:45.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e3cc42ab86c3770ee557e70bb34c838f0b522989","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 19:31:05.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c4a087d230c48baad5256e067abac73 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b159051e5ba4f6a82e899bbf4d1a68b : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/257156d9a0a94a6a8cf93d7e595c4f3e : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c0c3284157544f1b5577765614f27a0 : SUCCESS in 7m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aad03096b51c49bca0d614e5533ccafd : SUCCESS in 7m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4d15b7f7e6e44f10a0194c1aaba7fde8 : SUCCESS in 9m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/69d517db2c3f443c950ad7ce77428a64 : SUCCESS in 8m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/209f46710dab4f51ad5b9782d1292c9f : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/20eb8af1efc54dc8b1b6f2a18976f525 : SUCCESS in 14m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/291abcbebfa64dce9707b5a9c8f4e7cd : RETRY_LIMIT in 7m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5ef25365166d4139a9214b0a46cd82db : SUCCESS in 26m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/edeb6b503df049ffac4c5cacc07b4383 : SUCCESS in 18m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/22f8b2dfd2a74278a7591aed61c45e69 : SUCCESS in 17m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78a30d57752448c58f1083aff20b8cf9 : SUCCESS in 15m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/af7e855c736c4ea1bcdc70d5445eb7e5 : SUCCESS in 8m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d482d36bf268456eb8236495e655d212 : SUCCESS in 8m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0498578683034244a285320932aa75d9 : SUCCESS in 16m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34d7f96c3beb4e2fa1f92f9967d57527 : SUCCESS in 41m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a062ecf0008641638282ef923ed0f7c6 : SUCCESS in 1h 23m 41s","accounts_in_message":[],"_revision_number":4},{"id":"6791bde23b24e6aa0c6b3c4450128d344a81a39e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-17 09:43:23.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fc403047c674d86d760b84c2199a1e937f156ebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 11:26:49.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d073c4699974d98afb5646f17a06f03 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f5ad27478454120943ab15c3d8f1872 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4648dcfd7ee4598951073cea5bf087a : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f824861b845446dabfd297e3e47bb4f7 : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c78c419f5554875ad089ecbc4534c69 : SUCCESS in 8m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2c34655c747042b3bfcac9b4031d3e36 : SUCCESS in 10m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/16dd11536b4d47909744b65a9419a60d : SUCCESS in 5m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ead7ba688e74416a93638cdf708b14a2 : SUCCESS in 15m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5604af8edada4a8db6800225da541b9d : RETRY_LIMIT in 5m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f338a7b77715405fad75d32b12e80b08 : RETRY_LIMIT in 6m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/989f203dccac4e50abd904c5dcd5730b : SUCCESS in 35m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/244361dbe78f46c48075bee3e0a7daf2 : SUCCESS in 29m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5c2041a048f84cb5816fc7891cd1dab5 : SUCCESS in 13m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b232148b263349ea981ef561c8ff3116 : SUCCESS in 22m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/720bb752afd84b93a955b1baa0a5ee1f : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32e0d66a350c492b842571386907af15 : SUCCESS in 7m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fa7a01567ead4a98a43bcf49bfb859b5 : SUCCESS in 15m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5eb5df5b23164f45aec35e7441ef1210 : SUCCESS in 1h 12m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da648baa59eb4341b82ad9abdb478599 : SUCCESS in 1h 30m 26s","accounts_in_message":[],"_revision_number":4},{"id":"4de1432ba7ec2f667257ca54b8b270ae89817064","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-19 12:41:59.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fef978827c72d3cfe9ed40089a95213f71be7ea0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 13:59:37.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc902ffa04f84becbc9077ebe5286143 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc45c0fc44684df0b383152a5afe3060 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8df078c26ade420688dd78727484b28b : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5713b9852605486e8ec354522772bade : SUCCESS in 7m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ebe195c6f904109a8a54c9b37f80214 : SUCCESS in 7m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bdaffcf790cb4788ad15029207acd47a : FAILURE in 10m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/65d2ee37695a49c18dfbc1ab6ec41852 : SUCCESS in 10m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0f7b4dd530b04db4af23770adb28bc2b : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/afd40887d0474bebb584c7ec6152ac27 : RETRY_LIMIT in 6m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7cd59e9c382e404786481a6392a09237 : RETRY_LIMIT in 4m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3b72dcb32a104be180c7b40d3db2d1ef : SUCCESS in 16m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c49bf970b17548178b385bf90ce01e4c : SUCCESS in 31m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f71777630fa14bb39015d0f332b29998 : SUCCESS in 15m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8182d1fea4784268843075491a88e197 : SUCCESS in 19m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/16b72f654032467f93e7dbca2f6f0696 : SUCCESS in 9m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6ce287e9c464bb1ab49d4f8ecac95f5 : SUCCESS in 10m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ba7de9f67ac7466181d45d60a0679b61 : SUCCESS in 14m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c86f9613bdd84cc68bd273ef79895332 : SUCCESS in 50m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f39a2c42e9954191b1b1922bdbd6748a : SUCCESS in 1h 15m 27s","accounts_in_message":[],"_revision_number":4},{"id":"86c0421d202a16fb67db8e215f585b7043431924","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-22 07:24:16.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"7eb2fe45a6b00213860f53dab0a00d7f0672eb75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 09:02:41.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b6ad97a2304bd4b5aba589798ccd8a : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ed2d404566b41eeb10d913c201cdb27 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/681ae80f343742649632ecd58130c4e7 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1079191d7ad249afb89bdf8d9a3b2682 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5eb60b8313de4b6fbe0ca740bd207bbf : SUCCESS in 5m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/15a34857e60a4f858290a58a707c1a1d : SUCCESS in 8m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/39260cad0d8144c493406aadcf439136 : SUCCESS in 6m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/759d495939364dfaae15831e1656d5fd : SUCCESS in 11m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b25cea523bb64ddeac65c70e38113963 : RETRY_LIMIT in 4m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cd2a85f103c7401e8f24cd7e0e73e136 : RETRY_LIMIT in 4m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/994cbb92b9cd4b658dcf2787b261c96c : SUCCESS in 16m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e8f2e4bedddf4cbf93251a733f1a6293 : SUCCESS in 27m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4cd9def53d4a46b4b23c99d3074f4446 : SUCCESS in 15m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2727ebe6ab3a4bfc85113a05fa6d5cfe : SUCCESS in 20m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/965e4a9eb95b4b309068da9363cb1660 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/03b750a9924140a98132d9fa43bef543 : SUCCESS in 8m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7948e3084a98488ea96f1e0485298ce6 : SUCCESS in 10m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7abf1cf5493241769d74c43e6bb3e571 : SUCCESS in 58m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/edcf422fe8fd469980f547f492146ce2 : SUCCESS in 1h 32m 28s","accounts_in_message":[],"_revision_number":4},{"id":"2fc08bd2daee35fe8783877d3374446ccff451eb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-22 15:53:22.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"466c7c89ed63327fedf6e0160a98b50c1e9b1f71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 19:13:10.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f01262eb42d46be89be38dc9ac69280 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2f6e908c05a40a980a55b40121e530e : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16e5aa4a87c74ef4b41eb4573e15561d : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/464c880f52c741f3b7f061918cc16154 : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aae821e5a4854ba2b6d662cf05dc8c63 : SUCCESS in 3m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dab5dc07c78b4d54a610df90b0a8b52a : SUCCESS in 17m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0b140dff60d14a43b84491da6ab43996 : SUCCESS in 12m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a56ad4453a274929ba7002d92e11f3a7 : SUCCESS in 12m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/91519dd380934013bdc8ed0cffc1cfb9 : RETRY_LIMIT in 11m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7218ec79c364470f8840a417aa366349 : RETRY_LIMIT in 9m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bd241a405b7c401f99b255df112e61c9 : SUCCESS in 13m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/77925dd93aa1493aadbfd5ec82b4b7a9 : SUCCESS in 22m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d3b1389307984b7ab6ad326b9db9416e : SUCCESS in 17m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73fcf3a9634b425ebb509c2fa6e24d87 : SUCCESS in 30m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cdf521dc2a1443a9a4c1c85f5b0acab6 : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fa24fb810cd74080b67e9553b51e49f0 : SUCCESS in 7m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/07257f241bdb4b8d99c6cdd5de8f299c : SUCCESS in 15m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fee11c38dd2d4c7da93d30d91f5f8b52 : SUCCESS in 59m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/399a2a09cf564c009fa44fe870888518 : SUCCESS in 1h 13m 35s","accounts_in_message":[],"_revision_number":4},{"id":"7f536f62bb305426736e6ab5f9a1226f5fbb4012","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 19:53:05.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"b9a049e28ca3014d19f2f5ce69f45e6fe6e98290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 21:14:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/208f8bec3b284ac0a42be492bcb86d9c : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab99add6a564f8a97bb25264b81d9ad : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55511f812cb64f08ba2c122479eb1ddc : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7e546060e294c61b82f0a556c481625 : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/06ea7bfe05d044e282f5a6cd49e7fca9 : SUCCESS in 3m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d7a766ce1d5d40668e9162d7a73f2058 : SUCCESS in 9m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/52fe9b93bd1b42628b5881a31d135bd0 : SUCCESS in 5m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8e8ffebd28ea47159040a73b5b65606b : SUCCESS in 11m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d356fd277b6c453eb62fd59fe633be07 : RETRY_LIMIT in 9m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8200f1b8eead41e2a0df535c74fef337 : SUCCESS in 7m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/365d8abc4d3b4340acf76fda39c48d0b : SUCCESS in 30m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e08d5ecd5f7648baba3ae7b7d0c85137 : SUCCESS in 28m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/363c28997f8a4b48bc0cf7bbf7b6547e : SUCCESS in 21m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/09c143b7380a424997879960582dcb1e : SUCCESS in 23m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7bc9ea580754456aac5558f202abc9f7 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2fb7fbb3e7ca45c2ace73de5fab5e410 : SUCCESS in 11m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7ff73c3c57554f1996fa691d5d4ec8b7 : SUCCESS in 11m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/000e60b50d85461b969011bc9e931f7f : SUCCESS in 1h 04m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b606fb4123144ea9d250e5eebe73ba3 : SUCCESS in 1h 16m 16s","accounts_in_message":[],"_revision_number":5},{"id":"c77d08d22412c021585c85ae6c4f665f726c2575","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 21:20:57.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"3f9a487b4a5de6ddb24f7fb4cb164a2169e6b5fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 22:49:58.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90b0fbc6e4e84db9b4f85ee928a23fd9 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f692fe4b4df450ba21b58a2b80d4ef6 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bc312410bb944159905914e2793ab6c : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a927c4c02e534982950e81ca71fdfe03 : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8322827bb3b450087b1fa7fda1262f0 : SUCCESS in 5m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e1731921b7324ac3aa4f2591148cc5c7 : SUCCESS in 13m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/03fc7d8dc92e402fb1d9d8a7bdf4c695 : SUCCESS in 8m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2cc5398cde14499b909e32ca70f5e1e1 : SUCCESS in 12m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc74b6d81f18453c8e3ed09c223970c8 : FAILURE in 6m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3c8f87b66d914383b6a0970356b30f96 : SUCCESS in 12m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/28f615450f874202811b7bea92542c61 : SUCCESS in 27m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5a1aede238cd4ee182c8d11b6e5e44e6 : SUCCESS in 36m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6df7c8120922484e9ceb56dbf2e5bd66 : SUCCESS in 12m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dd47dda83e7c4fa48921edee566c4b9e : SUCCESS in 18m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3bbdf2adf1744dbaa4860c333e7356f4 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37b91a585f37489fb819ecdf3959a151 : SUCCESS in 6m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b59f34eb4c3a4dbe9d5c730e1dd361dd : SUCCESS in 11m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d68e32df9d2341d29807f21c5aedc107 : SUCCESS in 59m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3b91cb3e31d4c5e98633246e8ec25bf : SUCCESS in 1h 23m 32s","accounts_in_message":[],"_revision_number":6},{"id":"10a82548a4bd2946677778264a95c433c0c0af41","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 23:28:30.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"f9d14943df4a1287383ef10c734b97db8a580f43","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 23:28:58.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"b6ff127e5be471c1f5d4e2b052e8678353df134c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 00:55:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52b77edbb1b04283bd4d9a00075b46f1 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de377534be0a4fe2b0e1c6002e3cc018 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e789ac14c8164809bb6c6ece734d12c8 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9986f09c758c4127a9fb5c2dbb723707 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ecd2c1631ac45acafd436547e61ff33 : SUCCESS in 3m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5c5321122fdb4304aac0a1e63686477a : SUCCESS in 8m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dcbbd9d5de454e5a97acca5f83188175 : SUCCESS in 5m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5daee0d731ed480682a6f253df727fa6 : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9167b935c84b4eb6bf376de4475f3095 : SUCCESS in 6m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/adf42b578b1e4900aaf94914c4aa5772 : SUCCESS in 8m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aff96d7dbfbf43cca0643ad375d79323 : SUCCESS in 13m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c1ab365b87f0449c86d27f97bc529b31 : SUCCESS in 26m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7cb4ea238dae4e8db406e55851b607ee : SUCCESS in 13m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8c0030b4a2d466a9c3a767300fb89ca : SUCCESS in 20m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1f9f4f4b4e6b49d1ab7748ed4d0441f3 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/45f488571a184a4d9f2123063783683f : SUCCESS in 7m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3f6b3f5297d84f939015de97d6420b16 : SUCCESS in 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1bfaa831f11748159b742259e312dc67 : SUCCESS in 56m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e27bd510558446bbaded3b9b3bb212c5 : SUCCESS in 1h 18m 41s","accounts_in_message":[],"_revision_number":8},{"id":"2d96f48bc1e3b9aa7fa7e8953e703989e6cfdd2a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-23 00:59:32.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"9e268901a0f3089cdaf03fdbcae923d8e7a566aa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 20:21:22.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nLocal change for horizon job and ironic repo changes have merged. Should be good to go.","accounts_in_message":[],"_revision_number":8},{"id":"3c98c0568f69e547ab6208133520ec4e05bf39a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 20:21:46.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"807fdbee9b607aa1edb9583e19ce7f6865d9f59e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:31:43.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c84cafd55d0841259402963a672a6439 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ed96136b1eb4516a60e8eab4bf11a65 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5d7b4b8f659499eb47d53d35c04ee68 : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dc63f1ce7664811a1a2066cc56b120f : SUCCESS in 8m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47f07321468d4db9b1fb03d0a8a3e5ef : SUCCESS in 8m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/daf5b9870fb047a0b23e2b9ae7bce11b : SUCCESS in 13m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/40bee5cf2f2b40b2a40cf7834df83a0d : SUCCESS in 9m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3a581ccf28aa4e1697519a3f6620b3de : SUCCESS in 14m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e269cf760112437cba91a0071e07f8ce : SUCCESS in 13m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0ccd518687b14db8b5b746de2c258687 : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3a415765fa9440f8816fe04533e967a7 : SUCCESS in 25m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9880e038840f48bcbd700b67a6ba247e : SUCCESS in 27m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73b0b93796334a3682e4fee65889c81c : SUCCESS in 24m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5044b0bf235444a1abf96fac4387bf0d : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/283f56d6212941e2a91e5710341e0de2 : SUCCESS in 12m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7869ee1c88274fed9a165dc8218625a3 : SUCCESS in 21m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e2960c90a0fe484e872e787752004b56 : SUCCESS in 23m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/913f5d19a5b544ba95b95a59b759464d : SUCCESS in 1h 00m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b57811b02e8c460983d67eb1ec34b134 : SUCCESS in 1h 36m 00s","accounts_in_message":[],"_revision_number":8},{"id":"9d757807207e72dec1b22565db0d2eb95c3c4272","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:31:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"59c486c249a918d637efca8b8c5d431ae7912dd7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:35:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb2a745299b424497a78966eecb7bfd : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"32648949a48b8d84dbd032db7bd2726e4cc0c726","revisions":{"23002898e874960749349def878067716486ba31":{"kind":"REWORK","_number":1,"created":"2020-06-09 08:57:44.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/63/734463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/734463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/734463/1"}}},"commit":{"parents":[{"commit":"a1911d25a572794d60294587e716fc09b66ce7d9","subject":"Remove shade from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1911d25a572794d60294587e716fc09b66ce7d9"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-09 08:38:19.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-09 08:40:04.000000000","tz":120},"subject":"Add an ironic cross-job","message":"Add an ironic cross-job\n\nThis patch adds an ironic cross-job to help verify requirements\nchanges.\n\nChange-Id: Ie7ba23f845bc5de2c22da21c54becf6c5a0be0a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23002898e874960749349def878067716486ba31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23002898e874960749349def878067716486ba31"}]},"branch":"refs/heads/master"},"1324eb313f945303f44d74f8d1ff3f1950615dd1":{"kind":"REWORK","_number":2,"created":"2020-06-11 07:34:53.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/63/734463/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/734463/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/734463/2"}}},"commit":{"parents":[{"commit":"a1911d25a572794d60294587e716fc09b66ce7d9","subject":"Remove shade from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1911d25a572794d60294587e716fc09b66ce7d9"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-09 08:38:19.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-11 07:34:12.000000000","tz":120},"subject":"Add an ironic cross-job with py38","message":"Add an ironic cross-job with py38\n\nThis patch adds an ironic cross-job to help verify requirements\nchanges.\n\nChange-Id: Ie7ba23f845bc5de2c22da21c54becf6c5a0be0a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1324eb313f945303f44d74f8d1ff3f1950615dd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1324eb313f945303f44d74f8d1ff3f1950615dd1"}]},"branch":"refs/heads/master"},"03e937241ee8c74454d86b5d2d2aa31e0545469b":{"kind":"REWORK","_number":3,"created":"2020-06-11 07:36:08.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/63/734463/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/734463/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/734463/3"}}},"commit":{"parents":[{"commit":"a1911d25a572794d60294587e716fc09b66ce7d9","subject":"Remove shade from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1911d25a572794d60294587e716fc09b66ce7d9"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-09 08:38:19.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-11 07:35:55.000000000","tz":120},"subject":"Add an ironic cross-job with py38","message":"Add an ironic cross-job with py38\n\nThis patch adds an ironic cross-job to help verify requirements\nchanges.\n\nChange-Id: Ie7ba23f845bc5de2c22da21c54becf6c5a0be0a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03e937241ee8c74454d86b5d2d2aa31e0545469b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03e937241ee8c74454d86b5d2d2aa31e0545469b"}]},"branch":"refs/heads/master"},"2ba724b7c9d211ee0295d2cbea1ccf5d8de7b39d":{"kind":"REWORK","_number":4,"created":"2020-06-11 12:11:12.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/63/734463/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/734463/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/734463/4"}}},"commit":{"parents":[{"commit":"96d16e8742120ac88babb77a05bba6aa225b9310","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96d16e8742120ac88babb77a05bba6aa225b9310"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-09 08:38:19.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-11 12:10:33.000000000","tz":120},"subject":"Add an ironic cross-job with py38","message":"Add an ironic cross-job with py38\n\nThis patch adds an ironic cross-job to help verify requirements\nchanges.\nWe need to run that on Ubuntu Focal to have Python 3.8 installed\nas default interpreter.\n\nChange-Id: Ie7ba23f845bc5de2c22da21c54becf6c5a0be0a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ba724b7c9d211ee0295d2cbea1ccf5d8de7b39d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ba724b7c9d211ee0295d2cbea1ccf5d8de7b39d"}]},"branch":"refs/heads/master"},"9642574675a56a9e8af43f142e1c828136370ec3":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-06-22 19:53:05.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/63/734463/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/734463/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/734463/5"}}},"commit":{"parents":[{"commit":"96d16e8742120ac88babb77a05bba6aa225b9310","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96d16e8742120ac88babb77a05bba6aa225b9310"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-09 08:38:19.000000000","tz":120},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-22 19:52:34.000000000","tz":-300},"subject":"Add an ironic cross-job with py38","message":"Add an ironic cross-job with py38\n\nThis patch adds an ironic cross-job to help verify requirements\nchanges.\nWe need to run that on Ubuntu Focal to have Python 3.8 installed\nas default interpreter.\n\nDepends-on: https://review.opendev.org/737367\nDepends-on: https://review.opendev.org/737362\n\nChange-Id: Ie7ba23f845bc5de2c22da21c54becf6c5a0be0a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9642574675a56a9e8af43f142e1c828136370ec3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9642574675a56a9e8af43f142e1c828136370ec3"}]},"branch":"refs/heads/master"},"eb1c12514aaec4d22858636dbd282a63c9f480b0":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-06-22 21:20:57.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/63/734463/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/734463/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/734463/6"}}},"commit":{"parents":[{"commit":"d62b47954bf3b053daedc31aa429ccb9a02e41d7","subject":"Switch node tests to inherit nodejs-run-test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d62b47954bf3b053daedc31aa429ccb9a02e41d7"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-09 08:38:19.000000000","tz":120},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-22 21:20:57.000000000","tz":0},"subject":"Add an ironic cross-job with py38","message":"Add an ironic cross-job with py38\n\nThis patch adds an ironic cross-job to help verify requirements\nchanges.\nWe need to run that on Ubuntu Focal to have Python 3.8 installed\nas default interpreter.\n\nDepends-on: https://review.opendev.org/737367\nDepends-on: https://review.opendev.org/737362\n\nChange-Id: Ie7ba23f845bc5de2c22da21c54becf6c5a0be0a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb1c12514aaec4d22858636dbd282a63c9f480b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb1c12514aaec4d22858636dbd282a63c9f480b0"}]},"branch":"refs/heads/master"},"32efa7910ac5611b695318277701d17c8607e20d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-06-22 23:28:30.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/63/734463/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/734463/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/734463/7"}}},"commit":{"parents":[{"commit":"5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc","subject":"Switch node tests to inherit nodejs-run-test-browser.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-09 08:38:19.000000000","tz":120},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-22 23:28:30.000000000","tz":0},"subject":"Add an ironic cross-job with py38","message":"Add an ironic cross-job with py38\n\nThis patch adds an ironic cross-job to help verify requirements\nchanges.\nWe need to run that on Ubuntu Focal to have Python 3.8 installed\nas default interpreter.\n\nDepends-on: https://review.opendev.org/737367\nDepends-on: https://review.opendev.org/737362\n\nChange-Id: Ie7ba23f845bc5de2c22da21c54becf6c5a0be0a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32efa7910ac5611b695318277701d17c8607e20d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32efa7910ac5611b695318277701d17c8607e20d"}]},"branch":"refs/heads/master"},"32648949a48b8d84dbd032db7bd2726e4cc0c726":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2020-06-22 23:28:58.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/63/734463/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/734463/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/734463/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/734463/8"}}},"commit":{"parents":[{"commit":"5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc","subject":"Switch node tests to inherit nodejs-run-test-browser.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-09 08:38:19.000000000","tz":120},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-22 23:28:54.000000000","tz":0},"subject":"Add an ironic cross-job with py38","message":"Add an ironic cross-job with py38\n\nThis patch adds an ironic cross-job to help verify requirements\nchanges.\nWe need to run that on Ubuntu Focal to have Python 3.8 installed\nas default interpreter.\n\nDepends-on: https://review.opendev.org/737362\n\nChange-Id: Ie7ba23f845bc5de2c22da21c54becf6c5a0be0a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32648949a48b8d84dbd032db7bd2726e4cc0c726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32648949a48b8d84dbd032db7bd2726e4cc0c726"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
