)]}'
{"id":"openstack%2Frequirements~734629","triplet_id":"openstack%2Frequirements~master~Ic938198ab2c5e974fd7d55345723da2c066c63ec","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Ic938198ab2c5e974fd7d55345723da2c066c63ec","subject":"Remove cap on cmd2","status":"MERGED","created":"2020-06-09 15:38:07.000000000","updated":"2020-06-14 09:34:34.000000000","submitted":"2020-06-11 19:44:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734629-1591904697944-9cb1c9d8","meta_rev_id":"a7f4d523bd9a3ece768d612126655ed98d9353e2","_number":734629,"virtual_id_number":734629,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 19:44:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-10 00:45:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-11 18:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-11 18:20:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 00:45:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-11 18:20:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-11 19:44:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d787c316245c29bb89eaf2f6b52a8b5a50888e4","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-09 15:38:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3ff411884f916486c071ec96558a01050846bc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 17:30:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88df6afa4e884717a963464a5a67f58c : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db0ff5ceb5914b68a142d9908bf341f3 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eab98a4353bf4033b4cf6bddf77e3899 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a6ec8fe72a94be0b8fd5c1c5c2a917b : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95e415ae407b49348be234002aab3f20 : FAILURE in 5m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/309a8b0d9689478194f79ca78647cf90 : SUCCESS in 49m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2fb0fc0b40574f788f8fd173ca92debf : SUCCESS in 1h 00m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91eed7ba25524b55a651d586d7ff08fc : SUCCESS in 1h 21m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/834d69c1b458493d8274f8b487a137d1 : SUCCESS in 1h 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b97ec388b0d81e16c4c3b8681982ea32a8b98593","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-09 19:20:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b146996e8852208754596b7ee8272f0d6e2b63f5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-09 19:34:52.000000000","message":"Patch Set 2:\n\ncliff will also need a release","accounts_in_message":[],"_revision_number":2},{"id":"62a65d17fa497c22ee467ea1972f258341eed8c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 20:48:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d5351c15a24fdbb4ec415b864c1471 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/619f87f374244818b99abcb2c4e1f00b : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0589971ab5d444c792bcc3002053a2d7 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93a6e69296be4a23b1cc1e0082f06e8e : SUCCESS in 7m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3930ddb7f67e409a841cfc0c1e479820 : SUCCESS in 6m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/419c4d523b36469ba680d0b714c0b93d : SUCCESS in 5m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3ad2a0b2fc8640fbb6a1b734f192c4c6 : SUCCESS in 13m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7670fd13a868490db589cb4e23c4fabf : SUCCESS in 11m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7c38436b8dc6471b89f668436f76bb74 : SUCCESS in 13m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9430b5ffee634e9b8a1787295fc6125a : SUCCESS in 7m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/eb49a374a8ae464890b40e36f0b134c1 : SUCCESS in 12m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ee8838dac53c4247a59a0c67c0310665 : SUCCESS in 7m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c0759263bba140d59460f85d9ba2071f : SUCCESS in 16m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ee43633e5112436d9a47287a359ddab6 : SUCCESS in 26m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9100a1fb016a439b9f72311bf0ba85d9 : SUCCESS in 12m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c9ba3631a9c4134b14c9d8530c95ba0 : SUCCESS in 17m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/36b1b8b8f09948179d5bb60ef26e8db4 : SUCCESS in 8m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/65b00ba17f78471fb836917ea4c09437 : SUCCESS in 8m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/75a6e6eb841740dc9d8bdc69a1d584e5 : SUCCESS in 11m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8257cd91fa6649f3a663ebd68ecf028b : SUCCESS in 56m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9fb625bab214701b5f61b98b00937dd : SUCCESS in 1h 11m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dfdffee794b942b3a11cd260fd36edbf : SUCCESS in 1h 12m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/3f0db16fcfb044d1bed0b6d12f179f3a : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":2},{"id":"a84e516397e1a733a510c56d62ff7555fdd0100e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-10 00:45:49.000000000","message":"Patch Set 2:\n\nthough I guess it\u0027d be their master testing against our constraints, so lgtm","accounts_in_message":[],"_revision_number":2},{"id":"646577b68d225cbcbfb5161cbc057b56249d6abb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-10 00:45:53.000000000","message":"Patch Set 2: Code-Review+2\n\nthough I guess it\u0027d be their master testing against our constraints, so lgtm","accounts_in_message":[],"_revision_number":2},{"id":"b44079c716a4027ae86719c0c4eecd9435d297ac","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-11 18:20:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c1e113ef8a591bd6758516d9fd55edda93ba9660","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 18:20:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7a86f26b3e6b80e1a5a80119740fc8e669c6c745","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 19:44:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c49bb3d30104f2d9e8de2f01cb84c73 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c75152e6bb44b48e6e119ac01f57bf : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/614ab1cc61a74acebf34333f85049a04 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cea2b36ba04a43d38d53b5ad9f471fc9 : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa625b6e68ee401babd36f3b12a09e2e : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5ef275c93cde4a568248b6e7dc953830 : SUCCESS in 4m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ca7b4e0c023a4f26983aa9d860a639ce : SUCCESS in 8m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6e03b95eebd24ecc8dcb6c60cd409b68 : SUCCESS in 11m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a71114241d6548c180c746c5f24314b3 : SUCCESS in 14m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1256c2bee3a041e88a7d020d767d1fe3 : SUCCESS in 5m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d88c50fd97c543b1a60280d1d7510674 : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e4b5d4f0d94b46a197107200da45ec1e : SUCCESS in 15m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0725aaa821624642be204f0a01f3cbf0 : SUCCESS in 16m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dca28cb9d0a142b0960d14b86893375e : SUCCESS in 25m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cfdae4d5833242299bf4c529b948ee10 : SUCCESS in 20m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/183088cdd0974f6d95944b65e9372efe : SUCCESS in 18m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/be68817dc9454bfda4075a9572cb1dd1 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d2cc776910d64f308ee7c7f310d24495 : SUCCESS in 12m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2480e6623fe14624b31ab35c09b3b7eb : SUCCESS in 13m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3aadd3a372304bfd9b67ce8ccda3039e : SUCCESS in 13m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68195ccaf5ff4696b3e23ee4fce824fc : SUCCESS in 1h 04m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84c8890409d14834bf20287163f862ca : SUCCESS in 1h 23m 19s","accounts_in_message":[],"_revision_number":2},{"id":"d422a37648680ef164baf16b2a0d16d7d64f8d2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 19:44:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a3a8b9bd06c15c134098fff311e331508cac2793","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 19:48:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21ef95e9de0d4ba78d238ae2a2ba009a : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":2},{"id":"a7f4d523bd9a3ece768d612126655ed98d9353e2","author":{"_account_id":11692,"name":"Cédric OLLIVIER","email":"ollivier.cedric@gmail.com","username":"col"},"date":"2020-06-14 09:34:34.000000000","message":"Created a revert of this change as I0046c0c966fa5766113513b8577dd5a954e83a48","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b51ba9d7149307f484b02587b5d9e95c4c79fa9d","revisions":{"728296fd741f1dbafb602f6f9b99d5eb1076692b":{"kind":"REWORK","_number":1,"created":"2020-06-09 15:38:07.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/29/734629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/734629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/734629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/734629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/734629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/734629/1"}}},"commit":{"parents":[{"commit":"a1911d25a572794d60294587e716fc09b66ce7d9","subject":"Remove shade from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1911d25a572794d60294587e716fc09b66ce7d9"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-06-09 15:17:02.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-06-09 15:17:32.000000000","tz":-300},"subject":"Remove cap on cmd2","message":"Remove cap on cmd2\n\ncliff landed a patch fixing the parser for newer cmd2.\n\nDepends-On: https://review.opendev.org/712591\nChange-Id: Ic938198ab2c5e974fd7d55345723da2c066c63ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/728296fd741f1dbafb602f6f9b99d5eb1076692b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/728296fd741f1dbafb602f6f9b99d5eb1076692b"}]},"branch":"refs/heads/master"},"b51ba9d7149307f484b02587b5d9e95c4c79fa9d":{"kind":"REWORK","_number":2,"created":"2020-06-09 19:20:57.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/29/734629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/734629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/734629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/734629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/734629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/734629/2"}}},"commit":{"parents":[{"commit":"a1911d25a572794d60294587e716fc09b66ce7d9","subject":"Remove shade from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1911d25a572794d60294587e716fc09b66ce7d9"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-06-09 15:17:02.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-06-09 19:20:54.000000000","tz":-300},"subject":"Remove cap on cmd2","message":"Remove cap on cmd2\n\ncliff landed a patch fixing the parser for newer cmd2.\n\nDepends-On: https://review.opendev.org/712591\nChange-Id: Ic938198ab2c5e974fd7d55345723da2c066c63ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b51ba9d7149307f484b02587b5d9e95c4c79fa9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b51ba9d7149307f484b02587b5d9e95c4c79fa9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
