)]}'
{"id":"openstack%2Frequirements~734828","triplet_id":"openstack%2Frequirements~master~I8bf7446df2a46bbfa6a5e58f326cc9eb7b821ba2","project":"openstack/requirements","branch":"master","topic":"virtualbmc","hashtags":[],"change_id":"I8bf7446df2a46bbfa6a5e58f326cc9eb7b821ba2","subject":"Revert \"Remove virtualbmc package\"","status":"MERGED","created":"2020-06-10 13:08:46.000000000","updated":"2020-06-10 18:47:15.000000000","submitted":"2020-06-10 18:45:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734828-1591814713803-db77c85b","meta_rev_id":"391b1c498303f75e3b89f58df6867388cd9f4712","_number":734828,"virtual_id_number":734828,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 18:45:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-10 13:17:01.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-10 13:14:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-10 15:57:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-10 13:12:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-10 13:27:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 13:12:54.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-06-10 13:27:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-10 15:57:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-10 18:45:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa7b86456b365e011dea620de8124c5d501665da","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-06-10 13:08:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ff35bf68ee870a3982ca45cd2f46dc33934ccae","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-10 13:12:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b717ee36c062f2ce4d4b719b3e4846e3860fc311","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-10 13:14:30.000000000","message":"Patch Set 1: Code-Review+2\n\n+2, but we need to fix the issue of where this is being used without being properly declared first.","accounts_in_message":[],"_revision_number":1},{"id":"0be31ec214aaaf6300e8b0169c369c2534a3d31e","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-06-10 13:17:01.000000000","message":"Patch Set 1:\n\nOkay, where should we declared if it\u0027s used in the CI (not as a Python library)?","accounts_in_message":[],"_revision_number":1},{"id":"991c478d45c4ae82cd122bedd94b8bfc7fcc6e43","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-10 13:27:11.000000000","message":"Patch Set 1: Workflow+1\n\nSubmitted https://review.opendev.org/#/c/734836/ in ironic to work around this issue.\n\nPushing through since this is blocking the team.","accounts_in_message":[],"_revision_number":1},{"id":"54eefb21833f5f11cd2622fcca95f408816d39da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 14:36:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/718df51934014af2822284e9c6909a95 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f5555127ef34cd4829dcd7c6f5b7f41 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c66d9eb49c04cea8d9521a38f8de267 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5f9ef13e95044288c831afe38055983 : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a9b7dbae2ad47ae9d1d2bcab7f2edca : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9416dd6f168b4013881d11d01a298f7a : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c9a57208b8fe42a29234bb1b1e7865f4 : SUCCESS in 9m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/313a5c0a36a946b086d8536ed5bab5ea : SUCCESS in 9m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9500442670044a3b973b23f07820c409 : SUCCESS in 11m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bfe7de5ca191448b9d6024f722c2057e : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5eb920d6f2d749d3b963c82d7cf8ce48 : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e57878cb95344a6b9e828a8eca000e1b : SUCCESS in 11m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8ea8e461fc5e492f806581433950c040 : SUCCESS in 17m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3735b9da67ca401ba6118b50a75696f7 : SUCCESS in 19m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8dfb5beb24154ea3831a23eed3243405 : SUCCESS in 10m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ae46a4c9aeb4454a8ee548f791459333 : SUCCESS in 14m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c0d6f066de204815aaa0602cf6c88b44 : SUCCESS in 7m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eeb6af490c3a4378a60c5965c075ced8 : SUCCESS in 11m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d557839651604b8da352c18e12eb42c3 : SUCCESS in 9m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9c405f114ac43f1a090b09561d61d9a : SUCCESS in 59m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7970ee114a404ee3abc7a943e5c675f5 : SUCCESS in 1h 20m 21s","accounts_in_message":[],"_revision_number":1},{"id":"488da9100484368c12ec0ed2a03404427b331f6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 14:37:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c72d0dbcfb5689af3675f314b9452d4008ea44d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 15:05:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4416395ecc0e4a94b705256a1b79f1cc : SUCCESS in 1h 52m 51s","accounts_in_message":[],"_revision_number":1},{"id":"86d120c11a1a358a7f55b4014c05375e53e69d6f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-10 15:33:27.000000000","message":"Patch Set 1:\n\nis this still needed with your workaround?","accounts_in_message":[],"_revision_number":1},{"id":"e355ea729809633c3787d346055bf5865b55f88b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-10 15:56:07.000000000","message":"Patch Set 1:\n\nYes, the workaround is just a way to make sure this doesn\u0027t end up happening again.","accounts_in_message":[],"_revision_number":1},{"id":"6d991ea12d2369a01ed7b9711abffabe653403e4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-10 15:57:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"77e35cccfadca100413daa25f3ffc5d8b2574bf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 18:45:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f72eb3bf7e24bb3a036025b211016e9 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60d8477d4aed4d409fdff8ddebe19ee3 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5c89ecb65b043c2a7085180a6692d69 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fdfd9a5c078440b98f591d96e12c9e9 : SUCCESS in 8m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95d559f84b0248eab3a6135aa5cef12d : SUCCESS in 5m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0a7f3cbb36a439888cc91b3774efcc6 : SUCCESS in 7m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7a2a67ae9ee34585bc580b60cfc486b5 : SUCCESS in 15m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1606ccddb83b4320bada03ec684e9786 : SUCCESS in 10m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b2f2a91b7e1d4706bc04dbf745c177ee : SUCCESS in 11m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d73d9e69307643f3922f13f4cc6eb9f1 : SUCCESS in 6m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9f60cb2798b8429093bda0ca330ef630 : SUCCESS in 13m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/470da3c6e096400991dd627b359538ed : SUCCESS in 10m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/227da2cb40584e74bc40b17833fd2efc : SUCCESS in 25m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3f16dffd38cf459995e0610cbf53508c : SUCCESS in 30m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b4d7a243f3e04b07b7cef8ce8e049767 : SUCCESS in 18m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a5c6bff0f8274fdc90575ccc5b1914ce : SUCCESS in 18m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a2acb7088f864815a220cfe89c178954 : SUCCESS in 8m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/963bd79384e64385ac655808c4b17fc6 : SUCCESS in 9m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8155d192f93843d6a784a05fb81c2b48 : SUCCESS in 19m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5040410343bb4b6198da3be5cc11e194 : SUCCESS in 17m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf64e18e564e4599a114c6a938824735 : SUCCESS in 54m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b756741e1914ecd9b452845b54c45dd : SUCCESS in 1h 16m 50s","accounts_in_message":[],"_revision_number":1},{"id":"832238c7d8b0a9078460aca578c9a129128ea9d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 18:45:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8018faacde4b197d89dfbe8de6a4e051148c179b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 18:47:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8401c1b9db9043fa8b7aa0802b7b0d9b : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0029ce22ea87bfae6b170d9c6d1f1f5ceadc5a19","revisions":{"0029ce22ea87bfae6b170d9c6d1f1f5ceadc5a19":{"kind":"REWORK","_number":1,"created":"2020-06-10 13:08:46.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/28/734828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/734828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/734828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/734828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/734828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/734828/1"}}},"commit":{"parents":[{"commit":"bded0f0c3e945569dbbaf6a6fc44b407454a6408","subject":"Remove virtualbmc package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bded0f0c3e945569dbbaf6a6fc44b407454a6408"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-06-10 13:08:46.000000000","tz":0},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-06-10 13:08:46.000000000","tz":0},"subject":"Revert \"Remove virtualbmc package\"","message":"Revert \"Remove virtualbmc package\"\n\nIt is used by most ironic CI jobs, which are now broken.\n\nThis reverts commit bded0f0c3e945569dbbaf6a6fc44b407454a6408.\n\nChange-Id: I8bf7446df2a46bbfa6a5e58f326cc9eb7b821ba2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0029ce22ea87bfae6b170d9c6d1f1f5ceadc5a19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0029ce22ea87bfae6b170d9c6d1f1f5ceadc5a19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
