)]}'
{"id":"openstack%2Frequirements~734829","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ibfa97acab34cc8e17966b4eea87455482b24f4e0","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Ibfa97acab34cc8e17966b4eea87455482b24f4e0","subject":"update constraint for oslo.log to new release 3.42.5","status":"MERGED","created":"2020-06-10 13:12:26.000000000","updated":"2020-06-10 18:45:28.000000000","submitted":"2020-06-10 18:45:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734829-1591814728877-ffb2baeb","meta_rev_id":"8d35bc1a806a47ff8e58ce6548f58341232f12ed","_number":734829,"virtual_id_number":734829,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 18:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-10 13:16:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-10 15:57:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-10 15:57:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 13:16:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-10 15:57:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-10 18:45:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18b99ba5a765c6e044662365771380b0e2e56248","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-10 13:12:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d390eca0c871732693457622fd08e2ff7247ad4b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-10 13:16:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"46224ff2c099b71ff4d0251ac195edca17ea4d06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 15:08:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78e68a4e1cb9436588e4d913c1f5c298 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/894944a38c4c4002b00a03d93be6b69d : SUCCESS in 4m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0a0d1dc774b244a0a5791c6d382b1ef6 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a18619696224c9c9c08cb493445c651 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/77bc34b61a1a4e26948d2efc0f732001 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/80963beff2b24c80aabb6e2cdc292cc7 : SUCCESS in 4m 17s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/59997474420449f99cdf59782ecb49e3 : SUCCESS in 8m 31s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/d31cd77ed8f048aaaf2876c9ee2dfd01 : SUCCESS in 9m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/804076fbf4314fbeadf10e8525a6ec1b : SUCCESS in 9m 15s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6f7afa932a4b47369d384970867685d9 : SUCCESS in 8m 56s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/c74991e4323c44e28b79916ba8f0c7b8 : SUCCESS in 8m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/07385d5e8e3e45dca5deb234f33f21bb : SUCCESS in 5m 31s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/a4bfdb65edbd4564bd4a57ed13e128db : SUCCESS in 6m 16s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1fa834c35698441a819cd93688307f7b : SUCCESS in 13m 40s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/f2049acfaad949eba6c6c67d34c17413 : SUCCESS in 12m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e27679afdd334bb8abbc6350f8ba88a1 : SUCCESS in 11m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0460e1acd400435e881a68662abe414d : SUCCESS in 23m 29s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/ce7044f78ca24c5a81231d9619989f09 : SUCCESS in 30m 20s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f5075c5eae0e4aba81afba2190fc0a2f : SUCCESS in 26m 00s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/023143836769493c9344da886a1024e1 : SUCCESS in 23m 23s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/644f4231d554476db2c610b4546f5eb7 : SUCCESS in 14m 49s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/50b8b798394746a4ab10a35e46cc0714 : SUCCESS in 20m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/30b99c3c474542b5b6db37534a741587 : SUCCESS in 19m 57s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/fe2ffd5a45ca4f9aa576996a4f1a3a9d : SUCCESS in 7m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e5b67eb5fa394c20a1a992d26b29af4c : SUCCESS in 10m 02s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/8840a33778064ace84cdc122d2987dde : SUCCESS in 9m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e23c01d29f024941b92601835087f1a2 : SUCCESS in 1h 12m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c0ec129d2ccd4c9ea39df783ef90f815 : SUCCESS in 1h 49m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4d878b227a2646b680bf14ccf6bf2a50 : SUCCESS in 52m 34s","accounts_in_message":[],"_revision_number":1},{"id":"bb6caaeba05cd537d3dfc213e965eb3d5658b388","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-10 15:57:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d9e61e69b17a238aa95fc8dba270ef75bd76f64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 15:58:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b07ecc09df1536dd182a055ae3f20991d55d581","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 18:45:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc031ab46e3e47a9999180ef202749e9 : SUCCESS in 7m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2c2fe77c2d8f474e8e12c8b8f66832b3 : SUCCESS in 6m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/03b1ac9201b04bc09e44fbb2695d1f34 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7a55d451c624ab981136fb80bc46698 : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8480865e2a254ff29d7b4adfa747f6da : SUCCESS in 7m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e80fd2ce04f74a2f8ab2172a539fe678 : SUCCESS in 5m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ed398948ea4543f8992d2e9c0e91314b : SUCCESS in 10m 08s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/4ef304ec34c74f8d8cb7ec6edbaebfda : SUCCESS in 11m 58s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e183863dca994919bb9b54cd2e0e6049 : SUCCESS in 16m 27s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/c9f0d695e1ba47548d65c67cca6e75af : SUCCESS in 12m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/dee057ee9243452bab25ebfee6b27c58 : SUCCESS in 9m 59s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/a1f241b2de984740b87aeb342890a0fa : SUCCESS in 8m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6766c5992db048b4b2e0d6ec252fd147 : SUCCESS in 12m 50s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/94e042cb999944ca9e47ddf345c2c83c : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff9f59bcfd3c4baab87246af87ef1ee6 : SUCCESS in 8m 40s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/fdbf5299f8f44530978888df1add6fce : SUCCESS in 16m 24s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/05e3b8e243714d4f8af1041fcff7f908 : SUCCESS in 23m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/14472270bb154b989ece17dd470db730 : SUCCESS in 27m 20s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/7200d2a171dc431ea1b58b2fb80319be : SUCCESS in 24m 16s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/13f726c58106410dac5b0ad1ad38c0a1 : SUCCESS in 14m 26s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/8e05a508708c4147a712dea0a82cafff : SUCCESS in 15m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbc7b06ad16d466fb82845f7d811abf1 : SUCCESS in 21m 47s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/0b434480e7874093883f55775f22d844 : SUCCESS in 8m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4b5c0623d35d406ebe20c9bb75428020 : SUCCESS in 15m 39s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/3d256846a7334bbabe94cfa6d039accc : SUCCESS in 13m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f7ecc3bcf8644552a45d847e2ac3a217 : SUCCESS in 1h 29m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/24bffaa7455d43728033432ebe4d7169 : SUCCESS in 1h 16m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/76044778331a43d884b25cbc2c2025ef : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ccacc3f294bc77d41b91b37f8878df1fd30b4bb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 18:45:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d091e1ac3e1790d7633bd5e4592fd3e2a923ec6c","revisions":{"d091e1ac3e1790d7633bd5e4592fd3e2a923ec6c":{"kind":"REWORK","_number":1,"created":"2020-06-10 13:12:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/734829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/734829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/734829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/734829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/734829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/734829/1"}}},"commit":{"parents":[{"commit":"91f0aa3eca667c12b2c8ac9282486ddd27d7b323","subject":"update constraint for os-brick to new release 2.8.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91f0aa3eca667c12b2c8ac9282486ddd27d7b323"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-10 13:12:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-10 13:12:24.000000000","tz":0},"subject":"update constraint for oslo.log to new release 3.42.5","message":"update constraint for oslo.log to new release 3.42.5\n\nChange-Id: Ibfa97acab34cc8e17966b4eea87455482b24f4e0\nmeta:version: 3.42.5\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I19da401c42fd4453c91955c51baaca13a601b68d\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d091e1ac3e1790d7633bd5e4592fd3e2a923ec6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d091e1ac3e1790d7633bd5e4592fd3e2a923ec6c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
