)]}'
{"id":"openstack%2Frequirements~734831","triplet_id":"openstack%2Frequirements~master~I1fe60a9207c5d4517b701c48d98e1db853bf3a48","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1fe60a9207c5d4517b701c48d98e1db853bf3a48","subject":"update constraint for cliff to new release 3.2.0","status":"MERGED","created":"2020-06-10 13:17:24.000000000","updated":"2020-06-10 18:48:41.000000000","submitted":"2020-06-10 18:45:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734831-1591814730162-00e32a87","meta_rev_id":"597c2ef4a65522c4906e9011497d5285823c1117","_number":734831,"virtual_id_number":734831,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 18:45:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-10 13:17:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-10 15:57:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-10 15:57:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 13:17:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-10 15:57:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-10 18:45:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80c2317380fb091f3262052fed92aec36d958960","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-10 13:17:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8d5f3b642e49fde7182c8d91137e399648426c6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-10 13:17:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cfb2e906388960369300d7d5f252dd0f53930bda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 14:50:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/466ad360c4954169b20f362c610e9d09 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59cc4d9f2d0044328d040893fcbce64d : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/833999fb5de348feb6951be5bee5cd77 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/275820c7135f49ca88dc29a81b078c03 : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33f6f00497b9454baa57b67b528d328e : SUCCESS in 6m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8e18e8f66a984279ba6e9246b7bc6271 : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0bd491da5a004a19b1b9ceff0f15c515 : SUCCESS in 11m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/db43040acf7f4087a9e9142004dd8c2c : SUCCESS in 11m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/006372cdbd0b44b99f8f524e9ffa28f8 : SUCCESS in 13m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7914667ec86b4efbb1548a5952ac856b : SUCCESS in 7m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/02d4f98090854be092bdf654cf9a526b : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f92c62ef0b74472801199bf134371ad : SUCCESS in 9m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/01b4dafa89404df0b3269fb450a83129 : SUCCESS in 16m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9db25a96a61b4ab186fe2f17616789ee : SUCCESS in 29m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/74c3181e26e94926801254576f299a76 : SUCCESS in 26m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b31898ef43e04a1594277e28d067437b : SUCCESS in 21m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/283dda856df34c93aafea96e28df4a1f : SUCCESS in 7m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/808cee3b81cd4338b080b7574eec267f : SUCCESS in 8m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e72774ee35344dcaa67d716d8461e258 : SUCCESS in 10m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1611bc348d2c4c5fa198fde102695ef1 : SUCCESS in 54m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8cd9b65aea944b68932e37204bcf20e : SUCCESS in 1h 29m 32s","accounts_in_message":[],"_revision_number":1},{"id":"98bb4525a708678e3091efa941fc5cb0e3c84e69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 15:14:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e18a9446d6d54502970bf1a756a170b8 : SUCCESS in 1h 52m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5b6dfa39ca58d9b7ef9aab904bfd3c0f86c8f14c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-10 15:57:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb20a03de4346e0c7c24b97230c6a87383fea1bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 15:58:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4e2d6f3554043110690e01a6899b844bc862a4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 18:45:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87d0104605cc47baae4b2531f9a7c920 : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ea67ffc38b3486eb283ad0cede370d6 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c51d44c3704e4412b0a5e2b2b2c4a4a8 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be9871b1f1e74b1cb70d3c9543c0f0e8 : SUCCESS in 7m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2d25b678ae34ffcb264c0e5c36999ee : SUCCESS in 5m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7942e7fb399045e2b4d9904beca11561 : SUCCESS in 6m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/56ead0a369bc4efdb1dd4f8874ab7411 : SUCCESS in 9m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/018e2cebde57496dbf695e60546ec0cf : SUCCESS in 13m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/319908f1a188425da48a9b34b8cfe31b : SUCCESS in 13m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/99ed05c67d2346bf8f9bb077176feea7 : SUCCESS in 8m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/90136f5ef65d4610bbac42a67ef4a955 : SUCCESS in 13m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/11943bc3dd5140d1b7e1ea4300e2f030 : SUCCESS in 10m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c35b61de47b541779b4e6384bbbeeeb0 : SUCCESS in 18m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8a7ecd541746479191cfe9f6e65a3d74 : SUCCESS in 26m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/99a39bd4bf80408297cdc2af4c9b0298 : SUCCESS in 15m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/52eee29aece544e5a6c8c6565e0d3388 : SUCCESS in 20m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/80d5327824cc4a898d0799478ee54fab : SUCCESS in 9m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/456cd088294144e0b12f1b7177318f63 : SUCCESS in 10m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b79dfd2cb981496e8bea145daddb3af9 : SUCCESS in 13m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f62bfd66a8f445859765f41246e04a77 : SUCCESS in 13m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af6d3a2713824213860323a862f1be9c : SUCCESS in 1h 13m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0561853443d4beb83380a3f27639230 : SUCCESS in 1h 37m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d90fc494e613148a692d5b0e693862b326fefd54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 18:45:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"972607ca97fdeac00561ea8d94d269e0659aa7d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 18:48:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b2fa2b69454567a498412300f8f246 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63f772afe9bfde4429a4ed6a9352e565c4fcef68","revisions":{"63f772afe9bfde4429a4ed6a9352e565c4fcef68":{"kind":"REWORK","_number":1,"created":"2020-06-10 13:17:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/734831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/734831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/734831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/734831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/734831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/734831/1"}}},"commit":{"parents":[{"commit":"1b6e378b4644eae14d130ffc83df99d384a89f42","subject":"Merge \"Remove grenade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b6e378b4644eae14d130ffc83df99d384a89f42"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-10 13:17:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-10 13:17:21.000000000","tz":0},"subject":"update constraint for cliff to new release 3.2.0","message":"update constraint for cliff to new release 3.2.0\n\nChange-Id: I1fe60a9207c5d4517b701c48d98e1db853bf3a48\nmeta:version: 3.2.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Commit: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Change-Id: Icfdb876e7fdb174ec54b97d30d53b592339700e4\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63f772afe9bfde4429a4ed6a9352e565c4fcef68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/63f772afe9bfde4429a4ed6a9352e565c4fcef68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
