)]}'
{"id":"openstack%2Frequirements~735029","triplet_id":"openstack%2Frequirements~master~I5c6372fd1d72eb47a486846b2bb5a852055811ec","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I5c6372fd1d72eb47a486846b2bb5a852055811ec","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2020-06-11 01:26:40.000000000","updated":"2020-06-12 04:20:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7486a1b7f30c6886f962807e1edda512ddf2c431","_number":735029,"virtual_id_number":735029,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-11 07:03:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-11 05:56:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-06-11 05:56:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-11 05:56:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-11 07:03:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc6c596a09adad71516ae76e75e1b3f34adc2f26","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-11 01:26:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1eceb92ef3ee5374540d23a960be46066ba4a61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 02:40:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9019d6420c094e62a6e62e6f04885fde : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72719c38084b448bb1ec1fa0af7adcd1 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2029e2bfaf44e0598da7f6064997f34 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fadae20fcbfc43a4a96e088e1f673ab3 : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2eb3a8b48b341569f7acc60f4ebae9a : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b1fe0bffef454f7aba83285a0a3e9c6f : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8d38981a6a574994895ebcb570c6ac27 : FAILURE in 7m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c4868ce459b3426b80feafb835db0826 : FAILURE in 10m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b5a2aeabb40a4da2a7a3cc07db3ffc07 : SUCCESS in 9m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d68b2563ac8c4f698717eb4bde65be5f : SUCCESS in 4m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/12892b893d3a41c1873cd3cc892a7ada : SUCCESS in 8m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/829da5fce64b46419b3cf5743e147aae : SUCCESS in 8m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9e27f58dd58c408e85911d6ce19085e8 : SUCCESS in 18m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ce84c4bef5b34fbf83cfc7f88b0e5b8e : SUCCESS in 23m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b48fb37c9e024cc5b14f89080f183dc6 : FAILURE in 14m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/79ce97ba2aa74a1781e33831d74d6157 : SUCCESS in 15m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/48b873d97891462b9135df98c1866196 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ab07b4551df4af8a0313d0450749e99 : SUCCESS in 6m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e343e981798740a8ac835eb2a42171d9 : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9cc24efc9edb42999d80490599ebb7f6 : SUCCESS in 50m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a71d863d8c340a6aa0692d90ecf4250 : SUCCESS in 1h 12m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1a3a034a810ad76dc206d149dbd73644f5b157cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 03:33:44.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a8dcff5d147f414b8a46a7bd4b0b4c8b : TIMED_OUT in 2h 01m 00s","accounts_in_message":[],"_revision_number":1},{"id":"81bc02558469a4acfcff81fbdaacbd6e75a73c55","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-11 05:54:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c3a826c1069da11ceafef63f7859df1b67d753b0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-11 05:54:44.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/reclob","accounts_in_message":[],"_revision_number":2},{"id":"e0b8482f97964d196c35cab0d72c27f3b65a29af","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-11 05:56:08.000000000","message":"Topic changed from openstack/requirements/constraints/reclob to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"93a7758158d9075c6589a3db2b42d4d91e209b6f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-11 05:56:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d2584a222d73760b534bde9ecd67ddd0c33aede2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 07:03:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/079e32c9d8f540a68bdeb3d5cc15fb16 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee141cb0c7f417d9b5b7c2bbcc9dfb2 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ab26e784d104a82ac5babf48d12ec81 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70cce75c28c0427b83660d716444ee64 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e2e1d6faa3194a85a34c64e77dcfe548 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a67086c998fa4f2eadeabadb82681d9d : SUCCESS in 3m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2d7ba18bc54746ebaac0f2b5c51ff100 : FAILURE in 9m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3b79473913264bb5992cdb1654c73064 : FAILURE in 10m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7ed28bdb740e4257ba38010faeae48f8 : SUCCESS in 10m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0e7b8aed9c5e4d7693475a629e6b4406 : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a605cffe434a4da9bd379ab948c927be : SUCCESS in 12m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5da8c0a2503b41ce8316eb1cd2d95343 : SUCCESS in 8m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7cd57b4718094405b9f879e75155cbf7 : SUCCESS in 16m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cb11d2371a554a1f86f0386762953f06 : SUCCESS in 26m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5283aa9f791d46d0a3b1c780b2a2ebe3 : SUCCESS in 19m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8e19fa5c1b34136b9862d80d19d6c9a : SUCCESS in 17m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fbaa257b5f254fcfa427330547d7c181 : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3d53d2bb60594d69a45079fafeb1d9be : SUCCESS in 6m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c90cc503f75547beb07c81952f8ec583 : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/842858c0f41b462cb3fb1243a84ed94d : SUCCESS in 49m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/665855e56ada499395defe456fbc4e86 : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":2},{"id":"f29940164a3b95d174251d50ded7baddbfb0b3e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 07:59:49.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/92a0c18012364696a71961d609f066c4 : TIMED_OUT in 2h 01m 19s","accounts_in_message":[],"_revision_number":2},{"id":"8046c0e170ecf6b40587fd6bb743eb42708b8a7f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-11 15:34:08.000000000","message":"Patch Set 2:\n\npkg_resources.ContextualVersionConflict: (rsa 4.1 (/home/zuul/src/opendev.org/openstack/requirements/.tox/py36-check-uc/lib/python3.6/site-packages), Requirement.parse(\u0027rsa\u003c4.1,\u003e\u003d3.1.4\u0027), {\u0027google-auth\u0027})\n\nnice...","accounts_in_message":[],"_revision_number":2},{"id":"534cd1b5129679c3fd6a842a219a31a2de9865bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 17:06:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/498d6b3fbe4d4e39855515e60c298fdc : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/410ea57a224e4ff7a7e959d93184e26f : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbe3310de580423e8f9beb0914484b05 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab2205d7725f44f094fbc55243b3ece8 : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90e97a711a4b4570946c8c7bf020587f : SUCCESS in 7m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/01eca4d058e44baba1b49554804acc64 : SUCCESS in 5m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/76b05185e58b45b2ad2fd1bfdd549674 : FAILURE in 11m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9c57735d5cea425f9e8f628f25be1ab8 : FAILURE in 12m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ec001acc9f9e4311adbfff88071fd563 : SUCCESS in 10m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b5f8ee5680ee48cdbd145b71fd7e15ef : SUCCESS in 9m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/797a420308ff4d89a12db497138b9717 : SUCCESS in 14m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2ffd93b77f7c4964b1f4f75873d579e1 : SUCCESS in 11m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e2823dd02022445495d05bb003cea7cd : SUCCESS in 20m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5089f96c024b4d028fc80f8abafcb3e7 : SUCCESS in 20m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/297d6b312327496393abb76122ea23bd : SUCCESS in 19m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1753b9108fb94941a712df0bde22acb3 : SUCCESS in 21m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0ebb8431375146c38217f6afcc47129a : SUCCESS in 8m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fee12bf9aabb4cd0bd29fa3ff772dcda : SUCCESS in 8m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5f8822e7d4f940529d94e3d4fd6d0d32 : SUCCESS in 19m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d763a6049254982871ab0314b0fb264 : SUCCESS in 50m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4e5f7bef8154e5bba54ff65fa939777 : FAILURE in 1h 01m 43s","accounts_in_message":[],"_revision_number":2},{"id":"7486a1b7f30c6886f962807e1edda512ddf2c431","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-12 04:20:38.000000000","message":"Abandoned\n\nnah","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7ac4f42fc40e3128d27b5ec07fc2334bc1653811","revisions":{"df4e9c070b56afe7eaac7b37ccf569cecfe8b79c":{"kind":"REWORK","_number":1,"created":"2020-06-11 01:26:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/735029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/735029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/735029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/735029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/735029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/735029/1"}}},"commit":{"parents":[{"commit":"96d16e8742120ac88babb77a05bba6aa225b9310","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96d16e8742120ac88babb77a05bba6aa225b9310"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-11 01:26:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-11 01:26:38.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I5c6372fd1d72eb47a486846b2bb5a852055811ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df4e9c070b56afe7eaac7b37ccf569cecfe8b79c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df4e9c070b56afe7eaac7b37ccf569cecfe8b79c"}]},"branch":"refs/heads/master"},"7ac4f42fc40e3128d27b5ec07fc2334bc1653811":{"kind":"REWORK","_number":2,"created":"2020-06-11 05:54:34.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/29/735029/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/735029/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/735029/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/735029/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/735029/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/735029/2"}}},"commit":{"parents":[{"commit":"96d16e8742120ac88babb77a05bba6aa225b9310","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96d16e8742120ac88babb77a05bba6aa225b9310"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-11 01:26:38.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-06-11 05:54:02.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I5c6372fd1d72eb47a486846b2bb5a852055811ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ac4f42fc40e3128d27b5ec07fc2334bc1653811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ac4f42fc40e3128d27b5ec07fc2334bc1653811"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
