)]}'
{"id":"openstack%2Frequirements~735372","triplet_id":"openstack%2Frequirements~master~Ie462412293765a7cc639eb2a5e21f49e98f0c40c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie462412293765a7cc639eb2a5e21f49e98f0c40c","subject":"update constraint for osc-lib to new release 2.2.0","status":"MERGED","created":"2020-06-12 16:46:24.000000000","updated":"2020-06-12 19:27:05.000000000","submitted":"2020-06-12 19:25:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735372-1591989946240-e4b0d899","meta_rev_id":"da24a30e727372e9dd2ff0d2cf354aadafa96adc","_number":735372,"virtual_id_number":735372,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-12 19:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-12 17:06:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-12 17:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-12 16:48:05.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-06-12 16:56:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-12 17:06:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-12 16:48:05.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-06-12 16:56:48.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2020-06-12 17:06:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-12 17:41:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-12 19:25:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d925e1bc9b91746f3ed638a712c060162d2077b4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-12 16:46:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a656e62895d73b55446657d42ebdd94b09f8091","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-12 16:48:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3603749bf3070211adebd6c9aaebc35da33ce4ab","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2020-06-12 16:56:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"78416841268aadb2cdfece4f7570e3cd1716fa7e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-12 17:06:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"99321d25985ca1090b7d525cb6fb28477572774c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-12 17:41:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"64209189de80bd834c811b5a61102cb41447bae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 18:15:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2195d293703b4066b22228ba4a80457e : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8e433b34c1a40739fa75bcfda0ff7b4 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1e679038b514d8d8f160bc8ea99a39e : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2999f76911954c79aa61d2306cd2964b : SUCCESS in 3m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12d86a3e80d249009a2d812a8b65312a : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/08abd2380a344e31a77577a6a317306e : SUCCESS in 3m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5c53a5306f0746b7a02ec20c26bcad4f : SUCCESS in 8m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/597c4825c9fd45538fbabc43559cccba : SUCCESS in 9m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ee4088aa518a437ea57d45e46b423ec4 : SUCCESS in 9m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/18233dd2a6b14f86885af4bb11b9dac4 : SUCCESS in 4m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2176e6fabb2e48e3867556b25fd12bb3 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6caa8e2c4a7a43789630ae7024e0e966 : SUCCESS in 7m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fecea7b478ff4aba82d7fc2610a9b9a7 : SUCCESS in 13m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/aec20b288fbc4cc8b652a5283338242a : SUCCESS in 27m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0b25d64b2272481aa7d486aed4ccc7d1 : SUCCESS in 16m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a41aa29e83f490e876f4727fedcc283 : SUCCESS in 17m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/628ad5dacfd44a8ebd19299eeaa6eaca : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/432eca6485dc424fb547b8cba196f80c : SUCCESS in 7m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/83145ad2a0da4058928aceae33cce7ea : SUCCESS in 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9832dd53a0c845028b9ce3063004b5d5 : SUCCESS in 55m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8481b32d5b034adc8087ffa1001a3962 : SUCCESS in 1h 20m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bb5bf6b2aa73a7c2bdcdb53f8014182df5105186","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 18:15:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18cd886ee99c9bd2d6e73ab61bf690c837a79f2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 18:51:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1d9246ea566d4694a925f63bf24aa83d : SUCCESS in 1h 54m 37s","accounts_in_message":[],"_revision_number":1},{"id":"bb776299f5019e892a94413e5d2e922c09dadfb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 19:25:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecbd888bc5c349bebe8db20d2c037381 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b3c320b2ad44c5491461ac4d2b8ed12 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09cf73a28a884fde82a0acd862e71ba0 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c5f85dd74fe446bb315b2cedf0855c1 : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e108836f4fc945a89711dbc94720b198 : SUCCESS in 6m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e1a6eddfb7b419f8b67357da8fe1c20 : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/55c53ed8994a4ef0a16375596d2a0ad4 : SUCCESS in 8m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2434f41e05084231949540a126c1a57c : SUCCESS in 16m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/166c7ee28cc94697b5094acb8e555491 : SUCCESS in 8m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f9820ac414b44835a6fc9609a77d97f2 : SUCCESS in 4m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2c54710753964bc8ba71d8bf67f954e7 : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a0e485e52e774ec7913e75aa8df3c4a8 : SUCCESS in 11m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/372c860c6ad2412e9013011fbbcbc4eb : SUCCESS in 16m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c9d5b5a987b14cea83f89be84dd4a47b : SUCCESS in 26m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5817d8d3242b4189ae4220e860c4cce3 : SUCCESS in 13m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b4ca1ed2042f491d9986333d2d6e18b6 : SUCCESS in 19m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/955d4111458a425a913060ce178a5bdd : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f02de1cae4246f0b286c920cc3e8bbe : SUCCESS in 7m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0dcf2a3c2679494ba074b159b57d0680 : SUCCESS in 10m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0265c57e39e8493eacb871448b1572a8 : SUCCESS in 15m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82e78e954e86486dadd0273d9f3e8967 : SUCCESS in 59m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cedb961bb9734940a5f92ec9eff467b2 : SUCCESS in 1h 09m 01s","accounts_in_message":[],"_revision_number":1},{"id":"5556056af8dfc5ecdf30a140cf3f20c06e0739ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 19:25:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3acc84057139778920f2e18ad2a4a73180fa73da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 19:27:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c35e3912bbb4488c8c8f0e37ab02c705 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d429bb89c5f00fb8939d016dd9eec21689c32782","revisions":{"d429bb89c5f00fb8939d016dd9eec21689c32782":{"kind":"REWORK","_number":1,"created":"2020-06-12 16:46:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/735372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/735372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/735372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/735372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/735372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/735372/1"}}},"commit":{"parents":[{"commit":"713ba79507e0b25703811bc5f509b758c1551b31","subject":"Merge \"Remove cap on cmd2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/713ba79507e0b25703811bc5f509b758c1551b31"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-12 16:46:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-12 16:46:19.000000000","tz":0},"subject":"update constraint for osc-lib to new release 2.2.0","message":"update constraint for osc-lib to new release 2.2.0\n\nChange-Id: Ie462412293765a7cc639eb2a5e21f49e98f0c40c\nmeta:version: 2.2.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Commit: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Change-Id: Ia086e17d2604ec4a66451f77bb3b6046a91c246c\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d429bb89c5f00fb8939d016dd9eec21689c32782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d429bb89c5f00fb8939d016dd9eec21689c32782"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
