)]}'
{"id":"openstack%2Frequirements~735534","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I0ef8f8d83bf22a00b90087ec6ff9c424a24ddd20","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I0ef8f8d83bf22a00b90087ec6ff9c424a24ddd20","subject":"update constraint for python-neutronclient to new release 6.14.1","status":"MERGED","created":"2020-06-15 09:23:33.000000000","updated":"2020-06-25 09:04:38.000000000","submitted":"2020-06-25 09:04:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735534-1593075878866-39350062","meta_rev_id":"00495dca390d7f024d0e264c8fc82357bee261bc","_number":735534,"virtual_id_number":735534,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 09:04:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-24 20:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-25 02:45:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-24 20:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-25 02:45:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 16:29:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-22 15:58:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-25 02:45:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-25 09:04:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd5512bc061c02a3d94b48bee4edf11380456aec","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-15 09:23:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab9d2dca4e4c2ee101875f99a9201549011a85ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 09:52:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/614e7a2e231a4e47af21ef91c9e90e89 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7dbb7934a59b4d6c9fe4c046f958b2a5 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31afccf5dabc4d388822af791cfe0949 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c83efaa445a647a9ab5986d299da5dc7 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37d93a83b980481b80323bf1fc5a06de : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/716aba6b63204637982c3eeacfcbc9e6 : SUCCESS in 4m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1201040ba6e3487592cd34e8c66f96cb : SUCCESS in 8m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a4717ebd43364cb3987d78d6b7eb6172 : SUCCESS in 8m 49s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/62151bd7c7444a6c9a2f622cfc437300 : SUCCESS in 10m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d71d31131d504cf0828adffc5efdf5a7 : SUCCESS in 12m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1b3cd55f9c9e47aaa7cd0100f060b328 : SUCCESS in 5m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cb4f84d41db64c8e8b174646a1356c87 : SUCCESS in 6m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e55059d9c73f4bd7bb411b62415a0523 : SUCCESS in 8m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/883a0da55cb54af2a96eb3c3cf1885a7 : FAILURE in 3m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/646c008bcbbe4db6ad99fd00fd0eb776 : SUCCESS in 8m 33s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/3a8e9b5a9669430ebb58e795d99473b9 : SUCCESS in 15m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d0b7046d01da487fb251970134a66f14 : SUCCESS in 14m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/894118fc3cb542aaa1ac46f3e6a1f64d : SUCCESS in 17m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ed6d2a76fa62475d9bc79f704be3a436 : FAILURE in 3m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4ba156607d074b19bebf2fbd000cbee9 : SUCCESS in 14m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4cd16c0985404cd2a3e583f4834e264e : SUCCESS in 10m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/35939db7276a4121b7c917c95329bbb1 : SUCCESS in 15m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/819d029be61644859c1385d06f30b43f : SUCCESS in 21m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/366f3ec92f1b46098b8c8e174bae269e : SUCCESS in 6m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/592d2f3559b84439adcf06bfeef5debe : SUCCESS in 10m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/19984f6ea2b8409e98b486c3a66991c9 : SUCCESS in 12m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9cdc594f7ff1493a94821fed1e4883d8 : FAILURE in 19m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a8db5e2139d340a48377ece4210f8f58 : FAILURE in 13m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/db9314ec8a0b4547a764539e6d3cb052 : FAILURE in 11m 24s","accounts_in_message":[],"_revision_number":1},{"id":"001aa65c97a2841d55506e445e2a7d15193d8e24","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-15 16:19:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c2382a0a31cd6e0d91e6a743d183246bb3eaa85b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-15 16:29:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d518aae32143034e460f2e90780ec9c2cbc9b78a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:12:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77110b9cb3ee4e52a4c8eac2c2bc2676 : SUCCESS in 9m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2a191d4e74f460cbd771de308957fc1 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf7035d932354102a35e2ab5cd081828 : SUCCESS in 7m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b22db56ad86480fae9520e14c154971 : SUCCESS in 7m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43c5032e09d64c89aa6005b8108d4a8d : SUCCESS in 6m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/834a20da5a7241eba9e75859f9d6f3fd : SUCCESS in 7m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4516fadca9cc46ff87588315fb4015e6 : SUCCESS in 9m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/972ae68dd83a487584361b0c72f6e8c4 : SUCCESS in 10m 14s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1c4c25a4b23c4f9b8fa2b795bc0dfd67 : SUCCESS in 10m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/762b628749b7417eaca24a53e0643ef5 : SUCCESS in 10m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4b2eafdb53f044d6950a1de222b633db : SUCCESS in 11m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e10056d7e9f942c79357655b13217425 : SUCCESS in 10m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/082d9ddb712442849740c776b946a4fa : SUCCESS in 14m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b1a6ade9687843eeab34e2784064b690 : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b26f6cd763b743c6a9e102ff14cb2299 : SUCCESS in 13m 02s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f6d25dee91434cac910dfb87b06516b0 : SUCCESS in 38m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ff355facbb3b49f08e22b95bae0e23a3 : SUCCESS in 24m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d1727a614d514038b0001046ab09b877 : SUCCESS in 25m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4e1babd4d7c94e329415ca329cc59fba : SUCCESS in 27m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f3cfc21c21a04d89aeb05a718f08387a : SUCCESS in 19m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/32b987d3f7e64be6806ee0f8205238ae : SUCCESS in 17m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f9fbbf4572554efea96ff50040298504 : SUCCESS in 23m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/33c981e393e24462b3c678f6de9a491c : SUCCESS in 27m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a7cbb22948a24d18a798129e160a004b : SUCCESS in 8m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c54b219576f34b8f972cc7c1cd78e5af : SUCCESS in 21m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a1e787da8fdc4a34a14fd69828eef201 : SUCCESS in 11m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/870313e439aa4652ab8079d99295901b : FAILURE in 23m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e9635d045a014ca69eeb829ac52067f8 : FAILURE in 14m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4a4615c7bb164b7ca7e08e37f92b2a50 : FAILURE in 21m 52s","accounts_in_message":[],"_revision_number":1},{"id":"84a5d4b2ce46869b1a1f3a5fc7972a1ba259d2f0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-15 19:42:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9591af93488398d728bfbfe3d59f23d488f720c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 21:53:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cba558ccc4ab475883707762b4993dc0 : SUCCESS in 7m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/56adb98582934ada9351885a7889c95f : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e029da25fa24f72b1421d0e3b9af6cf : SUCCESS in 8m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9f5355da179449d889e9e4738401453 : SUCCESS in 8m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84888f73cd7341ad8fdcc5df05734575 : SUCCESS in 6m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ca95d3a5da80400494f14ce160d5268b : SUCCESS in 6m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bb4d182400384a1f877d141d147dad1f : SUCCESS in 12m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f438761968ce4aa88c58dcd7e4ceb90a : SUCCESS in 13m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/aee04c2b0bbf4414bacb8c92401fc64d : SUCCESS in 14m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/85b30e1bc959429dae28cb8cc7adce3c : SUCCESS in 14m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/13e4acbeb3f84061ad7f09dd8e5e1791 : SUCCESS in 10m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eca23a63e6914269af22ff5828120f8f : SUCCESS in 8m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b39ca5ce419a497a9403b1385d3aedec : SUCCESS in 12m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/506fc634b45d49f588765ea37eee9aca : SUCCESS in 11m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce8e9df6c9354f688b1a7a53309d1da5 : SUCCESS in 12m 20s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ad7e33d4129540329866a706d1ffe074 : SUCCESS in 21m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/876090e0ce5d44838972a29e4be9d0f4 : SUCCESS in 24m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4f54717f8cf14c2f88e136f403bde6d6 : SUCCESS in 28m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0df9a117079e47dfbe5d57d223c43ab7 : SUCCESS in 25m 27s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/857f08c56b244821a53d623f3359415c : SUCCESS in 19m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cc0c10f1721045e1a8cfa0b54af0325a : SUCCESS in 16m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e8d1f9816efc414b8462c886a7abf0c2 : SUCCESS in 22m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e3a9a8615b0944e4ac696743bddfd2fc : SUCCESS in 24m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ffc00e64037145caa3e070e47cb39205 : SUCCESS in 9m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4b3706f278164412b7a206675afcc30f : SUCCESS in 11m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/22167bfcfb6b40ff9c9a2aa535087132 : SUCCESS in 13m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b721be5b98d4bfd881e535bd7df7cd0 : FAILURE in 25m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d65436103322485683893b05f77d6bef : FAILURE in 15m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c66d468a0adc4b7885389da69615cefe : FAILURE in 23m 49s","accounts_in_message":[],"_revision_number":1},{"id":"259a3dc61dac7c63e2c54ee76f77c1515fedb76a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 15:47:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3e9e90c28999b06b994cc644b381ca996e67d275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 22:59:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4a518497cde40f6a936d747f8cc1f11 : SUCCESS in 7m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3683755b86ea49cc92ea49e5ccfa6bed : SUCCESS in 9m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0741b7f2a03409d92b3cafa71790a9e : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa3b4e8968b246fa9e73653231b43df7 : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9eb91abaa1e3481ba6c5998e05994af2 : SUCCESS in 8m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c523601e961450bb0587112c8c34452 : SUCCESS in 7m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b627e0f7387f4934a7a626eb9e4f238d : SUCCESS in 10m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8058ab3169cf4e069681f0b9a0a7c521 : SUCCESS in 11m 29s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c186a2e455b84276a2512dcd06b7e43a : SUCCESS in 19m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3d922ff73c36412b9b98ef26c42a9dca : SUCCESS in 18m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/504a69cf765240358465aa5d461caf2a : SUCCESS in 8m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2b5965c79764490f9b286697f39f57fe : SUCCESS in 8m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f2267b9bd3c94101b604ecd4bbb7fcbb : SUCCESS in 14m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9317e69d282c46819dceec980b17b031 : SUCCESS in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eaf98c9a73a3460abda40f3331c530ec : RETRY_LIMIT in 8m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ba7c02f185084444bd5c5ef4268acd3d : SUCCESS in 24m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8c43558a4060460ba8e2efea0166b1f9 : SUCCESS in 18m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6202d257ca82419da7377931f412f634 : SUCCESS in 25m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c35cdbe1cb0a4a2c9b2d886cfe44b482 : SUCCESS in 24m 29s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c0e40614ddca45c18d8c6f977d565d03 : SUCCESS in 15m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/122113ccc0454bb89e5408227cf8b0ed : SUCCESS in 16m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/96347a905bfe47678830c3faaeb62bae : SUCCESS in 18m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e3df7d5ee91d42f3b3c06460fbef7343 : SUCCESS in 24m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/51cc2645080140188a4c099927decc28 : SUCCESS in 11m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/883691b80e99489cbf779c3a61e87528 : SUCCESS in 17m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1a83856efbe546b99f1e471b5e65b286 : SUCCESS in 16m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/089cc4c0e94d47ebae5512a2539d8357 : SUCCESS in 1h 11m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4f2223e4ab944a09b19d2752bbae0bb3 : SUCCESS in 1h 45m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/523e18a616c248eb9615a1071c8c9855 : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":1},{"id":"051098b9f324a457c3a1a87270b7085900e3e72b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 23:37:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"66de217a38a45730e5ecfb30770b5125c554848d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 02:09:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4b16c06cbc448d193aec839d0bd4f56 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ac37c3d551a4320b490d92e69faa8ef : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d93e3e1d8af1478c8045b01a01c1627e : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4cb890f63be046fc8412e7ad6a30e27c : SUCCESS in 8m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/230eb34d1b7248cc80ef1b26f15bb154 : SUCCESS in 5m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/efe942fce66744fb8b3724ff0437fef4 : SUCCESS in 6m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/160bfb21557248d0b7eac5635fa141ce : SUCCESS in 8m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8c5ec2c1945a4bd782010cac670113aa : SUCCESS in 10m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0293ac2276224610b9df8f3142726674 : SUCCESS in 11m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/536dd6cf1b2d4fc2a33963702ca83de9 : SUCCESS in 11m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e46b3e2dbfc14773870a43f085e91ca4 : SUCCESS in 5m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4cdb01baf86641e09f19970344629bbe : SUCCESS in 7m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/290401e2a4e1483495ef4c92db0d5d4d : SUCCESS in 13m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/82753a40ad7843c7bb32dfd0868ded46 : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c70a52fe62146c7bf6eedb63e46c589 : SUCCESS in 11m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/94efd30e7ebc47e0accc678946adf1f7 : SUCCESS in 49m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/50d2103f493b4a1a83a6aa3d019ce361 : SUCCESS in 18m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6e6de467a8fc475b98d53483ecebc87c : SUCCESS in 22m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/aeeb39c03d704a059b08f6b8e2d669be : SUCCESS in 23m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dded51f8920f4199a8266e37d1de11ff : SUCCESS in 16m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/333e1b2f33414a5ba3e40a8fc1701f06 : SUCCESS in 19m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f83780a1e6f1417bb2d990bd67de9d8d : SUCCESS in 14m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/99665c1b477244c0b06f025dc2e4844a : SUCCESS in 23m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f6734fbc7c9e44e1a2ee674bc7852fd6 : SUCCESS in 6m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c8b4df4c025b4b11882346044b88bde0 : SUCCESS in 13m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/719cb4d05a5d48e5b60425e19f884405 : SUCCESS in 15m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc42b30a92264eab91d3de9e5b5085e2 : FAILURE in 58m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/021d8538398a4edb81a8b65a8d54e47a : SUCCESS in 1h 32m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bb025a9fb4364e98adb0ca8670bbe422 : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6270008b744515f61b2108a3e1d3366c5712f8f8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-18 02:10:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1c03a5bd30192298d035ced4d1bbe70deab70e0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 03:45:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c12d0c068f4a55bd4497f6c0effc82 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/767e02f05645422e9acdf8731f82b416 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/028635382a49489c8871abe78c1c4138 : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b69dbdd198574d6ebafe882acb624bc2 : SUCCESS in 5m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80660a4c89484ef7bdf6de13ac5791ac : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d35b998ad2e54cee860e0a76e0b9f45e : SUCCESS in 3m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/af09c7d2372c42c4a759945b2346cfde : SUCCESS in 7m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fcba471f77934ccca4f1363178d9ae90 : SUCCESS in 11m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/29c757847af24b92928bd61dc54dc6c2 : SUCCESS in 9m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/17b7e63ee1f545b0ab799682c50f3efc : SUCCESS in 13m 05s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d25b4cb5abcd4fa7b8ebf50eb4f5c300 : SUCCESS in 6m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/edc110b4e6b74ff2b48fe2ff0032e105 : SUCCESS in 6m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bf161e508fa04ad5962cce3bf39c6b5e : SUCCESS in 12m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1e32ba37b5294bf4aeeb72e7a907547f : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d67b54815c1460fabc838bfb739b383 : SUCCESS in 8m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/230068a2706642bdb075eb1fc35c7523 : SUCCESS in 27m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ad370f19b41d4a12801537005dddc7e0 : SUCCESS in 15m 57s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b637ada152b04fc99c31dfc39df7641c : SUCCESS in 21m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d28d003cf68e4c039d9edc7828f20f3f : SUCCESS in 25m 20s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/387d1980fc214fef8feb63bde4a2576d : SUCCESS in 21m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/caf3c806f15e47e5a5f9b00c7a8b0ab5 : SUCCESS in 14m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f5540c8987b14f359ab51db8e3381181 : SUCCESS in 13m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19848a6a80984fe8b2d0f822617dead8 : SUCCESS in 26m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/222e4dde50b04695a08a8a7c15f04a82 : SUCCESS in 7m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cfc581b210164204baa376feec07d980 : SUCCESS in 14m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b7a215ebbc1041d1a42b1887a4e45745 : SUCCESS in 9m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a443c9fd522f45a5a1d8a749ca31952e : FAILURE in 1h 06m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f1456f6fed0147a4bdb3664b417d6285 : SUCCESS in 1h 30m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0725ead9b2254631a5aafa5b1e20dc52 : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":1},{"id":"cac7e0401846566b9b410599ecc2cd881ba61537","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-19 02:39:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"90d45ae5212b0710baa1e996e80d3fe36e288f4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 04:36:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d935e6ce4674b25af042500eff8d0b7 : SUCCESS in 6m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09b6df05d0a44d74a714a5f0cb4682f5 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2923ef7296994cb38e01dffa5de44fea : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/632676c695e64e0b98d52847531607a3 : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9db592507e8648038756246b4c3e1b5d : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92e9db6016194a98a4132334fcaa6d9a : SUCCESS in 7m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/44457a2db89c4496ac6269555178a100 : SUCCESS in 10m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/193a2e065350425990538b96a4c0b56c : SUCCESS in 16m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/779c3a54eefa41b995684c4209439624 : SUCCESS in 14m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/62854fb510d8490a992c03938688839b : SUCCESS in 10m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ffc29d7c13704b06a5075eaa20e3e477 : SUCCESS in 5m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dac39704b89b48b7b254c71621e412b7 : SUCCESS in 8m 03s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/149685e153244ca5b17c69d215058f85 : SUCCESS in 11m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f12ddbec41714c5e87fea120cd568fa8 : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3a42960072c74d8baff09e514e60b795 : SUCCESS in 8m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a696e34b4d404fcc9de0f14dcfee08a4 : SUCCESS in 17m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/886651efb2704d12a1801767906f9d76 : SUCCESS in 15m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a3cad9e029b34d25b20c6ce71e934d9d : SUCCESS in 26m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6a298c4701814dcc971852ec50525bd9 : SUCCESS in 28m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/dacd1697dd9b485c92c59150559a99f4 : SUCCESS in 18m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a473b2b4cb9b4560a2b70c34f906cad6 : SUCCESS in 14m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a4e95181e85f43bc84fe303d6da5e04f : SUCCESS in 16m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2503b36852f445e080a76db8838768aa : SUCCESS in 28m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f55093f92afc41658dad4940b96f7186 : SUCCESS in 7m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/55473d8927a14ff0a5d3b8c44345ecc1 : SUCCESS in 16m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/285c8624b85641ec9e8463c32088fb8b : SUCCESS in 15m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/71ded88946b44352b4578baee1974faa : FAILURE in 1h 12m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d5a1898d74644ef586af5c26f86b0e22 : SUCCESS in 1h 55m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4495e20db0de4abeb820781f04a4480e : SUCCESS in 1h 10m 39s","accounts_in_message":[],"_revision_number":1},{"id":"be05ed36f797bd93464ef69fe5e3a2a8ba032f98","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-22 15:58:52.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a85029145579bff66693f1403e505b29527ef06f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 19:52:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb8918e3d4c74849976b0832a0cc34b0 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f41e88b8b68341d395f1e95c6bca07f4 : SUCCESS in 10m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5741546aa54f46d68d7f24270ba919a7 : SUCCESS in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4322b6b6edcc4bd68c11ede0858b5843 : SUCCESS in 6m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b094edf87d9411eaba3ef09ff4ff9c1 : SUCCESS in 7m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d283bd714f1046d6a4e06df12432b4fe : SUCCESS in 6m 16s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5b8930a9c85b4579ad9f0e9fe91d04fe : SUCCESS in 11m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/432a6ae33e2b4efab7e358c4ab6e9bcd : SUCCESS in 12m 17s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/05391a8967874cf9a93aac29aeb485cc : SUCCESS in 14m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8dac0c4a81b0487faea9e7c23a7f63a0 : SUCCESS in 11m 24s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ce73e19cac88470abec8979acfcd372e : SUCCESS in 7m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d405d87c2e9a4c1c86bdf19e18a271ea : SUCCESS in 7m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8f77da51e24f48e4b76f3f737e600b40 : SUCCESS in 13m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/934cf9f727ee4072a780cdecfe9b6ff8 : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/72fdd9eb878b4f5889f82a7b8121a5fa : RETRY_LIMIT in 5m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c8da4f0b68944e49b493158042d6da9d : SUCCESS in 15m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5e57a9e3128f4742be74b4bcf5bd8135 : SUCCESS in 17m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3a4f19220d274214994f3feaa1ae001c : SUCCESS in 25m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3ca6c8f1584e4e4f9ea5396be4b1d784 : SUCCESS in 26m 45s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b5ef75cd5599434bbf433c045dc1113e : SUCCESS in 22m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b6713d6f27f145e38beb15ae7f404090 : SUCCESS in 15m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/75e4f365110940348a6bf3e188ba0f8f : SUCCESS in 14m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54599cfd6b6642408df9fb4bfa2a2096 : SUCCESS in 24m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6dd15d1779a1410ea22a6c211d9741c1 : SUCCESS in 8m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8219c9c4ee5749df91ea333a53b64bd9 : SUCCESS in 10m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/727f461d98e84c828d35a001f45e2160 : SUCCESS in 16m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d4e1599ea91481a8af03857191be3e2 : SUCCESS in 1h 11m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/08eeabd8e10a4b91a4761f53c946686d : SUCCESS in 1h 23m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b1542955d3264381bdbabb411dfc9dd3 : SUCCESS in 1h 00m 31s","accounts_in_message":[],"_revision_number":1},{"id":"50e692454b626f864b74a4bfda90edbda7fbdf3f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 16:00:28.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5b616b1902cc2682f59f2ce9ad2195bd22cdc65d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 20:31:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"6a2702a6c76d931a323861616f06112248274992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:33:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/855906bb42c14ce882b0c2cb413d5031 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8a0134993be48aaa68eb95d18efc046 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8b5928d54b84d6f9aeedcffde24adc0 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18b4cf3ee47d45deab68cbbff615d0f1 : SUCCESS in 5m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e2f0f5e0da24b60a6c9997da92634ee : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c665f45ca41420c8a9be80b6e868f8a : SUCCESS in 4m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c282b13e95174fa6b363db2539899094 : SUCCESS in 12m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fd1e70807beb406d9dc31cba9b51857d : SUCCESS in 9m 48s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/896e80eb20da454d8d355e15018fbd9f : SUCCESS in 13m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/09896c5d062e4dc68c5070cf3264d66f : SUCCESS in 10m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9be902f9d59749519dcce5e43064f0fb : SUCCESS in 6m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f227e6764c264a3cb6617c319487761d : SUCCESS in 5m 57s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/038ec9460e774765b3d53faa43289927 : SUCCESS in 10m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/eecb6680752f4ca4bec34b55ee140be5 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e51dfde7e093480598d6fbe2e0efdfd8 : SUCCESS in 10m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d369e32230b94b6788eb64346dabcd96 : SUCCESS in 47m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f6776d104f5b49bea2f893c245cf28f8 : SUCCESS in 15m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/cbc8340705a0444d843890be451537bd : SUCCESS in 26m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c5b050225ae140dab6b400a90db5873c : SUCCESS in 16m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/82b9b447430745cdad77ce1ca20ce0c2 : SUCCESS in 13m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ae39f94c33714c079dce275f7ba458ca : FAILURE in 33m 41s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a2ac3090195b45daaf2fe44537a34bf5 : SUCCESS in 17m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3c52104e76754238b002ebc22ddffd32 : SUCCESS in 20m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53743ab175aa4372a2015639fc63629a : SUCCESS in 6m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/405a045202504209bc2f9f49c49c5172 : SUCCESS in 13m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/75a15181e5c44c1a88cfd28a25adbdd1 : SUCCESS in 13m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/737d3fbfc84d453bbb51be381ee71973 : SUCCESS in 1h 03m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/559bb13f9602447092f37f050a4ce147 : SUCCESS in 1h 44m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c2189abe5b5a4ff9b084db28a05ef46b : SUCCESS in 1h 29m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ecafba7a1697e202fb616cf359cec8865bced91e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-25 01:00:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"26f8f1d958f152ddb33d8a499d0e5c5832922846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 02:33:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d51c86f4c2b042509cbdcfac3b08d065 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7364640ad67544cd837044cb83f772d1 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bd3c15affd943afaa24a3e8f332f57a : SUCCESS in 10m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b965fa7d1f8441197f594bca6776712 : SUCCESS in 5m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c749bdeabdff481d87364dab7790acf2 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dd5062969e8a40348b86d8335019c0a5 : SUCCESS in 5m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/039fc66439ed4a5e9dc3ec5eaaa38f3d : SUCCESS in 10m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1a1b005261c449c297e4441384750274 : SUCCESS in 8m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f28c7583e4ab4e83874648b5f58ba200 : SUCCESS in 15m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6a8141dfe66944ee8117bc47a833a996 : SUCCESS in 15m 12s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/70661d1bcb944c83b62f65838b7e3d17 : SUCCESS in 8m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7cec3d5048a247508e5f634906ec5442 : SUCCESS in 7m 57s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/da32b9771a324e40aa0bbd03590496b4 : SUCCESS in 11m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1239e397748345feabdfa80e1f0b09b2 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ab019c7843804f44afc464e1524fa92b : SUCCESS in 8m 39s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/3b0c10aff72b4ffca23628dcb397b1dd : SUCCESS in 32m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/87d19d98e90e40278411f582f8282d6e : SUCCESS in 22m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ea5ec9a3dd28495eab79a7dc9f59f7e9 : SUCCESS in 14m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d5f55c399ba847ed9f1def595d046878 : SUCCESS in 15m 05s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0ac1366eeb5a47b1a2fee239428d8e3f : SUCCESS in 13m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/83c4d1b9a41e4f3fadb90040461dbcbf : SUCCESS in 11m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/62a8d42ef851499d812a5bbf3bfee697 : SUCCESS in 13m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d52cefa10ed44adba069515168fd332 : SUCCESS in 18m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf73dc8e53264ac68f80eae9d904856b : SUCCESS in 4m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1c0b00e249294078936f4131523dc14e : SUCCESS in 11m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9f38f1cf025d41f08e23e3e5fbda76e4 : SUCCESS in 11m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/32c808c060dd490497532c5878fe5e3b : SUCCESS in 1h 09m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dc1aa7cd6571460b8f93ea647f06d585 : SUCCESS in 1h 32m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f80a004e8d144f379703b442d12b1698 : SUCCESS in 1h 01m 02s","accounts_in_message":[],"_revision_number":2},{"id":"5df18647ccbcd3c8b432da09357119829cb1b72a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-25 02:45:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e0168b450dd7d286507046e45f3881f9ac1ba8e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 03:30:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c8fd4355939931c769894da62dc10bb095e1b78f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 09:04:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86b42edfb2e341f9b232ea63b986d9a9 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2f1ca4d7fa03459ea3317b3094536f23 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a410e87516fa4462872fdf8bf89e4426 : SUCCESS in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b544fcdd0d1486991cdf5b157bfaa4b : SUCCESS in 7m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/035fa5cae7a548c080eafe4dd34431db : SUCCESS in 8m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a95bad3a06884b189f94ae8c0f8b7ba7 : SUCCESS in 8m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/006a428df60e4207ad15036673774675 : SUCCESS in 12m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bf17aa3308ad4d95bde875c1ec648842 : SUCCESS in 15m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0eb5c7393a3840c48ab6581eb20cf29c : SUCCESS in 18m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/899bfd30bda44ba9bb5871830432b681 : SUCCESS in 14m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d03c510d2f5340efa0a2ce6f8c4ce515 : SUCCESS in 7m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/32846cb7da344a1ab4cbb3650251896d : SUCCESS in 9m 31s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d80520a52f9d40789349c07ac68e3ecc : SUCCESS in 13m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1623f1e8495f4d259427339a66c2d4b4 : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/57fb4e1fa8824745a5d0db94b6954044 : SUCCESS in 9m 40s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9e81e3f741e54fb09f25d5a100ccfb5f : SUCCESS in 15m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/96dac4bb75064cbb8c4df57cf73d413b : SUCCESS in 17m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0e428d723a20498cafe2c84fe8e7deb0 : SUCCESS in 26m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6e1166ae0c5043a491a8ec3487644da6 : SUCCESS in 23m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/73501cbc4a0f4d8c8de9c2656d9a65ca : SUCCESS in 16m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6af403137ed44ebeb1864814dddf46a2 : SUCCESS in 16m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1d0fea23771f4c50ba6b2910d1290d85 : SUCCESS in 11m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55eb61e04d254266b7c8e8d868fd68cc : SUCCESS in 16m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/308f17c33a5946fca6a4613e6f59852a : SUCCESS in 8m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7f01781e82df442ebe3a83dab908e230 : SUCCESS in 12m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e058978d028d43c78465cb4c07710ed5 : SUCCESS in 16m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7792e1d7ab914999b3421b40f2529d71 : SUCCESS in 1h 08m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/72d5ee9c94da4a64a8b451807cd66f8e : SUCCESS in 1h 22m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/30126e04a37149d0b7c584a7ff5807c5 : SUCCESS in 1h 05m 02s","accounts_in_message":[],"_revision_number":2},{"id":"8c5b5828754c30c272b6ddd5a2c2be18d42a79ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 09:04:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2ae1f2abf6cc977d704ec82db1ac4b77bf0a2fe0","revisions":{"fea831b3f00e5e0c4cf41e37f2fd58b4eb9e8639":{"kind":"REWORK","_number":1,"created":"2020-06-15 09:23:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/735534/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/735534/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/735534/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/735534/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/735534/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/735534/1"}}},"commit":{"parents":[{"commit":"7de7a2aa6ad56630f1a1ff950d0c597e983ba37a","subject":"update constraint for horizon to new release 16.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7de7a2aa6ad56630f1a1ff950d0c597e983ba37a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 09:23:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 09:23:29.000000000","tz":0},"subject":"update constraint for python-neutronclient to new release 6.14.1","message":"update constraint for python-neutronclient to new release 6.14.1\n\nChange-Id: I0ef8f8d83bf22a00b90087ec6ff9c424a24ddd20\nmeta:version: 6.14.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I61b997ec76114ab9f41a1b5a32d36f1070307b46\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fea831b3f00e5e0c4cf41e37f2fd58b4eb9e8639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fea831b3f00e5e0c4cf41e37f2fd58b4eb9e8639"}]},"branch":"refs/heads/stable/train"},"2ae1f2abf6cc977d704ec82db1ac4b77bf0a2fe0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-24 20:31:44.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/34/735534/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/735534/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/735534/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/735534/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/735534/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/735534/2"}}},"commit":{"parents":[{"commit":"5996ae1bd1fa68b8985b49c4a07380d9b723dd4b","subject":"Switch node tests to inherit nodejs-run-test-browser.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5996ae1bd1fa68b8985b49c4a07380d9b723dd4b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 09:23:29.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-24 20:31:44.000000000","tz":0},"subject":"update constraint for python-neutronclient to new release 6.14.1","message":"update constraint for python-neutronclient to new release 6.14.1\n\nChange-Id: I0ef8f8d83bf22a00b90087ec6ff9c424a24ddd20\nmeta:version: 6.14.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I61b997ec76114ab9f41a1b5a32d36f1070307b46\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ae1f2abf6cc977d704ec82db1ac4b77bf0a2fe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2ae1f2abf6cc977d704ec82db1ac4b77bf0a2fe0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
