)]}'
{"id":"openstack%2Frequirements~735537","triplet_id":"openstack%2Frequirements~stable%2Ftrain~If4dfb79305df0266f078612788b754132b0f6651","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"If4dfb79305df0266f078612788b754132b0f6651","subject":"update constraint for python-novaclient to new release 15.1.1","status":"MERGED","created":"2020-06-15 09:26:31.000000000","updated":"2020-06-25 08:55:09.000000000","submitted":"2020-06-25 08:55:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735537-1593075309151-5e2ae216","meta_rev_id":"7b39afd132fe03641bf3f03bda62845375f56c06","_number":735537,"virtual_id_number":735537,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 08:55:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-24 20:31:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-25 00:50:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-24 20:31:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-25 00:50:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 16:29:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-22 15:58:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-25 00:50:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-25 08:55:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25d92d03a380962f75cefb46d775deb1647cd8c3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-15 09:26:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f5d159c8584c55d98fcc5d04c12718482280cc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 09:59:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b290138bb72049d591632ea713fd4308 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93cd928ce8c64a20a02f0f223a386fe3 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d93fedaf4bd4455b73cd272cc0df6f2 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4640d0b293f4463eadb05564429867d6 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d2b04e80ade4c39acf6eba8a806a195 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbb5e10de2b94dd695fdfbcc077da724 : SUCCESS in 3m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/26a37afd9f1c4de4a7f45b321f6bdb8a : SUCCESS in 8m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/de982361087f4ac0adb882becd85eaee : SUCCESS in 8m 57s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/804580a66974448c9452cded96f9ccf9 : SUCCESS in 11m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/13dbb52b64b64a1395fe8f585ef3c845 : SUCCESS in 9m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c47ed693962d4f0f8e7c7b6f67493bac : SUCCESS in 5m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0b3fe269e2974e8ca09cf7b3f72e1ff5 : SUCCESS in 6m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a5ccf844c6c84e699ff0dc5f97acb029 : SUCCESS in 11m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/599a86716cf54f9baac4183ef944274f : SUCCESS in 7m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/987eb89f902346e7a28f64f56f2a36f6 : SUCCESS in 7m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7a166c5b0eeb4219955761a0fbf2f777 : SUCCESS in 15m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6f175197690945af886e2581485c76f0 : SUCCESS in 13m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f5fdb3313b404e19801e93e03b6f8cf2 : SUCCESS in 26m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a5a2606f81724abe87dba22059c33cb9 : SUCCESS in 23m 07s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1fc7e00734704096a5fc7b2fe6e080e6 : SUCCESS in 14m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bc6c846a30db42fbbd6ad4375ad69f20 : SUCCESS in 11m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/39a9a9967f8649d6ada5a2f73a0ace44 : SUCCESS in 15m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0f634646b11344b38a9b2ebc439ed3fd : SUCCESS in 20m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ab4e4ec883f462db24b6ce9c9d5963e : SUCCESS in 5m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ed469b931c3344aa8b4fc4254fc03124 : SUCCESS in 8m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/14889f48ae7c495794c3a1747539bf9b : SUCCESS in 10m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/70bbb621f9e0481b830b48b28bd2ca1d : FAILURE in 19m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2af3c6ffcfcf48298c2d547b6a8fbe6d : FAILURE in 15m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/38c588ba6f3d4f95968f713130b95b6a : FAILURE in 18m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3800712d85ae7ed0474177486fd7acfc64f282e2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-15 16:19:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7a85c411ca49e1897f75d2a66d6eabeec25d643a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-15 16:29:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"072ff30c705f8b4ea78ed4682a48b6223ec3cc18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:13:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bc9e3ffc81348bb9e8513f50581e2b8 : SUCCESS in 8m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/edb6ee534f1a47fabd616b9eef19a941 : SUCCESS in 9m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebda29d2e0c744d8a301dad0c9dd2c1c : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/886e0bdcedf34ba3a249a647a05e7fa0 : SUCCESS in 9m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/306640a878ea4cf5a6147e8d0e34a563 : SUCCESS in 6m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/46ad927b571c4a09ae31d12973ed05ac : SUCCESS in 6m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9d965c422da64e689142b9de191e0485 : SUCCESS in 11m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/96bdfb8cf3c74866be3d8c82c1e5185e : SUCCESS in 11m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/746e75385ae94dca859ad98abe356a85 : SUCCESS in 12m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4dd7494f5e4145dbb07392cfb3df8191 : SUCCESS in 11m 27s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d025da62941d46909c1df07ac97cf828 : SUCCESS in 8m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d509bf6203a94bb1922bc2ea79d4806c : SUCCESS in 8m 31s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8c953301c2804d7083305bd220dd3981 : SUCCESS in 16m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/54b31b2cad1d4d5cb7f5824026ecd170 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0cb37ebcc6544a0b06913d6de8cfa99 : SUCCESS in 12m 25s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/aefe10d0927e49078d462e153d1122b6 : FAILURE in 21m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f9b1a08abcf94de9ac1eaaa973397e8f : SUCCESS in 21m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/034a922b25c6429ab20d8fed07147544 : SUCCESS in 25m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cef83a1612f041e88ac7bf7a56acfccb : SUCCESS in 18m 48s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/22f15ded256a4cc6a04df40500f206fe : SUCCESS in 13m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/74872492e53e414dae0f5a4db7c18bf8 : SUCCESS in 13m 17s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1951cf209c804ecaa40cf831cb142883 : SUCCESS in 16m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5771e631da1648c9bf194edab8b20e90 : SUCCESS in 24m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/62fe6dcd266d4b66bc76cc01d6a2db26 : SUCCESS in 10m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1983dd4de7b84220b3a942209e86e15b : SUCCESS in 13m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/236df2ff49954e75950bf98985d873a0 : SUCCESS in 13m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e1c99db11d5f44ba93ef0601a066301a : FAILURE in 28m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/58a216938c5c4fb9b96d4a25763a1ba8 : FAILURE in 18m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c346c884c8b44d09a86a19e6ba894f7a : FAILURE in 18m 51s","accounts_in_message":[],"_revision_number":1},{"id":"65bab6ce5e3a867c63c10c5268859e9e696a2401","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-15 19:42:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ddf678cbbea5cb76cde372f7e15a511913dbdf3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 22:09:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d68c3d2defa047f9ad1982ea001e1ed9 : SUCCESS in 7m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6bc811ffc2f8417d822a2bd67cba1bba : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/913e96d0e27446be8c390f1483b540a8 : SUCCESS in 9m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/639f3f75560f42aa9c1809b3891c8c61 : SUCCESS in 9m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/78d33b2bbcae4cb3836313c04b5e5a02 : SUCCESS in 7m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/857924ebca084cf7ad5281c411195b99 : SUCCESS in 8m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6dc89f08a1294c9784756896f83ab7b5 : SUCCESS in 11m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a584d3b1548d43ca880195af57340602 : SUCCESS in 11m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/43ef128bd6fc473ca0d2a16b4fb0a4b5 : SUCCESS in 13m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/226c1eec5f464244bf752d08c7380abd : SUCCESS in 13m 34s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2be1c627e8954f29ad40d1a0f0bb4f61 : SUCCESS in 10m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e36ce7a08ea3433da510eed5298d9e67 : SUCCESS in 8m 31s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/030690dde2134a31a347dc4bfa0ac0ff : SUCCESS in 13m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b40e9f668bc6448cb1f26462ca166bc4 : SUCCESS in 12m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/45ce8a1d551e4b5b84815309ff0fbe98 : SUCCESS in 13m 36s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ed2ad3bd1f204d1486f286e8f8acc36b : SUCCESS in 47m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ef6907f54add403394a576376c441bdb : SUCCESS in 49m 31s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a6d0fc14b6e141a7b669c511ed525429 : SUCCESS in 28m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3148caf77eaf4e598c87ad0de5b218f5 : SUCCESS in 18m 25s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5231341f23ad4d9aa04043d63f943c6f : SUCCESS in 18m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fef6cb47224c4714a172e5de43fb664d : SUCCESS in 19m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/27275e85bb18420683c89a3c04be189d : SUCCESS in 20m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5741d562a3474f0ba13049bebeb704bc : SUCCESS in 21m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f034a3097a34eaf952703fc8eec3f8e : SUCCESS in 11m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/855cda563f2749e5a5689d974790af90 : SUCCESS in 13m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ba05a516e598411694934d8e227e5037 : SUCCESS in 13m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe8f7bcf45af4e30a54ba9c7d5b94394 : FAILURE in 24m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c9fdfd6964c449c48403a731f9c71d6f : FAILURE in 17m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a58d19a7bbe745598c4eb9a6b94ae56b : FAILURE in 22m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b0ca194a11570de28a7e496b380a02af96d7d7eb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 15:47:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8fe8060e1320c440c57caa5b8732d94545c36a53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 22:58:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d933cb17fa645b9bb636babccabdb7a : SUCCESS in 5m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e6b184570874737aed4a2f39ae66c58 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a83879378a784740bb01f115e0edf260 : SUCCESS in 8m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8973b7ee618646a8bd5d960441f45760 : SUCCESS in 11m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/78c50022566a4af481aca40ca633372d : SUCCESS in 10m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/507d1d53966349d18e30cb86cd54ce25 : SUCCESS in 7m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b9e052baf2874afd8f69031a13c6c7e1 : SUCCESS in 11m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d93b70e877ff489893569e849e3e8dfd : SUCCESS in 13m 27s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f90be2cfd0c7471c99f6b47175969339 : SUCCESS in 14m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2627fcedb18a442cb8737e42bec0341a : SUCCESS in 13m 22s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/672f969a21e042f6b13b793ae4b22cb6 : SUCCESS in 9m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8bcf9758353b4881a0d9946ad06abde5 : SUCCESS in 8m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8da32862fbdc4b0ea42037c202823d88 : SUCCESS in 16m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/67d5e36939c043869d26a029edd45317 : SUCCESS in 14m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/15397efce1fc4743bfeb858a722ed336 : SUCCESS in 11m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0a4143e08e984001b235cb97ea93cf5e : SUCCESS in 21m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c541821804ad470a9af2eb0584e4e06b : SUCCESS in 18m 29s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5d80ebcf8bc147b9b237af17b2375886 : SUCCESS in 20m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c5166edc8e8d467189b130d0402bb947 : SUCCESS in 18m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4ac07c3532524b978168495b5a5e4e8d : SUCCESS in 18m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9bba90ebe09b4c959f1ace14826ba5ce : SUCCESS in 12m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1aa6f5514d40427b8a1dca29b27de807 : SUCCESS in 19m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b4b77fe3bf94a398b4113b6478d993f : SUCCESS in 23m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d0e31fff72c940078a06e0b4fe365f1e : SUCCESS in 11m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1a704584f08c40d5a3fc5f2d2b34ef57 : SUCCESS in 12m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/393d6bcd28c44e1a9f61b9125396809e : SUCCESS in 12m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ad665a1100574ee89f96dc9d8c8978a4 : SUCCESS in 1h 20m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9baaa91dd097444c9a50752184f91f4f : SUCCESS in 1h 38m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/761faa5ee1f14e51a89991d0c769a9fb : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f8bebcfa9470e4974c694e42721e05f8cb0be3ad","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 23:36:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3c84510d4797d4119b438e7c9da02a2024483696","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:36:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"66cbc6c632cdabe0e22dac7dab52f0788a81f195","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 01:19:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90062f140d80488a91fcc080538e159c : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fde8309a6e304cbb92531737e24e7fd5 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e59fa98e1c5c44caa686fb68f49b909e : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a97b1ce2627483dab04e15a9b8afa30 : SUCCESS in 5m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2ba68ba0a0745f1b1901c2b8c4aaee7 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f9a4c68bd2034bf4918dd984858bced3 : SUCCESS in 4m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e141adbdea574459be8b60012aa7e259 : SUCCESS in 10m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db02793825f14c5083031ffac141d4f5 : SUCCESS in 10m 38s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/55afab1402274920b3d8fe5175dc5a26 : SUCCESS in 10m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/77d10b4371b6415cbcbfde01fd2c5698 : SUCCESS in 10m 33s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8435893f7947458eae8f39bb09219f71 : SUCCESS in 8m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f1b33c28291a4d749c4d7fffccc0f4e1 : SUCCESS in 7m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3bf07e5b4b8741b19d88e2bc54881b14 : SUCCESS in 10m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b60919d913f648718ecb2effba56689e : SUCCESS in 13m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4369ca29c8ab4b219d3242f57e64f43c : SUCCESS in 8m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f7bd32c3b0244ebead50b3cc34c25c01 : SUCCESS in 15m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f959e2922570426ca8c8bd6819e8b26c : SUCCESS in 21m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8a855cedd4b94e6fa42d2549deb2d628 : SUCCESS in 22m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c9df097f2dde4580a3c0da6bdc953317 : SUCCESS in 27m 48s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6b51b38c015d4175b7a368b440880d04 : SUCCESS in 14m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4d6df4e7557247af9e78bf06513d11f4 : SUCCESS in 21m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/919f5ed3bb5f4f28bea82db634039f58 : SUCCESS in 13m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/210cd30ca7424dee9df36f7359d2a0a5 : SUCCESS in 21m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/054ce2face794c86a220f87469e66487 : SUCCESS in 7m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3067847f23434f94a41c32f41aac0658 : SUCCESS in 17m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c11ba68d94894722883d0593a9c54de0 : SUCCESS in 12m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/743c467adffe4d808c438858038ddcf4 : FAILURE in 59m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2f5a12490e824a13bfc2700a4c361459 : SUCCESS in 1h 35m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dcd1e61e4fd548d99c7f4b672b1c3158 : SUCCESS in 1h 09m 40s","accounts_in_message":[],"_revision_number":1},{"id":"bb65880d7b8cab23b26c8fe84c2fffd6981144b6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-18 02:09:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"45078c64f1f12075a6f0217beafebe70a7157368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 03:36:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e9ae9a448b646c3ae801313594fe1ed : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b8ac095619544ce6989b28f79f37bab6 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d348f2f4a337459db63dc324a64d01c0 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34ce07a340ff402083f18c392c9d28af : SUCCESS in 6m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/270a23595db1421abb0fc03cb56a2d49 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d8611ce8101a40d3a26da4b13b34cbc5 : SUCCESS in 4m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f1f62d404f6d496aa9d1b6d6801a24ad : SUCCESS in 10m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ebb8942b73504c5ca1ba326b1d4fc7af : SUCCESS in 8m 34s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/608d104563c54e44a8822c252e3f7e7d : SUCCESS in 10m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9507f9a09d7341d0860ae4e3d583ccfd : SUCCESS in 11m 47s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/dc97fb8473344458a5bf6a881d43dc6e : SUCCESS in 6m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d392e0438ab84aedbf60e5a4034fbd8a : SUCCESS in 5m 13s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d24ea17176a34995a04006676a16bb09 : SUCCESS in 10m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0a036814cb0d40b6b8e88c3eef238ca4 : SUCCESS in 9m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c7439bd313e44c439027d624bb04c88b : SUCCESS in 7m 28s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a52bd058444c41c783eee77070d3674b : SUCCESS in 20m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/53db869f43a44ca8bc912bd061c30d1e : SUCCESS in 12m 56s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1f826a1096a648c3961ed7620d8e75ca : SUCCESS in 25m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1842bff3fb544b3dad4d6b5f424123cb : SUCCESS in 14m 00s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/522af63900334ee59a8c15ea6a90426c : SUCCESS in 26m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7edc88daf0ed4583afb38d483c9ecc20 : SUCCESS in 18m 57s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/39d05a906ecc4726b66cb4000c7788e1 : SUCCESS in 14m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/075509bf93054034a9a8361e85f3bea0 : SUCCESS in 14m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d72eef0b24174e7886684543d9a4474e : SUCCESS in 9m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dabd6bd04f724f7297e39ff0e735b873 : SUCCESS in 12m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ad86dd17adcb4b6197df420969c40b79 : SUCCESS in 8m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e33940a7b20543f5b075389d5240a8d5 : FAILURE in 1h 02m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/38006311806144a4ac32e6332cc79227 : SUCCESS in 1h 25m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/da6628fbbbf64584819a9981ec26393f : SUCCESS in 43m 43s","accounts_in_message":[],"_revision_number":1},{"id":"33ecbe21b78a0847194f6dbe5e41274e4f5e9e3a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-19 02:39:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4a89256a99385f10cf79d9a8399ae77404aa4f18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 04:31:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/775feca7660d42d8ae336bf6675c8c51 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec2f38eddbd646e69d98e654c7c59bc4 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c3985c4eeaa4797b60ebe37de8d32c3 : SUCCESS in 9m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f95a8462cec475dbef6033db44ac83f : SUCCESS in 6m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/565fd65f1b954fccb07f6b1ff951fc8e : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/37457ca4ac484295a481a175bd661bfa : SUCCESS in 4m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2474fb5a43504601880522f8e129e280 : SUCCESS in 13m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/92f2a8018b31478f883e8bf9077925ec : SUCCESS in 11m 17s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5fda4c592f6e45239793df371f292af5 : SUCCESS in 12m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/92e53067aadb480380ca9109ef5dfd84 : SUCCESS in 10m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/5a9c7365a2d54fb68d25ee6ec42f3a81 : SUCCESS in 5m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b81170ad2aa5478396f4e6fbc96cafd5 : SUCCESS in 7m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/944a98d1a50f4cf8826cbe1c3372008f : SUCCESS in 11m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/32d9d0cf892d404f9bb532d71e635ebb : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bed8ef89627c43a7b1b22c898080cc21 : RETRY_LIMIT in 4m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/937d9ca5b5524b6e8a147e15e916a9b7 : SUCCESS in 34m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6a1ec6b6f70841d582da2dee17eb963a : SUCCESS in 15m 40s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d95fd968142e4e0daada9f7d3b61c33b : SUCCESS in 21m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/68c7b57da9ee4e24826adb36d65111bb : SUCCESS in 26m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7ec31133df83438c8bcf66ba4f118beb : SUCCESS in 14m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ec33d866cdb947a4b9a4644aa83a2266 : SUCCESS in 20m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c16466bcbf4541a18daef19a0702e06d : SUCCESS in 15m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/161a080d57a04138a15e7088b17ea1d0 : SUCCESS in 22m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67d8590d03134f668f402f97ea973b7c : SUCCESS in 6m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1e04fa39ae5941d9880dc03107d69a97 : SUCCESS in 9m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3c40762031a14ed4bf51dd5d8eeeb019 : SUCCESS in 15m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d75c39ab5eb451084085a689bd88695 : FAILURE in 1h 04m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2620b79b5579433388296a25f0dff92c : SUCCESS in 1h 46m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a7ee743f383c4f129c9fb91ca8e554e1 : SUCCESS in 54m 57s","accounts_in_message":[],"_revision_number":1},{"id":"84121a96420c284a937246187b8e3b0e793c3add","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-22 15:58:35.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eb0cf92abd5ade762a84d6252ae701f2cc2e6009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 20:05:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4846124da9d244a2b6a7cf1e3095735e : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70a5e4a9792b4b31b2518d371f51aa06 : SUCCESS in 9m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9038db49eac54ca5af1fc30d56d082d7 : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78d6227ee12c4783b8a9f5fd142b44c5 : SUCCESS in 8m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70011a0db1394c09927073e6aa37b356 : SUCCESS in 9m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/13db2645796e4e068eb9152e3a816a51 : SUCCESS in 5m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/56182af9fc234e1193c329ad7ac5c8c1 : SUCCESS in 9m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2eace4a314b942d68ce50c246b776afa : SUCCESS in 11m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5a903ae524c049a1826fa8fb4b2326d4 : SUCCESS in 13m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a1686f84ed084b34a9a30e4fea623cae : SUCCESS in 12m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/848e1000253040ba990ed0394c5187f1 : SUCCESS in 7m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/637a06b8d7304a0b8efa359cb3aa6976 : SUCCESS in 7m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/375924424f61452fab5d5499afd951c1 : SUCCESS in 11m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/34b7d941cb954231a0a8405e387efbdb : SUCCESS in 15m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c4d9ced6c1a94998ba4f113c0cf28e3b : RETRY_LIMIT in 7m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c176a8b545b94cd58b41945ee1503b7c : SUCCESS in 19m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6308f7f8d2764cedb4df7266e34896ea : SUCCESS in 20m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c92570707c0c4952ba8cc5fe10b13f8e : SUCCESS in 28m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2aea84fa25e144b2b3d4532deb1a15ba : SUCCESS in 21m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ef932b9678bb4409b1c924bb10bf775d : SUCCESS in 16m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3ac93b4e7d7e4e02acbc5aa70523fc12 : SUCCESS in 14m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3d147426b149484c9398a527299179f9 : SUCCESS in 14m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/60a576bf0cdc4c33b2f62d82ec810b35 : SUCCESS in 25m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4a50fe3a51774288a9736f3155b804c6 : SUCCESS in 6m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d612e9051b524b50bbbe690d60370ac6 : SUCCESS in 18m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/66a9aeca990d4d80a94c4c2665400caa : SUCCESS in 9m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e05283a316145c7b1a0ec5e2ca715b5 : SUCCESS in 1h 16m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/025a789e8a9e4bc09ee083c25a3cd37d : SUCCESS in 1h 35m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/42740692b6e948808a413d2138e4ac2c : SUCCESS in 1h 00m 18s","accounts_in_message":[],"_revision_number":1},{"id":"16513edd7f765b8d471be0fb6e1e561ae243e5c8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 16:00:30.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c1040e49fd14bf7ab292f2ba3dd8f59383c019e7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 20:31:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3cb3b5e19c9b1978d2ebd8106c5b5ec7dcf6bcc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:22:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db89cff951ff4e74bd00d2db4ec89106 : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a0c2ad10ba5c4134952c58eb30391746 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1cc1b79fe554b3daad6984e9c90be68 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16463b268b4e427ca19f02a0d6ddde11 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/64da52e9f7964175b826aebb9b665074 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fdfa655f9bcf4929ac49e625b4e2007f : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5cce50fb4e3942ddaa899fe176beabe7 : SUCCESS in 11m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3464b3a9e7024e4bab802d985ff115ac : SUCCESS in 10m 11s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/95009880133a4b2a967d80945c7e52b6 : SUCCESS in 10m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/03fb884b6ee64ca1a75738a2b83a9301 : SUCCESS in 7m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a6fa7e26bf1f4d079b278c7f9ce5fe66 : SUCCESS in 8m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/09c16ec9724f4bf9a1dd3cd41784614b : SUCCESS in 5m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3d07171cd1dc470ca179e3b60a68f310 : SUCCESS in 10m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c9fabb6d3f3c475db4fae8b9d274e15d : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e21909468cdb4f4ea59582eb383453ac : SUCCESS in 8m 14s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cce82d7d9b2c4933bd94ff81aa474e33 : SUCCESS in 14m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e7b8ebc0bd3c40169a6a96a8d5257422 : SUCCESS in 13m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/22cf2a7828b24beb87b61a7fcd3fa456 : SUCCESS in 16m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/897ca400de1c459c8bb0b9f6264cb50f : SUCCESS in 22m 16s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e711f5d09e64425d8fc39e9af6df7b93 : SUCCESS in 17m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0f97cb89471449498ee922973d471d81 : SUCCESS in 14m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f5b6928fc85140e5836a5c8ca517b6dc : SUCCESS in 16m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/15e70a8d1f2e46c49b74b487d4c2ae8a : SUCCESS in 19m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/132461ab80a84184ab24a967d5b0e311 : SUCCESS in 6m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ec37daacabbe48bf93bf5c602873bfbd : SUCCESS in 10m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3ace0a8d315c4f1498fd4bc92ccc3edb : SUCCESS in 9m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b7f6baecc064abd8d291318bba78f2e : SUCCESS in 1h 08m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a08b1aa51b3341e0800f23a10710b7cc : SUCCESS in 1h 43m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d3bf3a5dba7d42d7aeba93569403e5e3 : SUCCESS in 1h 10m 03s","accounts_in_message":[],"_revision_number":2},{"id":"411cc167658779ecabfee242069f915017223f4d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-25 00:50:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a4bea2e1a94ae0f7a2e20cd973c801d7d96ec365","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:51:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d015b49ffb411489491cdfdfbc2746ed023cc84f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:55:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fca10afc0104a859c932708e7c64bf2 : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/921151c06ccb4676b018b1699c5d666b : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14a1835075c94907ab12797551db6a6f : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45c784e6a283460e9de032e32161da5f : SUCCESS in 9m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bad7a08118b14e7ebd95177af478aa56 : SUCCESS in 8m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8dee65966bfc44b8b70cd75f1a53748e : SUCCESS in 6m 30s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a804b3acddfa4ca597d43af9fa904c99 : SUCCESS in 10m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2b51d15b0b1d4ec2b8cf872ad79df771 : SUCCESS in 12m 03s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5bf29104f5c9416795159ccb46215e8a : SUCCESS in 14m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dd005c4f79ec4722bd7a95b333cb78f4 : SUCCESS in 13m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f727ddfecd1e476a84798f2fdc57e1ce : SUCCESS in 8m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/26fb55abbbf14d23b16158ab0eea38f8 : SUCCESS in 8m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e24916dcbfd64d4b99a546b587baea0a : SUCCESS in 14m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0ed45ccf62f247d4bfec75f92bdb376e : SUCCESS in 12m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87f340372ec8475589315d4f941f6578 : SUCCESS in 12m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cab563a8d63a4457a51ce8de25d7cdec : SUCCESS in 14m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fff3118f0c3f423ebba7b417df15bfd8 : SUCCESS in 15m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/ddef87cb44fd4f6ba954a0149a105740 : SUCCESS in 27m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a680bdb0d9934486ae9ec1ce0d4c5607 : SUCCESS in 27m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1752f6d193f0444cb7fdcabd19ea2741 : SUCCESS in 19m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3e2ebe722ab841cab0beb58c16d67785 : SUCCESS in 19m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9a88318c27564481be232289cc0b6771 : SUCCESS in 21m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95239a2abbb64b478269506ad5cf44c9 : SUCCESS in 23m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a8aefa7ca934854b1ee7fee642c6cba : SUCCESS in 8m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4533fb9a9a5a426dbfef80bb05cb3f20 : SUCCESS in 19m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a109d9fc5054421aac5c5a9534ce12c1 : SUCCESS in 18m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7aaeed74336e4a0296a34e3c1cebd853 : SUCCESS in 1h 10m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e18920310c364eb9a182f2dabdc15195 : SUCCESS in 2h 01m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/704e0b2ff26946779f22e5131ca1c3e7 : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7a0f231b600ae8016aeac0f33c330f794b814ad0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:55:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"264440d39795427e03741c881e782e63f5803818","revisions":{"017048aaad5ec3610d85c16eae91a30fb0871c90":{"kind":"REWORK","_number":1,"created":"2020-06-15 09:26:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/735537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/735537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/735537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/735537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/735537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/735537/1"}}},"commit":{"parents":[{"commit":"7de7a2aa6ad56630f1a1ff950d0c597e983ba37a","subject":"update constraint for horizon to new release 16.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7de7a2aa6ad56630f1a1ff950d0c597e983ba37a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 09:26:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 09:26:28.000000000","tz":0},"subject":"update constraint for python-novaclient to new release 15.1.1","message":"update constraint for python-novaclient to new release 15.1.1\n\nChange-Id: If4dfb79305df0266f078612788b754132b0f6651\nmeta:version: 15.1.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ie84d8b4ee79b7b1345028faf0d283fd6bc60ac3b\nmeta:release:Code-Review+1: Balazs Gibizer \u003cbalazs.gibizer@est.tech\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/017048aaad5ec3610d85c16eae91a30fb0871c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/017048aaad5ec3610d85c16eae91a30fb0871c90"}]},"branch":"refs/heads/stable/train"},"264440d39795427e03741c881e782e63f5803818":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-24 20:31:38.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/37/735537/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/735537/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/735537/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/735537/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/735537/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/735537/2"}}},"commit":{"parents":[{"commit":"5996ae1bd1fa68b8985b49c4a07380d9b723dd4b","subject":"Switch node tests to inherit nodejs-run-test-browser.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5996ae1bd1fa68b8985b49c4a07380d9b723dd4b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 09:26:28.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-24 20:31:38.000000000","tz":0},"subject":"update constraint for python-novaclient to new release 15.1.1","message":"update constraint for python-novaclient to new release 15.1.1\n\nChange-Id: If4dfb79305df0266f078612788b754132b0f6651\nmeta:version: 15.1.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ie84d8b4ee79b7b1345028faf0d283fd6bc60ac3b\nmeta:release:Code-Review+1: Balazs Gibizer \u003cbalazs.gibizer@est.tech\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/264440d39795427e03741c881e782e63f5803818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/264440d39795427e03741c881e782e63f5803818"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
