)]}'
{"id":"openstack%2Frequirements~735543","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I921e67e159f062d72e88dcc9cfb58a9ed9841ac7","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I921e67e159f062d72e88dcc9cfb58a9ed9841ac7","subject":"update constraint for os-win to new release 4.3.3","status":"MERGED","created":"2020-06-15 09:39:13.000000000","updated":"2020-07-11 20:56:05.000000000","submitted":"2020-07-11 20:56:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735543-1594500965727-72fcc681","meta_rev_id":"c6b5814fa1260deae93245685af892ca7c26f9a2","_number":735543,"virtual_id_number":735543,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-11 20:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-24 20:20:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-24 20:20:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-24 20:20:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-24 20:20:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 23:36:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-22 15:58:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-24 20:20:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-11 20:56:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"145fd8b94ca992b54354a15fdc08c4db5ccd14cc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-15 09:39:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"608e51942effac97951c6165444c4ea8899b6f0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 10:08:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4693453cb71499fab240f796b184718 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bc627fe6928c47038f2cc86860eefadf : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5104ceae5f6f4bc8840a0f7f71c1058d : SUCCESS in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9809d89c4e094f88b45a184d9e6f7653 : SUCCESS in 6m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/234b5be07a7341759848c465800d300e : SUCCESS in 6m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c909429426846f28cf7d3a1d133edc0 : SUCCESS in 4m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9cf96d0fce734eddbd184332862bcc72 : SUCCESS in 10m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/66a8274131194e5092d627699debaa8e : SUCCESS in 7m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/129645f822b74d9b803ecb75093eb305 : SUCCESS in 11m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/172f6d6e813d4196a9d396bf1e2acc23 : SUCCESS in 11m 55s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a74b97dbb27c4c8bb43d2d4ebee18042 : SUCCESS in 7m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cfb22444e6dc4eea8f5ed4f5a620a832 : SUCCESS in 8m 29s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/65ea5c08c3ef420494322dcd0f62cd31 : SUCCESS in 11m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d2cee6811acd44dd8b7f5a380d5417e3 : SUCCESS in 11m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e513a1724f446b0b83e128acc23de70 : SUCCESS in 10m 40s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cdc9f1970298494485da18040311f7db : SUCCESS in 19m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f3cb36ab738b4822b57019cec04fa769 : FAILURE in 15m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5fd7409777014c7abe72ac63033c675c : SUCCESS in 15m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/61520216e76c43a8b26e4986f74a2d51 : SUCCESS in 23m 51s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/cb54be056de94efd8ce827a4d2f2b35b : SUCCESS in 16m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e9b63ea64d424345aa752041a888e3e6 : SUCCESS in 16m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2f1badbec3e64caeaf7c58a46c52df63 : SUCCESS in 14m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ca49570f0554deda51bf5d1d2bcebb4 : SUCCESS in 22m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee618d70defb42529d26a7cb17b991d3 : SUCCESS in 6m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a911274112be4163a20a0a333e0b28df : SUCCESS in 10m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0f125170887f4595be23e166f102541f : SUCCESS in 8m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5bc31b05d95f430f9c8a69aac5632962 : FAILURE in 23m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ca54b7f4f2304efcac71b6662211c39e : FAILURE in 17m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1851be3113c54d4ba8df5852a48e765c : FAILURE in 16m 16s","accounts_in_message":[],"_revision_number":1},{"id":"3d0786d24047f045c457d4d356197fbb86c04b37","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-15 16:19:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d50f05b645a95e6da80afe9dde23e46bd2d6ab0e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-15 16:29:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1de61764f3caf3abbcf4e6d56e178e41a4a1e859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:24:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a7acac1e5b04bf4ab265ce03d529d06 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a743ba774b214d42af9757e3b1bfbeca : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8099322980914ba98d8ba6dcbd6709e0 : SUCCESS in 6m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c40a2287e2547148ed59a21719a2033 : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d37dedfff3da4d778c5e00523024ae5e : SUCCESS in 9m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/14d35f8cb496461097c203173fef558a : SUCCESS in 9m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c1e235d666e440439099dcc91725597e : SUCCESS in 14m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/76c2942c40434fbc95e228f63376c634 : SUCCESS in 13m 48s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1671b3d367924cdb9eb8abd2ebc1096c : SUCCESS in 12m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a31c3b2df71d42ae9cd2ffb2b44ba798 : SUCCESS in 16m 44s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ae20c4622c954cf1af69509956886a87 : SUCCESS in 8m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/adf41229604f48279f14f5ad075aba51 : SUCCESS in 13m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8a8180316a0a497cb35ca959c53b46f1 : SUCCESS in 17m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/66aa9d84bb694318bbf981f10ffced90 : SUCCESS in 16m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/758fe2eb010a41128e153a3b72268079 : SUCCESS in 9m 45s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1a22052e40564a70863c5069d63a8efe : SUCCESS in 24m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dd7c91355626418ba44944a66181c934 : SUCCESS in 40m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/30dd794333554b328833011754e080ab : SUCCESS in 27m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/37571160ffc14f29a2fd04b919603ecf : SUCCESS in 24m 54s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f005a3fca7df467692b1814e2fb51a24 : SUCCESS in 20m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/25952b211d3444a68f5090707a0b0285 : SUCCESS in 17m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d8c42e2474fe4ccba8c9e605bc10280c : SUCCESS in 16m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2344207c46b447ba8c307a6101586ee9 : SUCCESS in 25m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3994752def1b4ff0a61639dc60920c18 : SUCCESS in 9m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/07fe6c25211c4fe8b4dda40fc67f128e : SUCCESS in 11m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bb83f56c5aa74ed18464fde1dd7bd924 : SUCCESS in 14m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f5087dec9bd94880a6ae7f5eb52413bd : FAILURE in 21m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5932155d18244135b787171a6406fd4a : FAILURE in 18m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/503d121348fe4c10b1b3b9039a5e1959 : FAILURE in 16m 24s","accounts_in_message":[],"_revision_number":1},{"id":"f3d1cd38241a8ae29f56cf9f9e4e49c3479d89f3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-15 19:42:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"760933603eaab11d03b2a9c4c44e6fa5d4fbc33c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 22:04:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09e74577f5d488fb1571588e3ecbc37 : SUCCESS in 8m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3d6fd41695946d79ad503000115a553 : SUCCESS in 9m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76479498461740fb82687adef4dce140 : SUCCESS in 9m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a3cef2d596b43c0bfe1d467d42139d9 : SUCCESS in 9m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f336b1852ee34cc192359cffec4a7e09 : SUCCESS in 7m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/08edaad1880345688287e7d630afac27 : SUCCESS in 9m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/dd18a63e987b41a6bfbaee89a1398e61 : SUCCESS in 10m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b5d47a553c6b403192e765fc91ebbc49 : SUCCESS in 12m 04s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/28ea8deae9e44bc7a1ac3a00cb8da21a : SUCCESS in 11m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/54a56383128348968c813439d5593ba3 : SUCCESS in 14m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0cefeb5abc0449748d6f64e293c0e24c : SUCCESS in 9m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/59f6f9e6a3c64bdf908b9014ae88e235 : SUCCESS in 8m 22s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/13efe2c2901e45b3a287d234212e2c6e : SUCCESS in 12m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e944f5c1014d4647abc4d18f3c79a314 : SUCCESS in 11m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e9d61265978a46848fa508eeecbbd7cd : SUCCESS in 9m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ad10f9e6277d4b2ea7bce6524fccb126 : SUCCESS in 25m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1d3441e77fe9435098d61a1dec7d2fe1 : SUCCESS in 25m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/0eb3f9a4312e443cb8d3a5e1bfa42ccd : SUCCESS in 29m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5005f8c4546d4c98a59d3bc4508444b8 : SUCCESS in 23m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/272214e595fb4371bc602349e390f8cc : SUCCESS in 18m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a2214a806b0b43159a9471e3bdf7590b : SUCCESS in 15m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0a8a0a13c0604b699bb709f537d7272c : SUCCESS in 16m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/27fc10c119d646b2b22045839e691cf5 : SUCCESS in 23m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b499380e3984d65b695db1d724e7759 : SUCCESS in 7m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b80bb5be24994d10894cbf0e7cc819aa : SUCCESS in 11m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9e36964f5a614d0c98711fcc54528f51 : SUCCESS in 10m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e21f806c2b6448e83e555b194fdf7d5 : FAILURE in 27m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/48ea461c19d848f1b37fc854b9c64dac : FAILURE in 17m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bf9841d8289c4a13b0f7c49657367fc0 : FAILURE in 20m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5109a0faeadeb3d6459b90bff96903b677a31d7e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 15:47:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"581704e130fdd80f20e17533e66d9c97a9fe6dff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:23:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c0216ec950f43e19a5036cdbfdd7840 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00506af9bb0641b59b5582a72d8d3524 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45e0b9cb1a84413a96031f63fc73bcc0 : SUCCESS in 7m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/006dd4c9759e4f7c8ea849be144906fd : SUCCESS in 10m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b6a62c6932e54ec3b69ec73b82df9cdb : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e8d688a03734cbd89dbb6202bac5895 : SUCCESS in 7m 33s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/86b00e051b2c47569f085a3922f395ed : SUCCESS in 10m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3d10d2af7302488a8e7fba8c8f868f7a : SUCCESS in 10m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a72f8a61afd3438d93625f93cb8cb367 : SUCCESS in 12m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/18c9772b2820432dbc109c59c73ee9fa : SUCCESS in 9m 20s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ff62a3da27e240bcb0468b6241706156 : SUCCESS in 9m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d96a1bf9ae3747d9aaec4108c2045129 : SUCCESS in 11m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c641aad869da447c8ad13f40bd4e1998 : SUCCESS in 14m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6e9e5fe2ed1c4f0893ed923b5848507d : SUCCESS in 14m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b267e74ef914f8ba3fef510e092ea1f : SUCCESS in 13m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/740541b78258489fad6a4afbb07356f4 : SUCCESS in 15m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/130986d42c774f38a4a3a2c881be81e9 : SUCCESS in 17m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/83b2b8f0d1084d80903252d03e27e20e : SUCCESS in 27m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bce1fea70b164b91bf63afcfcd53edd7 : SUCCESS in 24m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/290c5dce114f40e0bbb3358377f218dc : SUCCESS in 17m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/93b8b52f66e649c593a4c894373c1e9c : SUCCESS in 15m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4bf2a67e82b5431d8da3ee4e7a63dcea : SUCCESS in 16m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7929d0f8e5994db891bdc3bafc5b074b : SUCCESS in 20m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab9e08508a8a4f5c9301ec5812ab5408 : SUCCESS in 9m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2381d033db7b4caa8601fcd225386205 : SUCCESS in 13m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ff4086207f8a4f558af8eeb1257e89be : SUCCESS in 10m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a2e132988494733bd51fccdf3b2727f : SUCCESS in 51m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/827b9c9348d644d68f5157aa90c3e2db : SUCCESS in 1h 38m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fc72b309ec264fa591525ccc321b4966 : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7300ccd739f3e4fe126cffc63fd533644b452b79","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 23:36:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3af3d017a1eef0f9f8780c559269277053adc3d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:37:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d310f315e5f26de298f771a74bf2ab7f408327f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 02:18:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ec16e66cdae4319a2b81c6e0cf50422 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8bbafdf4a705429cacdbaf976bf1b85d : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7005a56fd5f14280b3de0e7000fee51d : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0245df64a97348f686e53858525228a3 : SUCCESS in 5m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0f83d8c76ab46d38c7b7382899b3a02 : SUCCESS in 6m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/197c3e1f523b4b3a8ccbe5e3e3593430 : SUCCESS in 8m 16s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/425755f4d0884fb7ad7f0abc2ef9d4be : SUCCESS in 15m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/99595702e692462b8010181cd04362cc : SUCCESS in 8m 40s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3af7a9594ded4c6fa5d179a1f12211a2 : SUCCESS in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a143e881ee834e239094997ffb4a1e35 : SUCCESS in 13m 53s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e95fd959f65044a2aedc56f0bd965693 : SUCCESS in 8m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5d5d9c2268694f728e959145b205033d : SUCCESS in 10m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e655239007fd417cbb81d86aa268b9ac : SUCCESS in 10m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8183819ba9e64d9380a3f589e87ed1eb : SUCCESS in 9m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fe24d1e0ab474f54a453642e1c803235 : SUCCESS in 11m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/af7db52f56a14e0ca688e4028787d2fa : SUCCESS in 17m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0ed1ca53a99248888785ad166ea4a2eb : SUCCESS in 17m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e4cf0bc9983146f0a67a4bd7de8b9b96 : SUCCESS in 21m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/18d3bab376c24354800b81bc8af65ff2 : SUCCESS in 24m 45s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e3b5f3604a4740c7a01c1c14b11a1f8d : SUCCESS in 17m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9a6437cb183343468f75ab198aa6f020 : SUCCESS in 16m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d19ad403d8f5434a8d0000bee7b60b1b : SUCCESS in 16m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4077f3d102b542cfb82204a735f25d78 : SUCCESS in 20m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/daedebfbc312431189347293fc446957 : SUCCESS in 6m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e16f994b4e754bc7aeee37a8b86c8042 : SUCCESS in 12m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/07f706da29f24d7ba9dee78ed381e589 : SUCCESS in 12m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6b68c0f1d9f04c17b45bb763e04ae889 : FAILURE in 1h 01m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8dff5e63928a4ba9bb531062f64e31c7 : SUCCESS in 1h 33m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dc4f03f2b4074514adf83e2a7399bbf8 : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a22314cd7a00152f2a1819e6b3ccd587e059bea3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-18 02:33:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4be83c2ceb2e0826892d82ffe898179de556f517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 04:12:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da5c078ef7df40989466a4b98761fffa : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4cf139251d2430c856f70c910642159 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96a2aa988b6c43b7b3d5f568c26b36e8 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/555448936f644af997f2df5d6b7d6f55 : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85ef606e80ff4eec9b7a7270b4233e85 : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/67269648e8e14e2e86f7da592a0e4362 : SUCCESS in 4m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/297096e3971549aa9c4bf9e50f2f63db : SUCCESS in 8m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e6d1f893461441bf8f9b6b5d20b56b4e : SUCCESS in 9m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c745d3402b2c4cfbaab48f7b0863160e : SUCCESS in 10m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b03bab36683a459aa91ef9e0f8db812c : SUCCESS in 11m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4310decc108547fcb8569085c5639178 : SUCCESS in 7m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/68eb63fd335f49579e9fb6429e592ed3 : SUCCESS in 6m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/796625167b834a2a924787256f22d7a6 : SUCCESS in 10m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0cb10ee75c6442368caefcc1e0c33269 : SUCCESS in 8m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/832a504fc96143d1b5e382c08b31fa87 : RETRY_LIMIT in 5m 34s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/95371191004448bd881279a8b6700fbd : SUCCESS in 33m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fd9e6d990ba34d8db75bed2e99cc0d9e : SUCCESS in 31m 47s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/13ccfbeccd344c0e8de836b1f64d3948 : SUCCESS in 19m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d3781f2611e34747b5a6916b4cdc7f51 : SUCCESS in 22m 57s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/069722c32eaf4f1bb9746fab30665068 : SUCCESS in 22m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1599ffc9ed354bb8bc0a4002bd2f4fd9 : SUCCESS in 9m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/628c36583169415b9730b01e8b6a88a8 : SUCCESS in 13m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2668ae405625459fb546e9b303041422 : SUCCESS in 22m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb11fb7115fc4985915007e23cd68e47 : SUCCESS in 5m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d0a8301c463448ac958f913eb86c213a : SUCCESS in 10m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a42056c3517c4ebbbca0778133a2e72e : SUCCESS in 9m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ff4502248984dfaafe15dca59c3f760 : FAILURE in 59m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fb935b6bcedf438ea468a0bcf6c10405 : SUCCESS in 1h 31m 28s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4c36d0779f52424b9248b922b8758d0c : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c8d8a1bc23ee9b8f6ad80be66486dd2434a8e92b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-19 02:39:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cbcd557df5e2441c6acc92742bad1167940a2566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 04:20:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/565ef39ca0e34e04a0dd30894fdd3563 : SUCCESS in 7m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10c360a694654aeb8e5402b596c5bbf8 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c436d47c346346a69474cf04cf0ad7da : SUCCESS in 8m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bec6a2042daf4fb8ad4937b1649bd431 : SUCCESS in 7m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5229dbe097347f4a1ecac8725877e65 : SUCCESS in 8m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/48e35f3ef4f8423985165839be7b3501 : SUCCESS in 4m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6668f4cac9624ed188fd32cb6926508c : SUCCESS in 7m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7d637e6d9f4a4b87b9fecd76afaa57ce : SUCCESS in 13m 34s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/5a7d59fd7725496eac1b41b03743cb27 : SUCCESS in 16m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/16add1e9a4ab44b2b1a30fadedcfd4d9 : SUCCESS in 13m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/dce329a45144499488bf825799508d44 : SUCCESS in 6m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d6a1fa1b8f464766a858882f2fdd7a50 : SUCCESS in 9m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/308da1550fcc45d2ad1a79ec1f01a1c6 : SUCCESS in 11m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/490d40e48f484dd5a279c94b92a6cad7 : SUCCESS in 7m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fbd492a29c964b9587a2cf4e038245b3 : SUCCESS in 9m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2740ef0ba54f4d96a3b0d7f6670f6190 : SUCCESS in 34m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/43c8457af336459c9d83dad35a949cd0 : SUCCESS in 26m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d365480e26ef499480ffe4af37001210 : SUCCESS in 27m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bad8f84af9cd4da9b34c00196b6cd399 : SUCCESS in 25m 08s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5346ed4b39b143a198db9b71196af41d : SUCCESS in 22m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ad6fa8cd813a4ec6a1e7875aa8b54822 : SUCCESS in 13m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7b6bb1a546c7461e80572534606bfc89 : SUCCESS in 15m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2070eb4ef5cd42de9d13dd41156ba1b3 : SUCCESS in 21m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/63ab7d24a73f4315871d83f2b7c9ef76 : SUCCESS in 9m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5d7c0462be7e46d3aea82dad0155c711 : SUCCESS in 11m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5803c87e33a34996b077302763d94d1b : SUCCESS in 18m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/82851dcfc5de45ec9169ddf4b6ef1813 : FAILURE in 1h 04m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/751760b2f4f642ca95c63bcac9f30625 : SUCCESS in 1h 39m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3f92cfd4e4604f31a9839dc2909324d9 : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c521f828985059adee8c830f95ef513e2d8ba317","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-22 15:58:17.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5505da7ec943e5e277096f8e142c927993a2bf57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 19:53:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/551950d2803a43d0be110021f0fdd677 : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e81f2f4a7ef46f49f12a9ba3259fc98 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08d918a80c8c411dbdcfed6114d198b7 : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bca5aa61fd2d487a95b6ea3871166a6d : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d6b2b0d9da148278662895a0beb6380 : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/36641e078d3f4ce9a1d169f2f035e609 : SUCCESS in 4m 37s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b0cf4782d96d413796618c19f7dae742 : SUCCESS in 10m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8e26c39fd2414c4d932f1c6abfc918da : SUCCESS in 9m 02s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/56c031d41fe648d98b1a4c938b014763 : SUCCESS in 9m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/12840bb787274d139764516f5f923dd2 : SUCCESS in 9m 55s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b290d2ea353c407cba444d35bbe584df : SUCCESS in 9m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c52187533acf402ca6528d7b7af3f100 : SUCCESS in 7m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/87e9a0ad204c4d48b7e61f0071122401 : SUCCESS in 11m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a54a1f2dda9d4c0e8fd69bbee1f9a5d9 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c339c4efd644d29abf63de1473d47e4 : RETRY_LIMIT in 7m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/878b85599aa14b8aad5e97c423a298a7 : SUCCESS in 14m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aeb96cdc82964fd290a494854109bdfb : SUCCESS in 22m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/10ac6641caca4b76a3884c26b2269985 : SUCCESS in 26m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/450a145b585a485c9e5729fe0f1d3783 : SUCCESS in 25m 15s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/43deb58d5681452fbf064cd3abcc595c : SUCCESS in 19m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b9aa3f9b21ed48f2af2f896b4006b87e : SUCCESS in 18m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e25ecd001a034d69b3064f6a82aaea69 : SUCCESS in 18m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a45b4adb3ca3464daec524785d3c6136 : SUCCESS in 23m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d4c0c945fb34289a747439ff7cb877e : SUCCESS in 9m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f42eff359e1545e19b897d56d77071a7 : SUCCESS in 15m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fd6421f13af348cdaace875860372bdb : SUCCESS in 12m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/73f4199a4ba64169aa91d2331bff5ab5 : SUCCESS in 1h 05m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6a22a4b32ffe4ab386d9ea743e2c945a : SUCCESS in 1h 30m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e69f45439e07428fa37b5e09c160ef23 : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":1},{"id":"790a5372ef77ec98f8bf39eac61003f6e6354125","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 15:53:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"26e54cf3100233fea45e19b2f3855e6231e43184","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 20:20:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"46730cd04c4192dc40e1137055f0655d7ac6de49","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 20:20:25.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"60945e20d521b910daf018364a95bcc3d9305cb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 23:15:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35551b877d24cf4bf8e62f98db61a2d : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a58efacaf2c34121817ba98f8ba0b413 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd8899d3366c464684d6c894a372b9e6 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b675b1b408bd4b05a474542d36d4df64 : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a94539412ef43bcba1eeb2d627a9858 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ce2d410fb3749bfa9ca760c727cfd9d : SUCCESS in 4m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e3b0aad2459d4b749b5943b05dad1f22 : SUCCESS in 9m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/40ed5c3651704a8696c3c9c4559c2ac0 : SUCCESS in 8m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ac7fc2f20a344387ae983fcaa7bf9745 : SUCCESS in 14m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/017875f6981c47e99ad53f5063d2c51f : SUCCESS in 9m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c1e0bc5801e9450f84a59545bfd03c9d : SUCCESS in 5m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/258dca1d6a31455a9b03010544635012 : SUCCESS in 7m 30s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/890fbe8795724d3b8ec5c4c9b63ac654 : SUCCESS in 12m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ec4490c02fd94519a22c4229ae3f8511 : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0fdb1861e4d74fa0a57c18a139b24909 : RETRY_LIMIT in 4m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/20659b542dba44258f06e5c89dc66772 : SUCCESS in 23m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d60fb5bab209406d88e9012f0bbcd669 : TIMED_OUT in 1h 02m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/47e0e592fa794927885a0eee3a4891f0 : SUCCESS in 23m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f1494dd3213c4ffda40828598138d1bf : SUCCESS in 28m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1e2545a067854c58b6563cdbcd2bde47 : SUCCESS in 13m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cb03a4d197ee4aaa882b5f86d5df1270 : SUCCESS in 16m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c32f3d1dc2804f14bf959967cbcf9d74 : SUCCESS in 16m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04919e68de294ef0bb5465ff75fb1078 : SUCCESS in 24m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/614efbd87a0748d2a3a1944157cc11aa : SUCCESS in 5m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b9600974351b40db87685a365fa4df0a : SUCCESS in 11m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0d76ca622f4240e0aa586a662495ca13 : SUCCESS in 15m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0dca5cdb8134daa83e677e722946b73 : SUCCESS in 1h 08m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c5906d625a82434da8bda1d6533c3d93 : SUCCESS in 1h 10m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9ac8e19ec91b40b8be875980806247b4 : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":2},{"id":"6a591561e28d0bec2a10c6cb73024fee8004ba7e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-11 17:53:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1095f627cbd9cb693f547ce636202f1ce386bd3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 19:26:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1d27c53e3854d799f47be6f3b85ad09 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54178663a1094ea1b8ae993456a2b74a : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/966a49355eaf46128faf638dbbf2cf60 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fef24933de5b489dab50d50b078fc743 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7970cfaf9455496399aea9b15fbd5fec : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6baf8e99c40849aa8d77a0b9f58e14b4 : SUCCESS in 3m 54s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fe11cb1c40e045579571884ae9ac4237 : SUCCESS in 8m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/42cc13e732234f43b72116924596f196 : SUCCESS in 9m 22s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/98d3d0d8bceb4ca8a67545987158a138 : SUCCESS in 9m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3a227a3f18144d2487a386a56067d4dc : SUCCESS in 9m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ab734ef0bf1e4c9dba8092c0c14ec0c1 : SUCCESS in 5m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0d292eb181ee4c3eb88bf227e2638a9d : SUCCESS in 5m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2516a2c79d394247b9aff4ffc9ad8900 : SUCCESS in 9m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/105ab4fd1f80441a9823f2b014de8677 : SUCCESS in 8m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83b3979e644b4fc499a53b592a675649 : SUCCESS in 7m 58s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9e904a66144946de95db2d4150ec911d : SUCCESS in 18m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1db20c6c1065437388a38e96b0c0face : SUCCESS in 18m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4f3b7b68d92b41748b733968ad5a421a : SUCCESS in 20m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/61b38346036146f49a7217e0695ecbca : SUCCESS in 19m 36s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a63b5a51d73c45e28d6888f5e070ad5b : SUCCESS in 14m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/485b315eba86413087131cd4181f91a0 : SUCCESS in 13m 12s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/691a81feccdd4eb4abaf8ab2b9990635 : SUCCESS in 13m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/84717439f95145148d70cc14f359d6b5 : SUCCESS in 18m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55415c582c86459aac427c8095569dd7 : SUCCESS in 5m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6b3d91bc9e0d4b6d90ed928d06f9b3b4 : SUCCESS in 11m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d5c884d1d5b04a35aa8045b5a9037b8f : SUCCESS in 11m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f7d87efd7b6485b8675827799a952a3 : SUCCESS in 1h 05m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5077adfcf77745b786e4036c29849095 : SUCCESS in 1h 30m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/43435fc06366475cb17b0d13f3e42cea : SUCCESS in 54m 14s","accounts_in_message":[],"_revision_number":2},{"id":"7273aeb48bf195176d0f4ab288cd1cfceba27069","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 19:26:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9e2ca04ddd449b83ee565a2e29ca5c8e5eb259f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 20:56:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818dfc50b7b1409cb1f3b9e78e298180 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b8c2ee5b23a8446991cf4bfbdda606b0 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01f60d63e5374d73826b2701075196d7 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74bcc0861ebc444a82786520240de968 : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/753482224c22458a9c3eab5d5ac70a04 : SUCCESS in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e99c18a203034e848203792498ea16d3 : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/825d6ba1c4e94bcdbfad67199253c590 : SUCCESS in 14m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/014dc2427c7142aa9c81cd26031c77be : SUCCESS in 9m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e5582c4470934d9e863a15ac1fca6413 : SUCCESS in 14m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2ff5ee6da5ea4c95a5e131ce216c2fe2 : SUCCESS in 9m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1582d09c975d4afbbc5332d6837667e2 : SUCCESS in 5m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9a1d28b5f3f6461e97bf3900643a3b3b : SUCCESS in 4m 36s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0ba17453ddb042cea704e728b9b8b97e : SUCCESS in 13m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6e063bb94d694226ad3b317e2f5e3f83 : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad35ba492a3847c58db0f3fe19ecebc8 : SUCCESS in 7m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7b57afecaa6d453e94b905c2256d5ab7 : SUCCESS in 25m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7e0bd69df2aa430da447aacc17d540b6 : SUCCESS in 12m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c9a97bc5842b41348d882e974906d055 : SUCCESS in 27m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/90d156a7c7c14055acce958a9a42f967 : SUCCESS in 19m 21s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/aa987046d41b4451a1757880220a2a97 : SUCCESS in 13m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e2ec530848124d70bff665cb42adf48c : SUCCESS in 13m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/10e27fcf786e433596799f5d9659dcb8 : SUCCESS in 21m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1fb87e232ffa4732af5a519f49738491 : SUCCESS in 19m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cceeb6110a784caaa4327b2847482df8 : SUCCESS in 5m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9efd90d6ca4a4b7f9ae5e199254561c9 : SUCCESS in 8m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/390a4b0935224fd4b8b47b5702fcfde4 : SUCCESS in 9m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb7cbe944417488593113a019d3981a5 : SUCCESS in 1h 07m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6f0d04daa17749b0bd57fa01651bea6b : SUCCESS in 1h 28m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ecfe956d0611472eb31b164d02d41aa8 : SUCCESS in 56m 04s","accounts_in_message":[],"_revision_number":2},{"id":"3c052005cfe5f6c23952b8872de3d9abf16676a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 20:56:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"61894e506d8b7ce3c75532be0a7bae13cbdbd856","revisions":{"29a8558059f61755548e366a7c85b822d137bbbf":{"kind":"REWORK","_number":1,"created":"2020-06-15 09:39:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/735543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/735543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/735543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/735543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/735543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/735543/1"}}},"commit":{"parents":[{"commit":"7de7a2aa6ad56630f1a1ff950d0c597e983ba37a","subject":"update constraint for horizon to new release 16.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7de7a2aa6ad56630f1a1ff950d0c597e983ba37a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 09:39:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 09:39:10.000000000","tz":0},"subject":"update constraint for os-win to new release 4.3.3","message":"update constraint for os-win to new release 4.3.3\n\nChange-Id: I921e67e159f062d72e88dcc9cfb58a9ed9841ac7\nmeta:version: 4.3.3\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: If4db5d7646f910a56dc96a18d22eb5ceccfaeb00\nmeta:release:Code-Review+1: Lucian Petrut \u003clpetrut@cloudbasesolutions.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29a8558059f61755548e366a7c85b822d137bbbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/29a8558059f61755548e366a7c85b822d137bbbf"}]},"branch":"refs/heads/stable/train"},"61894e506d8b7ce3c75532be0a7bae13cbdbd856":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-24 20:20:23.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/43/735543/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/735543/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/735543/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/735543/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/735543/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/735543/2"}}},"commit":{"parents":[{"commit":"d43e0a585719930516fd82504d84e3867512c7f6","subject":"Ensure pip is installed locally","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d43e0a585719930516fd82504d84e3867512c7f6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 09:39:10.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-24 20:20:23.000000000","tz":0},"subject":"update constraint for os-win to new release 4.3.3","message":"update constraint for os-win to new release 4.3.3\n\nChange-Id: I921e67e159f062d72e88dcc9cfb58a9ed9841ac7\nmeta:version: 4.3.3\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: If4db5d7646f910a56dc96a18d22eb5ceccfaeb00\nmeta:release:Code-Review+1: Lucian Petrut \u003clpetrut@cloudbasesolutions.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61894e506d8b7ce3c75532be0a7bae13cbdbd856"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61894e506d8b7ce3c75532be0a7bae13cbdbd856"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
