)]}'
{"id":"openstack%2Frequirements~735544","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I8f349fc384ef1b7d50f1b75b63f30f35402a58b4","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I8f349fc384ef1b7d50f1b75b63f30f35402a58b4","subject":"update constraint for ceilometer to new release 13.1.1","status":"MERGED","created":"2020-06-15 09:42:35.000000000","updated":"2020-06-25 08:55:10.000000000","submitted":"2020-06-25 08:55:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735544-1593075310531-39350062","meta_rev_id":"cd58ea247953c7b54b0c369cd65a0927c56997e4","_number":735544,"virtual_id_number":735544,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 08:55:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-24 20:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-25 00:50:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-24 20:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-25 00:50:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 16:29:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-22 15:59:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-25 00:50:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-25 08:55:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2942a93d1639541296dbbdf03fb17faa5ecd1cc6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-15 09:42:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f5742e077189395093eb2a01bd31e9d5cdf66ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 10:11:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f0af8f286cb4d2d9fe7b860ae816399 : SUCCESS in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cff9965fa4004f6781b729a4548cebf8 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48d96926f6894e97b48801399ae6f6fd : SUCCESS in 7m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1ae600e62774099a1199df4f162d021 : SUCCESS in 9m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73e851ac3d304d43be19fcdd656a4edd : SUCCESS in 7m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b20c1a3b413b44279ecbc30556c8df02 : SUCCESS in 8m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/85f3ed0894444467ac9d4fbccdf389fb : SUCCESS in 12m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6f4fb3777c5748a485656358addca1b1 : SUCCESS in 12m 06s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d42358bce40e428e9696ca979b60ed9c : SUCCESS in 11m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c9071a08b11949a1961f9399e0d37bcf : SUCCESS in 12m 22s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/fd7ea0bbc2a14e7f9a461e87346b5ddb : SUCCESS in 6m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/256d9a43ce9a481c8be8f9fd66ffe735 : SUCCESS in 8m 55s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/af5f27c56855492c90ab3bace98d3596 : SUCCESS in 11m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6b884c4c26b94e1ba28234e8f0e90f9b : SUCCESS in 12m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b8502e28e0e742a198f4a8a3150ff906 : SUCCESS in 8m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/861075844d524fbca30faa17bbb6c58f : SUCCESS in 14m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/55c594fe702146e5bb4640d2d9430973 : SUCCESS in 17m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/acaba701b4ac4517809598af1fd3c653 : SUCCESS in 23m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5993de03840443ca99a0c31a1797a1c5 : SUCCESS in 23m 37s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1ed37126c38a4682ab2fddfb5b9f5e54 : SUCCESS in 13m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f725283e505243aeaafeee3e3d4700c7 : SUCCESS in 14m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4499fd3379dd4bb48e04f0fb80617267 : SUCCESS in 15m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce6a39a64d4448c4b76318efccb5ffdd : SUCCESS in 20m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/33534b216857493097a69eea42cf4298 : SUCCESS in 7m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f57f7352356144d3b731cbb1f73cc1ed : SUCCESS in 12m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2acb16b183af4f55a7b2599ad014027b : SUCCESS in 11m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3ad5842193044f1beeb5fb3c4a1a373 : FAILURE in 18m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/28b9dffaeacb46c28031054e4ccb50ce : FAILURE in 16m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7c9265e9fb104ef58253937c7d8d6845 : FAILURE in 18m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c12e40c2f2601ed824ca51d67e498d31f66a853c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-15 16:19:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c8af6ea803f3f2390a0e89d3790ac74ff20c2b89","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-15 16:29:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"18a6401b47b5c4471c05edb557bb4992ba315c67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:31:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11d6cf79305a4cacb88ff12759b40f0b : SUCCESS in 7m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/273efd9d43364454b6d4c1eac34179cb : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8165732f72ef41f1b44ef659cb0bb4c0 : SUCCESS in 7m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01a719078d124b1e84666c6653a9a688 : SUCCESS in 12m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70f2ea6a97674ade8966bec02cb8fa1c : SUCCESS in 7m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d7c85076633a42e0a37fdd6575e9c3db : SUCCESS in 8m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/11d26a86b6294624be39a68257e0d65a : SUCCESS in 11m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e20f70257e4a4e8eb31f8cb420cfdfef : SUCCESS in 19m 54s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4f6ed7eae99e4b3994b3b9176b57cb9b : SUCCESS in 16m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/24faad63b4ee4fb3be92a9bb2c50dd83 : SUCCESS in 14m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ba8e50121c11409f9044a8cc8a495f2d : SUCCESS in 9m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/01a957fd9b034035971f777ecf2887b1 : SUCCESS in 8m 57s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b6fe76f4c7ca4cd5b4e23c3a37b65ad2 : SUCCESS in 16m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e195babe0b62411c872ee7c222631ee7 : SUCCESS in 14m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/77e55a9d1fa6464ea77d495d71965b5a : SUCCESS in 14m 20s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/87ca8ac9316a45a8a29f318f6d593dbf : SUCCESS in 43m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1712f75148a14d6dafa381e8527169dd : SUCCESS in 41m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6e5852a30ff54e2bbdccf464bdcceb7f : SUCCESS in 36m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f50788bb31c84682b9c439e74ef81ba0 : SUCCESS in 35m 48s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/efdc6684553e45c19c2c2f41ad88e51c : SUCCESS in 30m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1321755293ec42d8b2086fe9d3902384 : SUCCESS in 29m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5864358b9bfc4b02b24ca16c0132f908 : SUCCESS in 17m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42ad8a90ca9b42bcaf89bf15be9f9668 : SUCCESS in 28m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c48122b80f784a2f8e93a1425b5d56f2 : SUCCESS in 10m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/41044718cb03429c9e6c74dd55dfbb82 : SUCCESS in 14m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/112bbaae7c0344da9f5c6c530268782d : SUCCESS in 12m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/423cd68202ac45aa87765367c4ead281 : FAILURE in 17m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a8163938d7604d4ea18125797d40252b : FAILURE in 19m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/69bf6f1943764909a6c8b1086b79a3b5 : FAILURE in 14m 29s","accounts_in_message":[],"_revision_number":1},{"id":"89debcedffd267ade2237745a6a220e9cbf293fb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-15 19:42:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7df70c306b74dac78248efffee5b59a0d9c1fb45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 22:01:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b15bcc54b04eb79816558c3b90a1fb : SUCCESS in 8m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/db4e14b2cf26493c9cb42c07979f0db3 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24bcca81fe0c4026a16a92097c1d18a0 : SUCCESS in 7m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4164f72bcc8049619a950485e6fd39cc : SUCCESS in 8m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c520169f47a84dc3b9a0f376e27a6935 : SUCCESS in 9m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/938a2f9f718b435996826cbc288b208d : SUCCESS in 7m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f1b075b3fa004a69b1c548d449d96ddd : SUCCESS in 11m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/368ba23f11cd47939542dbdcd24a02a3 : SUCCESS in 10m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a865769ac91b441ab8da1c825f0497b5 : SUCCESS in 11m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cb95748c5eaf449cb562b51c637b8e4d : SUCCESS in 10m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9f8d4dd6b88a4e73b7359e618d25866f : SUCCESS in 8m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6cf9c227f6d040929ff5175389931cfa : SUCCESS in 8m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/639fa74d2f5c459ca9f2481b6da3906a : SUCCESS in 11m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3fd1939d968542ef86b7d599550b83a8 : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6151aac3373c4ea693156b5c419d3a72 : SUCCESS in 10m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5a4983d782c84638b84878fac9945361 : SUCCESS in 16m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1192988223c5440190250260e5552bae : SUCCESS in 15m 57s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f43b7e5d3e094023ab322b81b6fc9aea : SUCCESS in 26m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6645ec57039845fe93db58ddf1fa4ced : SUCCESS in 28m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5b8f7ce38b4c4826b22110ea8b8992e0 : SUCCESS in 19m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6407823d0b614cf88dc028a1a2778f02 : SUCCESS in 19m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fc862304fd304534b575cbf612d1f330 : SUCCESS in 20m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1cbedb1219dc410dad5a36afffc7113b : SUCCESS in 25m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b326341e9d104725bda7a8b649555f4a : SUCCESS in 9m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8681ec30720e4b9a96094a78907109aa : SUCCESS in 12m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f4bba223fad74d82861b05f57a7787af : SUCCESS in 13m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/787fa9fffe4c44bf8f6dda4e34691159 : FAILURE in 22m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2d68d5a732a4467883bfc81cc2955db3 : FAILURE in 20m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b0d39d4f6e4c4cbfb12da0dae0f546ab : FAILURE in 20m 24s","accounts_in_message":[],"_revision_number":1},{"id":"6e13dee82ee070607a122f827436a457b68b89a6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 15:47:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c5a6d7d508e724e4393603d99d95c70c3de28432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:31:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4fd16b44f5f4b38baa944f1dec402e8 : SUCCESS in 6m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc873d22c85741d38a3519138914958d : SUCCESS in 10m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1ea6945502341c4ab8d297d0a143981 : SUCCESS in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac7eba751f5f403fa8c5d01155a806f2 : SUCCESS in 10m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf6db048ca124ca0b8af478316b53cb1 : SUCCESS in 8m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ddd767eee4f64728a7c8c02b9f567887 : SUCCESS in 6m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d5bcc5a290cf4dba8d6c344b1bca275f : SUCCESS in 10m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b79041ce49734dab8446f535928583b7 : SUCCESS in 11m 38s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0c76315bb23a426fa1e88083ca47d5ad : SUCCESS in 12m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/932061735c09460dbacf8aeec39d1e64 : SUCCESS in 15m 45s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/40afe5a49bbc4b29befb447f7c265dbc : SUCCESS in 7m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2c5b5a86e8634bd3a615e897ed27617a : SUCCESS in 7m 31s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8fca0cdc57694d6d8db48bbda27e40ce : SUCCESS in 15m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6302e1071fa341fbb22fb8e3f6a97740 : SUCCESS in 14m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a4194fd885a24568a8dade153757ecdf : SUCCESS in 12m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/33899e6e15a8450b996240d5e13f1561 : SUCCESS in 18m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e54e7547b19d4d3392061f14bcbcf42a : SUCCESS in 17m 28s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/22498ad12a384c6295a680c894271845 : SUCCESS in 26m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0cbe4a12f5434ebe89bf4c704e495156 : SUCCESS in 27m 07s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c8e9971cf7b647a086b84de365e6fad9 : SUCCESS in 16m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5d274db0ec764324a5ea4fbdd3f04b43 : SUCCESS in 17m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/625576fb23ef4a1986a34d880092c7f3 : SUCCESS in 20m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5d6ce1cd8f5c4a6e957d39864d39b5ca : SUCCESS in 21m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8689aa72401c485aaccd229b188d0b4e : SUCCESS in 9m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3a2b69b275064c20b7906c35859b8c36 : SUCCESS in 13m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/56c142ed82e147518be0279201653061 : SUCCESS in 14m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/673319e2bd47423fb8f8a73731ab009d : SUCCESS in 1h 14m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c120c03d812147b58f44027b74d9e943 : SUCCESS in 1h 43m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d5467b9c183245a3ba1ed600ea366b18 : SUCCESS in 44m 46s","accounts_in_message":[],"_revision_number":1},{"id":"87a2bde394c990cf10f43468df6f0dcdef5e5c71","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 23:36:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db7e85bf2b1fde57f7e52f4c00f9f6fe142bdb39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:37:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2aebe6a6864a2131ffd712686e952fc8535a0423","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 03:49:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcb8757b2b14477bb3c13e1af723f4a8 : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e1d13e3997e4f1384a2a4e9f42d03c7 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e1e195db0e04acd8ccd0d2e96f0ff10 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b7e7d5e3ef24f958431e35bafc70013 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f16bb4f05484b6395a5cae87bed461f : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3618ee07d7747b2a1ad48186e79a565 : SUCCESS in 4m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e19054630c88460dba0b18da4d349d3d : SUCCESS in 8m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/282492222ccb4a9a8eeba03884df3c90 : SUCCESS in 7m 44s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9677068c1c894c5a8ac8afaf1aeed540 : SUCCESS in 10m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c24ee64bfd9f49d0b4533eeabe0ce316 : SUCCESS in 10m 15s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3d6def63724448a38d9fca0ca5fbea4f : SUCCESS in 5m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/315d0a563ff749a386bc019a9f2ae2df : SUCCESS in 5m 20s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c1220c30dc974f80a6c90701512457b3 : SUCCESS in 12m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ac51c20fe8c84b5e812f9e1448846164 : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d48feae5dc514188bc45642034e143b7 : SUCCESS in 7m 16s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/52ff44c9c33e4af8bb0756178828f421 : SUCCESS in 15m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/83d24c481e494b0f9b39577bbdf9418f : SUCCESS in 13m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/94be2edec9cb410088fb28c5ba555062 : SUCCESS in 23m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dac803adc9404fc28b1e45cae45be2c8 : SUCCESS in 24m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/51ec3d08e3a44848baac708abad4ed81 : SUCCESS in 17m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aff51793ef874b4b911a0ee77b300384 : SUCCESS in 12m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d107bd2654ea4dcb87202ca84557491c : SUCCESS in 17m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aafc7e05ae804fc1a0fa5d6884a54098 : FAILURE in 26m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55d521231dab40538e4cc6c9b516007e : SUCCESS in 7m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4cdc563e93804b4a8a8e7bbe381050fa : SUCCESS in 9m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2a5d718afd494128b79f52dc564f20d9 : SUCCESS in 21m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/68c335cbc88c48309dcda4c7a341ce0e : SUCCESS in 1h 11m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/081ecfd3e06f4a6298b58d4d35eb453f : TIMED_OUT in 2h 04m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6dfdcb6e411e49ec9883e40ce3b9436a : SUCCESS in 58m 46s","accounts_in_message":[],"_revision_number":1},{"id":"739568434718518dbe2bc643a10658e29f5777f1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-19 02:39:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d0e0a579cc8dee350012313e10cda606db8c5a72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 04:22:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a92fe511752412f8cec33ff6623e7ac : SUCCESS in 6m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b963d77b444342ebbd4e99739cb22216 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54e6e959ddf3478e9a8c0aa91933a4af : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a00746f8d3874eaebc6fd03aecf1add3 : SUCCESS in 7m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/499f643d403f4e50924f8c75e3ee8d20 : SUCCESS in 5m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/87bda206bbdf43569c54738867aeada9 : SUCCESS in 5m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/07fd8702cd9442deb694fea989259604 : SUCCESS in 11m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d5fe5b500c5843288edbb1f23a5d956b : SUCCESS in 11m 27s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8c8881d2e1c549fd8d97da861b5e91a4 : SUCCESS in 12m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/17b7d559904f44dcb41944cd2d5a8fcb : SUCCESS in 12m 45s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cb693c24eb614220adb322c14790a4d5 : SUCCESS in 6m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/297458eb701047b6b711c769040909cf : SUCCESS in 7m 46s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/741516c37a354fa382717ea2bcef828c : SUCCESS in 12m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/397e576ceb8f4496ba811f1f9599bbbc : SUCCESS in 13m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e9f84f779cb14af5bb8dd38ae206d620 : SUCCESS in 10m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e11722d7308d4846a057ad0008e45f6a : SUCCESS in 18m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8ec2b7cfa3144301ac32cdde669e057d : SUCCESS in 17m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/bc7a8cea8170406b830d34fcd809e0ae : SUCCESS in 26m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/64aa18e3364447a5a741a1af995d54d3 : SUCCESS in 25m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a45c2c3c2c7e4947bda77f9b9f854866 : SUCCESS in 16m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b901146e0b68475faa560bd2745e1685 : SUCCESS in 14m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/64e3d48dc7ad4bef9a434feda05d0a2d : SUCCESS in 17m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e32583640aee4d6f82ed6ac5dfba1451 : SUCCESS in 21m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a5c710c38dc14e048200800a9af0274a : SUCCESS in 9m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1b38f3e769414d59b739948ead9bf63b : SUCCESS in 11m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e484de4a444e4a4293b018223e1ab4e1 : SUCCESS in 11m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a35cd1f0fe3845f5aca57b95a7e1320c : SUCCESS in 1h 15m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/297ea9552190422bb1272b05cd95af47 : SUCCESS in 1h 36m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c7b6febd2a4f4209950c0555ad3217bc : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":1},{"id":"e7fef4758805a7c3011319f1e45c369757fcd859","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 04:23:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba7f6b25cd8e4f6ffd4b80829530ba1777978375","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 06:10:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/020b49026bb547ca8e7f761c9734aca3 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d15ceaca3a424d1db87045fa41b89fc1 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca2d8d6954fc4976b7a1def005a94fe9 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43dc33e969d5409ca182862ee9b6dfbb : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/693aae11f3d94bc7a4d352547e6ce31c : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1bf6a9593ae44ece8b14c9537d4d3b5e : SUCCESS in 4m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5664f787e3084ce09cfa06491b7700ac : SUCCESS in 8m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0bb19c0a564b48f58f7365df9553d158 : SUCCESS in 9m 13s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/14b538cf9c32463992a87417d22a4804 : SUCCESS in 9m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f6d70e3b53114c5a87a59db37556e9c1 : SUCCESS in 10m 38s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c62a7fb355ff415db9b8298f86e0d914 : SUCCESS in 4m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bbb5faccc74f478f914d8bb3af5c4954 : SUCCESS in 5m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7d24ebd0f7584bdaae2b93cc23845e04 : SUCCESS in 14m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/437c0c2b21aa4fab97b5ee4bed6432bc : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/11ea30c2aacc4fe89228f8991b9ad363 : SUCCESS in 7m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a0fc695e72014e0e95e3bb8e4f3f2f77 : SUCCESS in 16m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a9f68f0e5eaf427da7a7a50af4df683b : SUCCESS in 16m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/29796fa3f1614ebf91711d536259e4b0 : SUCCESS in 25m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9d88b1b5fe8545bfadd1e3881245eecd : SUCCESS in 24m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/cdca237434254221bc7606a5d816b358 : SUCCESS in 17m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/48848068de214d819017fa25edbd550d : SUCCESS in 11m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/15c84d4d349947e68e73678b4a0592ad : SUCCESS in 15m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/43c5f6cf8d734fb18e567a591b96d422 : SUCCESS in 21m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/40de72a236da4021bd4e5f650d460546 : SUCCESS in 5m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8f35d49af35e43f1905b8c1aa021e8ea : SUCCESS in 8m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6fe4d89fe1ca4f2d8d976a626132d5bb : SUCCESS in 9m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1cbd625be0cd434cac39c2e8b36411b4 : FAILURE in 1h 10m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b257f6eda1ea4122bff55a79576f4eda : SUCCESS in 1h 37m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/585b723ea7704cbc8f0b4b2eb1ca3ecf : SUCCESS in 54m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d887ce48c0809d4bdfb5c518907e8c3591154621","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-22 15:59:58.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4835d458e9357a341ac16f0e5981819c2fa674a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 20:08:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2c1cff95a334052822fc821087a2cea : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7c88a188b81d44feaa69507078895ba1 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e40a510f7fba4907b890c299e9a5dba9 : SUCCESS in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37506c2ab6d54a96ade148c981f792d9 : SUCCESS in 8m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9bf0d574acee4798b368ef45efd280c0 : SUCCESS in 7m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/51313558b9d84df3819a50be55516489 : SUCCESS in 4m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8b7a5934ba874b56b18b51e49ad789e9 : SUCCESS in 7m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0ea8028d1da544ee87f86883fdeb7ac7 : SUCCESS in 7m 34s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1295cd31bddc439a90ffc5d18a90195c : SUCCESS in 12m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4d2cb5d2312e4d2f8a65ba142dbfe331 : SUCCESS in 13m 56s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0dcca356cfbf4bf08286601e5bcc0d08 : SUCCESS in 8m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/400ffa51e2cc46489c200c5137d3087c : SUCCESS in 7m 49s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4c062cd1fd3947f2b223ed659f7986da : SUCCESS in 12m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e5a2e73f49f342d09b48155b02bdc7a7 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/74c31d8078ae45468d3304e23c43cfab : RETRY_LIMIT in 6m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e4c914a931d54bc0bab0ccea760dff71 : SUCCESS in 13m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/37845b900e6d4f538c50801786aaebb3 : SUCCESS in 15m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fd9746af9c64404ca42dd8c02b601113 : SUCCESS in 24m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cd9a8678b80149198608299254592344 : SUCCESS in 22m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f224cb6d9a0a449fa7078f626edad764 : SUCCESS in 16m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f20a79fd6b5b446a87e2014ee5238baa : SUCCESS in 13m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cb005fb5578f436ea5312dba6c44627f : SUCCESS in 17m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/852eadcde9c84a2bbf05e8c999f41be5 : SUCCESS in 23m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3b2efca187cb45aba6cd4995f82042f9 : SUCCESS in 7m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/502355e678794deca2e0738469098df1 : SUCCESS in 10m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8e54aef253e34dcead1f9e8fa0db5381 : SUCCESS in 11m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a368c0558106453488a106f197f58498 : SUCCESS in 1h 05m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5866e034550e47beb30f2e945c96509a : SUCCESS in 1h 26m 49s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b237dd8bc09d4dd0bf43ed2c1383b670 : SUCCESS in 53m 04s","accounts_in_message":[],"_revision_number":1},{"id":"83b6658b59f5a29251e9de227e3e76fc9fd5542d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 16:00:35.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"904e64d4950c3e2cf026bbfc61f6e5cbaa421637","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 20:32:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"dcdd8f7aa4b67256fd1d1f0667ca21dc1896068f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:35:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e2c2fc89d894df1bc0cd28bf4fe0680 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0b3aa347773b436a92d484a97615e931 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d35b2dc5d9a34fe0bbe0a612175852c8 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f475c0d7d2024580ac1a736b1efe24a1 : SUCCESS in 7m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e0d5f8d875d848a6b6a61052d11b5b8a : SUCCESS in 6m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a716041e26894334966088f0aefb5988 : SUCCESS in 7m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/fcce4398141640358eeee7fb02d34986 : SUCCESS in 8m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b40760c41e9244589a4e16dad1e198a6 : SUCCESS in 17m 43s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/346854255b7c4c868992fe59b9a09d42 : SUCCESS in 11m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bad4e877c7e4431d86292d272a900e62 : SUCCESS in 10m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b8cd242128e749a79f6b902404b8e721 : SUCCESS in 7m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9f5931fdb8c843938c0d05022b06489d : SUCCESS in 6m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d46ebd856a7a495db788d22e934cdf25 : SUCCESS in 11m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/512b588ee95b4917837fcc0fdbb690e2 : SUCCESS in 12m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/022f40c9737a4ca0b31614c98e36ca16 : SUCCESS in 10m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c360547fbcda4586bb3f6f2d9e409587 : SUCCESS in 19m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7e2749ea9f8c42e0b9471661f47da024 : SUCCESS in 17m 58s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/110279834e9e441eb95daab5dc77da9a : SUCCESS in 23m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fb08d7545d9c4a9dbea12bc6d8ebaaa5 : SUCCESS in 22m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e2262de53ad041bcbe0c982dec49b66a : SUCCESS in 16m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/64d5e1624e1c480bb59e5e73abbed97c : SUCCESS in 16m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/743830471be24b88b3a4e7d1b1a5ce19 : SUCCESS in 16m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a7af031d76ca49ddbbca393f161edfd2 : SUCCESS in 20m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ca5116260592417f944eb0ab00285a58 : SUCCESS in 8m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8cad64473af54e828c31bc390b63a945 : SUCCESS in 15m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bd9e7abd930742aebbd14b36461e73d3 : SUCCESS in 15m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f247e8025754102a5285d43a304c32f : SUCCESS in 1h 13m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fb97ef0898274784ad51887dc3f0a1ad : SUCCESS in 1h 42m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/def1bf133d5d48439b2a332404a79fda : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":2},{"id":"368136f4567a54024c67630f8cd4904cb8614a1d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-25 00:50:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"963b7e532ddce79ab4b12216709d982b48772bea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:51:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cbbc92d4918c82f8972dc8e7a377365a0f2f4639","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:55:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01f39e3d851a4262806c9bd5809e1a02 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6c11787c0d64a14a156993209d51108 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1434f71146f6445eaa40f6722f7ec09d : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e5fbfebb3a84b24a4ab2080304069ac : SUCCESS in 7m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4fb7427283f44330b87d7767f9387ddb : SUCCESS in 10m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dcb575471fff4a5ca3535895bb8e808b : SUCCESS in 9m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ae223f6b1c8c45d3981d6dc9c855d1c5 : SUCCESS in 13m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b1d9220c873141aeb0cc6aae46019c50 : SUCCESS in 16m 01s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/182066fef2be4827a877c23c1e0792d2 : SUCCESS in 18m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/72f10b88fc574ba483c105b4cda9a841 : SUCCESS in 17m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c93cfcdc4de540a5926d36a9f85f3b66 : SUCCESS in 11m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/81d7362407c540da9c215aa00914d4c6 : SUCCESS in 9m 41s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/960c7fe16dec44c994ee1691c7239bb8 : SUCCESS in 10m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d0bbc225639d4185ac8dae2f5bd280e2 : SUCCESS in 11m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd22bb144a334584928fbb3b6f1c8d8f : SUCCESS in 12m 30s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/20c69658836544949761e469451afec0 : SUCCESS in 25m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3c3c095ae93d4e9480d79b42a0fef07c : SUCCESS in 25m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fa608133fc874f2cacf3c719869f5339 : SUCCESS in 24m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/355b8fbc61f8454289685bed2afb36a1 : SUCCESS in 17m 01s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d4b703e6647a4efb84fbf0c08418085a : SUCCESS in 17m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ee2eb0eb63c64a339681f0bbf59a8f2e : SUCCESS in 17m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e8773e24762e4e9d986d742eb297eb98 : SUCCESS in 18m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42e415d5562b4036ab9612df19c1af6a : SUCCESS in 25m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f330820342f741709df333a5f6040053 : SUCCESS in 9m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9d985b18a8a04d7a9ae2680b2a1c3a44 : SUCCESS in 13m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ee9d3fff68954b919ff79b29f160f7a3 : SUCCESS in 12m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d4b44300b5f46faac3633fd913eddef : SUCCESS in 1h 17m 06s\n- tempest-full https://zuul.opendev.org/t/openstack/build/735ad7a5d8d5407292df18d183b08ccb : SUCCESS in 1h 27m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fcfbba9f2fe44481a65ba73e64a551e8 : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":2},{"id":"da45989aa89c7c8f04ca7a8ab37ea13b8e22b9ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:55:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a99e3ffef7de4d43a16d90417bda92ba5df1b4e","revisions":{"fe0837cd6add479e1d3d9831cab9d5c7ae221d1b":{"kind":"REWORK","_number":1,"created":"2020-06-15 09:42:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/735544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/735544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/735544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/735544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/735544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/735544/1"}}},"commit":{"parents":[{"commit":"7de7a2aa6ad56630f1a1ff950d0c597e983ba37a","subject":"update constraint for horizon to new release 16.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7de7a2aa6ad56630f1a1ff950d0c597e983ba37a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 09:42:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 09:42:32.000000000","tz":0},"subject":"update constraint for ceilometer to new release 13.1.1","message":"update constraint for ceilometer to new release 13.1.1\n\nChange-Id: I8f349fc384ef1b7d50f1b75b63f30f35402a58b4\nmeta:version: 13.1.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I851a917158a2d1f51e203cf8b207ddfcb031cb2e\nmeta:release:Code-Review+1: zhurong \u003caaronzhu1121@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe0837cd6add479e1d3d9831cab9d5c7ae221d1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fe0837cd6add479e1d3d9831cab9d5c7ae221d1b"}]},"branch":"refs/heads/stable/train"},"6a99e3ffef7de4d43a16d90417bda92ba5df1b4e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-24 20:32:16.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/44/735544/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/735544/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/735544/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/735544/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/735544/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/735544/2"}}},"commit":{"parents":[{"commit":"5996ae1bd1fa68b8985b49c4a07380d9b723dd4b","subject":"Switch node tests to inherit nodejs-run-test-browser.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5996ae1bd1fa68b8985b49c4a07380d9b723dd4b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 09:42:32.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-24 20:32:16.000000000","tz":0},"subject":"update constraint for ceilometer to new release 13.1.1","message":"update constraint for ceilometer to new release 13.1.1\n\nChange-Id: I8f349fc384ef1b7d50f1b75b63f30f35402a58b4\nmeta:version: 13.1.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I851a917158a2d1f51e203cf8b207ddfcb031cb2e\nmeta:release:Code-Review+1: zhurong \u003caaronzhu1121@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a99e3ffef7de4d43a16d90417bda92ba5df1b4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a99e3ffef7de4d43a16d90417bda92ba5df1b4e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
