)]}'
{"id":"openstack%2Frequirements~735550","triplet_id":"openstack%2Frequirements~stable%2Ftrain~If33f2f78ec192a932b45d8bcd7b69df9ef88819c","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"If33f2f78ec192a932b45d8bcd7b69df9ef88819c","subject":"update constraint for python-octaviaclient to new release 1.10.1","status":"MERGED","created":"2020-06-15 10:02:04.000000000","updated":"2020-06-25 08:26:06.000000000","submitted":"2020-06-25 08:26:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735550-1593073566301-1a3eceda","meta_rev_id":"44db84a9707206e92e26edee4ce3924a8a819ced","_number":735550,"virtual_id_number":735550,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 08:26:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-24 20:31:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-25 00:19:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-24 20:31:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-25 00:19:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 16:28:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-22 15:57:57.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-25 00:19:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-25 08:26:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6f34fbb03207c8066fa3a50f108e46a91196450","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-15 10:02:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa44de36fd9517a8ea8e878c5cee63f1305612fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 10:36:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25e0feb533bf4352a3c25c293f71b35f : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/31401804bb184a1cb0e29a901cc4c878 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f349b1cc1d54af3ad061dc57d81ff00 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb01221a8f24474f884e0b08804a0715 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7cfa8426c974179aebe119c24b198b1 : SUCCESS in 5m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/afe3425caef74de08148e2acc58ba004 : SUCCESS in 4m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4bf60ab6a22740f19745eb46688849ad : SUCCESS in 9m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fce9086d789d4814bcf3fb36b922d87f : SUCCESS in 8m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1b43879f99f24f3eaee36067a1a19797 : SUCCESS in 11m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/24a4212634734f1895a56b4592216f26 : SUCCESS in 10m 35s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ba84904e1aa243a4896e3a13f44fcb18 : SUCCESS in 5m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/77a4c25decab48c597a72cb9af00afae : SUCCESS in 6m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fd6bf96891604047b1068363790c4257 : SUCCESS in 12m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/26dc511ba9d34dc0b964f94c1822cfd9 : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49a99317797642368fe60360735b665a : SUCCESS in 7m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8b8239c26aad4c28b147c1dae0802984 : SUCCESS in 12m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d2a94a2447154a3397031c9be815b711 : SUCCESS in 14m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5a04cd7e610e436aa369da68516a8db9 : SUCCESS in 26m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c75012c19e55430c8f4aa05107c5d175 : SUCCESS in 23m 11s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d038952189c04b3185b484223c3862a0 : SUCCESS in 15m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b71b0561cb9e40dbb03182e6afdd2ace : SUCCESS in 13m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/37303154e1944325b8da96012a095f71 : SUCCESS in 13m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb80f740d80948e0a5cb3ccf4f7c0816 : SUCCESS in 20m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0623fd86a9ba4db39b55701e0e20aef7 : SUCCESS in 6m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2e0c279dbb344a71af7f6d9f6af6612d : SUCCESS in 9m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3e87f0d3418041bd94ba8f8075053dd7 : SUCCESS in 10m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3592ee917fef49eea2d3faea2c5b2fe6 : FAILURE in 14m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1a117ef70633448396cea64da683e3cb : FAILURE in 16m 21s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8e4e7a48621540b0abaf87cfea8159b6 : FAILURE in 12m 56s","accounts_in_message":[],"_revision_number":1},{"id":"b7b3c9acfd03a45b453bbc261d96a731f61b764c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-15 16:19:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5a9ed245d92e6c3842fcb8bab3d120dd61ef60f7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-15 16:28:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7c5395f26b6e59d72a21b1e18de5d6b9aa4fed6d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-15 19:42:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"43c640cdba2ef9c7ba086fb1126015d163d80435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:48:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb73ac93ea4346cd8b2498529cac1314 : SUCCESS in 9m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/88c7092714e940ff9707f63ed0117eb6 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8774ef4721c341e3a2b25943fbbbc0d1 : SUCCESS in 8m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ff9d55426974abda00681c7e92d7b82 : SUCCESS in 9m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/68993dcc9c59469cbecfc92a3e958f3f : SUCCESS in 7m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b1db014c84ce41318331979ef958d4d5 : SUCCESS in 7m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/156738b827254897b10c87ca0ae612e7 : SUCCESS in 16m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/20c62672cb8b4129bbec0299a7dd590a : SUCCESS in 17m 29s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8b160620583249d7823a649b1f7ec0b9 : SUCCESS in 14m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b709330f71bc4e528653afd0c03bfadc : SUCCESS in 15m 20s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/48bd7721b9dc4ad2b8683968444dd475 : SUCCESS in 10m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0ff1fbee404d4306b141a791e1af3030 : SUCCESS in 10m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/672f6129a4294dfc8c6f0fca7e81c158 : SUCCESS in 9m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/96fab0c1f10a4b9db62e710cad706dc3 : SUCCESS in 14m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7cde4d99a25d4620aec5023a41676852 : SUCCESS in 13m 09s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1fbbef2148dd4e14936423e803724d5c : SUCCESS in 30m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/461127d4c5df4c59b029f79b049152b2 : SUCCESS in 25m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/010ea84b513c42db9ab5bd5f2dbc7ff0 : SUCCESS in 30m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/182f3bb1173b40ca9039313c919f94e9 : SUCCESS in 23m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/fe334513db1b4d96b8db943945f75605 : SUCCESS in 17m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2a98f43ee2364fccb1eedf44762184cf : SUCCESS in 18m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/88482a0f99e04191a8c63139eef0fe6a : SUCCESS in 24m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e236ff9515f74970be3c98098e1616a8 : SUCCESS in 25m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/45ef1f74339f40acab5a4f9cce150d02 : SUCCESS in 15m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4476cfbe74e843d3bafb1fb3f9f462da : SUCCESS in 12m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6826d516d96e4730b9b0eee924f9ad12 : SUCCESS in 14m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a0499e36a56349e09362e4054d3fc6ec : FAILURE in 29m 22s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a66fd99a1a3b461bba66672034e0eff2 : FAILURE in 20m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7bb9852e76884132945b9279de70f03b : FAILURE in 17m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d15243dfb9414741f731a467ddebf185d87121dd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 15:47:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"93897526053270e7b962146578e3b873a5b884bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:20:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c41b9e3678148808163b02c660a1a4d : SUCCESS in 7m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/927adbc05edf4370b7e7e79ad64ee8d6 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f5de71151e44aee8376aef7d09c19b9 : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7b5f6299d824847af1b1edfb71d7407 : SUCCESS in 8m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a7e8dbb43b048db8687fe7ab3a2d5d8 : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f0063343d1d4362a6dac314ab7e9a9a : SUCCESS in 5m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ba7198af5de84a7fa76b82c3357ea0c7 : SUCCESS in 10m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f610d50b0cd7412ab73cd864bb387819 : SUCCESS in 15m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3db2249733974c8fa66fb6a1854f37d4 : SUCCESS in 16m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a48e321b79c34caea1ef6d43a3a4ca96 : SUCCESS in 15m 04s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b998724585bb4ceb8b706a7cd8529e22 : SUCCESS in 11m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f0d02621957347e290d1a75395ba90c5 : SUCCESS in 10m 24s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7bc1723a6f634646b1a6458fc3ef2662 : SUCCESS in 14m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/47c010ab6e4f48009a3e2cb4a391abe6 : SUCCESS in 11m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63b98b0b675847ff8e9745ee922c525e : SUCCESS in 11m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9b4b90a726e443a685cb7636e59ec0d6 : SUCCESS in 18m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/23e2be2863af464c8452c32e6e3cd94e : SUCCESS in 17m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/bd9ccf56fbf74a849315439bd8aa13c7 : SUCCESS in 27m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cef0e5c14ffc4aeab61290d42b7cbe7a : SUCCESS in 30m 09s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/77a233569b5a4749a8d5b82000813879 : SUCCESS in 15m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/10a61602d3394a4e98f4fbc5195aeebb : SUCCESS in 14m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d515d2113e3f4370a296a15c95106a0e : SUCCESS in 16m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/511d0fd98c1e44d2b74766b7362ae09d : SUCCESS in 22m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5beabd2ac1c2445cb40d5c16b1c2f93b : SUCCESS in 8m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f6b70254d2e94ac196df942e80849f7b : SUCCESS in 12m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6551392aec3840e5b7905de917093178 : SUCCESS in 13m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e74516004ba425dbcb2a0f0326494a6 : SUCCESS in 1h 10m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/57b817b03bb244ada3a6efa828e67ea6 : SUCCESS in 1h 23m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0adbdccfa4d84eec80f1dd358017358c : SUCCESS in 57m 53s","accounts_in_message":[],"_revision_number":1},{"id":"8c9ab0b631eea1cf55f5809cb18ffbad4f40ab64","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 23:36:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0acd51913aab8ed3127a7688d1593f6b21c6460d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:37:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd3a974af68cf4cf16e35f5c461e034f47d2788f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 04:58:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d64c5710ccbe421bb2337b9f7d0564c7 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57ca16ec80cf40cc8ab5185e11ced0fb : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19f617425ddd4b69a7ac2242f8230d8a : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e58370960b2441c2b5d69225977dffd4 : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c24bb17288b143b0b117273cb812cfd5 : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ddd4b6f6e10c4171a393db087524d751 : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e659464f0d2b43c4b5b4b5944bcedd23 : SUCCESS in 7m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/03c107a996cf4a90a5dca80d33efdeaf : SUCCESS in 7m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f41b35ccdb224d988a9d75ab348df816 : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ce8a216027a6446e8b4d59d3125e4582 : SUCCESS in 10m 00s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/487be605f2014677869c7026529c0ba8 : SUCCESS in 4m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/07995321a0e040659fa43d50c3bb3a93 : SUCCESS in 5m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3ccb193e61fa408fa4d838ecc521df42 : SUCCESS in 10m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5b04b5655c5f471aa155f194418953aa : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ceac35e58884f158c231998331b49a7 : RETRY_LIMIT in 5m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a2e2748e8db9462886587f93b537279b : SUCCESS in 15m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6a3a33322446433cb486f7fd9aa5e1a5 : SUCCESS in 12m 56s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/5f28b24a6db84af6a9aa9fe7d1ae77c4 : SUCCESS in 27m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f8aa0d1ee98548308d7a5675a812487b : SUCCESS in 21m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9d4eb158fb964be18d7692d114ca70c7 : SUCCESS in 19m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/170f91674ede4ef8a6fecad2b097b19f : SUCCESS in 14m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/389c5e51700c4807ba30a3fc056528df : SUCCESS in 13m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6ab83c707c3f48729c64e27a2a1655bf : SUCCESS in 23m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/714f7a1a8f7143509e93cba1e8b136d2 : SUCCESS in 6m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2977ac80740949a4a8ce105413471d63 : SUCCESS in 9m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3b734a6165a64ccc94240d1d0bdb6e0a : SUCCESS in 10m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bafeb2ba726e4ce1aada857df8ac464b : SUCCESS in 1h 15m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f98e6fa574cf4651b3793faa2f15d43d : SUCCESS in 1h 35m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fc68248dd3b34d5ba3f32f44ff167c96 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":1},{"id":"cdd5ac2780025cf21213d799b7496a3e4e717229","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-18 05:15:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a10dc4493bbfdc05c8b8c7d6558359a2d4872d66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 06:22:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1241877dbd24ecba7b75d87de4c71b6 : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea1c1f747480475699733beac7c13408 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31fa6ae906f44fee97537658da9b846f : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38442e3f6a0d4c0d8fc06095095922a1 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b5d583f51474f7aa56967378e0b7f0b : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1163745982d745d981a270c2599a637c : SUCCESS in 4m 53s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4a64daf8e440401986313c2c0c4fa07b : SUCCESS in 9m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a7abe25bf4934bf097d6407ab3119a1c : SUCCESS in 10m 05s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0a965b5129364d61941c2e42b76ef9bd : SUCCESS in 8m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bcc72d8a962b45f99b6872019824b12e : SUCCESS in 10m 27s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/98fb2d49bc4c43e8b7571a5632c9f93d : SUCCESS in 6m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/106862e335234ba696209f1adba71bf4 : SUCCESS in 4m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c8e6fe1d4c06459aab107a9f98513f98 : SUCCESS in 7m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/64b329f1d52a40fca0fcff68daa98dfa : SUCCESS in 6m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/edce51f58b77402db3e966f43dd5a635 : SUCCESS in 9m 22s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d611368bdb4f451b8556b8c443a7834e : SUCCESS in 11m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b33e981557ef484093ecb913edba3f7a : SUCCESS in 14m 14s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4fedfb62005046309ff52a8136541591 : SUCCESS in 22m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b63ee6f6de86480385b9a7decfbe1f8c : SUCCESS in 14m 36s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/527bdcfe21c44662a74b7dff8b92fe73 : SUCCESS in 15m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1bb622f78fca4a779b23b95e787452c2 : SUCCESS in 13m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/078d5ee029f540adb9c4dbaf4eb1d148 : SUCCESS in 14m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9c598afcab14d06b455a44560352139 : SUCCESS in 20m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/660d5f5f8d3e40b9a4ab489d89e5ff94 : SUCCESS in 4m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/64704e78a85d41ee89fdb413b635fed2 : SUCCESS in 9m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/78ee60ac10984e5e9570b46f029dea73 : SUCCESS in 11m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ceb5aab0facb4cb58bbb530aa5165eb8 : FAILURE in 1h 01m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5ba2c7825fd34ba8b33ff42eac630022 : SUCCESS in 1h 05m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7eeae3600d204c9a821cbb26df0501d5 : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e44df2f4c6a55f53985094762b3302b3f75e0585","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-19 02:39:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a997ca7000302e3e583893129b2ee2f14d97afe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 04:10:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d852298f99b4c6fae59e9b84b331b22 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea16c46bc33c404695b2adf703cb1b94 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a723bb322b64db09b8be00f63783adc : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/309d9cbae094423188e352a7ce861b04 : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc028a3fd91246e99ad1dfd084715d71 : SUCCESS in 5m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9457191677514a03854859b4d72c802b : SUCCESS in 7m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6b2caef8dba742b683085c6a922fcfaa : SUCCESS in 11m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/04fad47aac2d4d9b90fa2bcb9117110a : SUCCESS in 14m 53s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/424345f08191426b8187791ed701457a : SUCCESS in 11m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/81d21705dfe548789ee3a68187560006 : SUCCESS in 13m 22s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b515ff26c3af40aba0289a1a64f2e0fd : SUCCESS in 9m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8089e80c9fef44c9b25796722dfff79a : SUCCESS in 6m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9f73f890f20b48b6bde884d09b8d92f9 : SUCCESS in 13m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d5ef2a9a588542c29c14839e7a47ac31 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3799436c88e94977b0f5003f4b57d7be : RETRY_LIMIT in 5m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/73af725eb9d845ed8c629d98da26b57b : SUCCESS in 14m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e2ee630a6cb2485c87797ef1310bc86a : SUCCESS in 25m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9397c34ccf934a75b4a6729ec1ed0214 : SUCCESS in 24m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2d197d7f59394a15b7d7e43a7b627bce : SUCCESS in 23m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/09a3bd99fda84dce8ceb688a3bcff9b8 : SUCCESS in 13m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ac6d0b9504df4a589173c09405feadb2 : SUCCESS in 21m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e2baf281c6434c43a632bdf9be1f1bad : SUCCESS in 16m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c33da81dd9234747a3550a3c114947db : SUCCESS in 22m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/61429cb36abd4e6b89be5d83d5878f60 : SUCCESS in 8m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4ac2d51b65a343bbbf972da48c3be64e : SUCCESS in 15m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/872ddd930b8e4564b463d5f52b42a2d5 : SUCCESS in 16m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c3edbc9800a4f4b9c0d1912d894c64e : FAILURE in 1h 08m 08s\n- tempest-full https://zuul.opendev.org/t/openstack/build/04a84ce7940e45e8a99b7535ddf291bb : SUCCESS in 1h 25m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ee1e2986a9cb428b9aeff1e5323dec14 : SUCCESS in 1h 02m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a0f7bf746c354346f1d03b82f573a6a3733a7e9e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-22 15:57:57.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f610f17ed60505cbdbd0834cd864f4ef1adbafb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 20:09:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/528364e9bffc4399b4b21dc07554c385 : SUCCESS in 6m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3abaf34f6cc4b13bcd97abc2bb35542 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bee84efbf36442539a789f75763bbb8f : SUCCESS in 6m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bd02b94537c430cb7ca33baf16c628d : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f26cab385fea44eaad0921843e9547f5 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5afbef9393c640bd9ab67f7dacb56a41 : SUCCESS in 5m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/37922ad9848e4ce893a6e76588d6062a : SUCCESS in 14m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d6fdf6767fc34462bbc0e04e5ad215de : SUCCESS in 11m 54s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0c15113d70114d35a0846bde12fb4341 : SUCCESS in 8m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/98cefd7f164541158a808de6ef87e2b4 : SUCCESS in 8m 37s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/fa1c9d72298040ef92b072a8841198ab : SUCCESS in 6m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ad0baccba9964b5b84b3c8770704d58a : SUCCESS in 8m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c350e7849e564bebb4a1a441357189ff : SUCCESS in 17m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b8fc6e59250c4bb18f071f2f35578caa : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/beca91d5baf3422f95de536a96528e97 : RETRY_LIMIT in 11m 20s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/991eeb14e35c4dbb93cc608bff7e72a3 : SUCCESS in 14m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d0e68d8f68024d55b6de32f7a1f1cc58 : SUCCESS in 23m 51s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c959ea5d1d0047b9b3a24f97faa88fcd : SUCCESS in 26m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/52f1768302d8419bb521908fa0ecf6d8 : SUCCESS in 23m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a733706427fb40468ba7557548cbda42 : SUCCESS in 11m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6d84ca30e6ba4c3f92aa6a221678ec88 : SUCCESS in 20m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0b0de83dcb904f7c8ee42404b7999c89 : SUCCESS in 14m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e10b95d9ba274b00a88ec5bafc5d6d62 : SUCCESS in 26m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/135d08be7a1c422aac5fbc39ec4adcfe : SUCCESS in 9m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ce89a8880ec54956a0fc1017b28a5107 : SUCCESS in 10m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/391e8e58184f450fbd155920f104a128 : SUCCESS in 11m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1afa1a9bd2854820a8e4db1cb25141ef : SUCCESS in 1h 11m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/690dda7979d942deb0594573fe17c8a1 : SUCCESS in 1h 31m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/de05004caf384afeb3b3cad894a0718a : SUCCESS in 1h 05m 45s","accounts_in_message":[],"_revision_number":1},{"id":"eefb08198be6586687af369f9a1504895ec1d0ec","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 16:00:40.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de8f6332d82f2979d82d60e9ea6fcbf67d6bb0c6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 20:31:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e0f0ca47d70185ee8de7b83519a3aee582d03bbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:01:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d2f68595ec14956958f7cf0b3eae12f : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/603b5611dd324c6b8c1d736d2c3c93e6 : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c3635fd4e4346d1a3cf95f8a4e57f6b : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b5a292a669948afa465102d1729a3b6 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4bdf619556ef4f6fbc5c40ba8d6cd5be : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/625590ea3ad641fbb5712e25deccc0c0 : SUCCESS in 5m 21s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/23c5a0b2011746a3b1e97bb328458549 : SUCCESS in 8m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5a5703423e75441a8a8639b580bd8989 : SUCCESS in 12m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a9e7a2fe7fff4253b3c7571ebb1dc170 : SUCCESS in 9m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ff5ad4c8e3af4408b5f65579828aa2b3 : SUCCESS in 10m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ca0aa371145c459a9c621a9dcf53844b : SUCCESS in 5m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fdd53959595940548bbaae890424116b : SUCCESS in 5m 27s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/58f9082863bc4dd1871e3dbafb77f8dc : SUCCESS in 10m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d950284ce61a4ea48d43bd03b2cba7b2 : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2bd1c7eb7aa448d685891180417777a7 : SUCCESS in 7m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/04089207704d40d886eac39f4f23da6f : SUCCESS in 13m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/757919a015b941479c49ac57d56d69b4 : SUCCESS in 12m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b20f1a74ea2c4916bb7f359f52522e60 : SUCCESS in 28m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/97a186f56dab451094860ac75de129cf : SUCCESS in 17m 16s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/79b03966b6dc424eab0c15f5910c8b14 : SUCCESS in 14m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e371db8b36a74ff0af9c2767127e5d00 : SUCCESS in 13m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d263c91de8a140ed9656c83804c32710 : SUCCESS in 17m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/281a9ee08c32412490de1729db61e790 : SUCCESS in 19m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e2e062597c4d4cba81056f58440b1d2f : SUCCESS in 6m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/eb4b1892a6f54c49ae79ecad9a323c86 : SUCCESS in 20m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b0837d56fa0d4ba5bf89492da164ac80 : SUCCESS in 13m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b6d67557020541cbaef16b9d70b95b13 : SUCCESS in 48m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6ef0a95063a348e69914f24aa878264a : SUCCESS in 1h 36m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3d0b409a56fd4210bf616998625bd7e6 : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":2},{"id":"04ec4e1c722a1969af8786809bd0984419dbb379","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-25 00:19:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8cdf3b89dc762e0ef7203f20ad8b31b240dedb33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:19:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a7ddec2ddf7eb2456ff5c1f1872d7a3c8c5e0300","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:26:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/208afbc16bcc42b4814308ab7419ea1f : SUCCESS in 7m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3fbba24e020340ef8a0ede3d2d4300d6 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/707802f2fa694964b9f40e52d7c2b869 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd6675ae52ca447fb2bda4598a85ec57 : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cdaa71f0ea4d48818dcc347c1a2c519d : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab1a168608464613a3e99c9424856976 : SUCCESS in 5m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2c25b1a3447d45c5af0e0e112b607e4e : SUCCESS in 10m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5203579e01b84ce987872932d0bf76e7 : SUCCESS in 10m 41s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4c6798e218c54d8d90d52b9875eaff08 : SUCCESS in 11m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6f0f25a30fda4e4e8b26900b774f9de8 : SUCCESS in 10m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c9e175279fd34ad3bae895b5998bde99 : SUCCESS in 6m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d6d24375fcd0455491a31267d1f78a26 : SUCCESS in 6m 55s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/53cfadc6980d4a01b4be61000ec59bbc : SUCCESS in 12m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9c773beb604249cd96c56f6d6c477aac : SUCCESS in 12m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd31e3feb7294c2a9e7f6b5254b8e202 : SUCCESS in 8m 08s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6e050e01d81a4db9a041dea7c4779dc0 : SUCCESS in 17m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2b03de8593d64d409eac460b174d4b9c : SUCCESS in 16m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/239b9ae0e95745d5bf3d2c193f9466a4 : SUCCESS in 26m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bed9f9eb5259453ba283d91dd119bd03 : SUCCESS in 26m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f9e1cae802014a3bb680c0f6c80e5518 : SUCCESS in 16m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f7af8f114df8409ea1d3253bc4077d24 : SUCCESS in 19m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/17cb607912c844229a34fd9e4c9a9f27 : SUCCESS in 19m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1a0d4d6080324c83b57d1f041ad82b75 : SUCCESS in 23m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83804d835c424a30b9c9cf68e7f1521b : SUCCESS in 8m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/da3193fc20e24fd89cd174ba44febd1e : SUCCESS in 15m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/24904bbbdd3c46548ea858674f882aaf : SUCCESS in 15m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/976da89d0f444aa990ff6c0f0ebe6745 : SUCCESS in 1h 13m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9aea6b90c07e4fdf90f0e4a593a860c7 : SUCCESS in 1h 26m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2d53ac40a68140fdb2175741d8e0745b : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":2},{"id":"0bdb9310dda193aec853d72c94f2fdabec57f3ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:26:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e3eb4ded7bb341c8e743815e98a7707a7cb68475","revisions":{"d658b1f399fa0f5ba364cf63c1b1dc7cf3f9cea5":{"kind":"REWORK","_number":1,"created":"2020-06-15 10:02:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/735550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/735550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/735550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/735550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/735550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/735550/1"}}},"commit":{"parents":[{"commit":"7de7a2aa6ad56630f1a1ff950d0c597e983ba37a","subject":"update constraint for horizon to new release 16.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7de7a2aa6ad56630f1a1ff950d0c597e983ba37a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 10:02:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 10:02:03.000000000","tz":0},"subject":"update constraint for python-octaviaclient to new release 1.10.1","message":"update constraint for python-octaviaclient to new release 1.10.1\n\nChange-Id: If33f2f78ec192a932b45d8bcd7b69df9ef88819c\nmeta:version: 1.10.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I066b31ee2e2faad2ad9ffea18f7f15384f324540\nmeta:release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta:release:Code-Review+1: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d658b1f399fa0f5ba364cf63c1b1dc7cf3f9cea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d658b1f399fa0f5ba364cf63c1b1dc7cf3f9cea5"}]},"branch":"refs/heads/stable/train"},"e3eb4ded7bb341c8e743815e98a7707a7cb68475":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-24 20:31:06.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/50/735550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/735550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/735550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/735550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/735550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/735550/2"}}},"commit":{"parents":[{"commit":"5996ae1bd1fa68b8985b49c4a07380d9b723dd4b","subject":"Switch node tests to inherit nodejs-run-test-browser.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5996ae1bd1fa68b8985b49c4a07380d9b723dd4b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-15 10:02:03.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-24 20:31:06.000000000","tz":0},"subject":"update constraint for python-octaviaclient to new release 1.10.1","message":"update constraint for python-octaviaclient to new release 1.10.1\n\nChange-Id: If33f2f78ec192a932b45d8bcd7b69df9ef88819c\nmeta:version: 1.10.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I066b31ee2e2faad2ad9ffea18f7f15384f324540\nmeta:release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta:release:Code-Review+1: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3eb4ded7bb341c8e743815e98a7707a7cb68475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3eb4ded7bb341c8e743815e98a7707a7cb68475"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
