)]}'
{"id":"openstack%2Frequirements~735599","triplet_id":"openstack%2Frequirements~master~I147047f7a13ac40acd5fa000be502f291648a6b9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I147047f7a13ac40acd5fa000be502f291648a6b9","subject":"update constraint for python-troveclient to new release 4.1.0","status":"MERGED","created":"2020-06-15 14:20:56.000000000","updated":"2020-06-15 18:03:30.000000000","submitted":"2020-06-15 17:59:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735599-1592243954504-8bc90565","meta_rev_id":"eda102e5db5de10a6b9f7803cd1aae163e952464","_number":735599,"virtual_id_number":735599,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-15 17:59:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-15 16:19:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-06-15 16:19:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 16:19:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-15 17:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e299536fba8d18eebb1d7399c13d3ae7b8097002","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-15 14:20:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20ea291d022a4cb4935be55d08f5a0994c5a8f88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 14:50:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3038d7d6030445e98134a20f689ba36c : SUCCESS in 18m 13s","accounts_in_message":[],"_revision_number":1},{"id":"55e4996ece060015ec82b6b4d69d0e6f2fc401dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 16:11:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf10c16b8ac4e26a780bb90d7836f5b : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e793ae5f7be4fd19fcef400abd82a98 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76df6590fc0e446f956c027561daf7f2 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd6cf71192744910b735798513f4d14f : SUCCESS in 8m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/725fe5a7a77345edb6d67c4a5b17b789 : SUCCESS in 9m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ffeded3957be4f7ca86bf2818ea62f51 : SUCCESS in 7m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/931ab432e3db408d81f0060b9e871b6b : SUCCESS in 9m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/82fdaf3ad54541cda675321e282df987 : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b7cd469fd8124ec0ab62d4260d342443 : SUCCESS in 10m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2c62bb8b12ef45ad91575e5fad9aa5b5 : SUCCESS in 5m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/66c5b907ae9a471fad5fd66e8c9d3123 : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b29f856bdf8d459680a714aa55546b3e : SUCCESS in 15m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0f18821b14c54c00830771ac110d89c7 : SUCCESS in 41m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d1f2af9ed3f742f38bc76ea63d505e7c : SUCCESS in 30m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7f44d8bd3c9d48b9b774cc9021734a8e : SUCCESS in 16m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6576b39aa7e64bbca9f7f6969ae5957b : SUCCESS in 21m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/45be1c745b594e4a82d1b46acc0db0e8 : SUCCESS in 6m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86c85bc84a4149faa053aafbaba54914 : SUCCESS in 13m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f0562ffc49db4f349fc1cb93875bf74e : SUCCESS in 17m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3af281a05f2a49e2add66f22d72cc553 : SUCCESS in 53m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56346a92d80346d4bf66b4673947b77c : SUCCESS in 1h 19m 54s","accounts_in_message":[],"_revision_number":1},{"id":"5662dd77d705cfd98709e05483dd93324f028de4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-15 16:19:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9544f1e4622f432bb41795a527b1473bbf5a97d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 16:19:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffd5e38f29e3c38922e362852131d8c5228fb728","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 17:59:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b3b18bf7aa6fdc462006d658990b5b2cbe906f76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 17:59:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0c3c0b4d644826969d183071f76127 : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceea9a1c400f40549390773449bac24f : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a0302e0a8ea4176bc1da1b1de568662 : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fec1319be8ce4d30a6e3ee4540cecf7c : SUCCESS in 8m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56dcb760a3ae43ddb70a8ae43c4893b6 : SUCCESS in 9m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/58e0b5e021e74534a33cc46ffa6f5ab8 : SUCCESS in 8m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bff294a1f39643fd82bc2ff6f41de939 : SUCCESS in 9m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5ae40db89749482d893cbc9ce8cb4cb7 : SUCCESS in 10m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/69cc28b46ebe4662af312bc21bb47d11 : SUCCESS in 11m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a8ca10f283f1453b87e6a459a765d7da : SUCCESS in 8m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ecc7bd4f1e434edcb31d59ccd84c22ce : SUCCESS in 13m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b2e2ef7a3294d71babbca9240937f26 : SUCCESS in 10m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/593c1a78029b4ef9b680b5607d130904 : SUCCESS in 18m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fb4387c91d9840cfaabf35cd88e3fff3 : SUCCESS in 32m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fbeacb37caf34902ae86785b5e1f69f7 : SUCCESS in 22m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3071b6f5681c4896b176447c06c116a1 : SUCCESS in 21m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a6b90a1ba1c8402bae23c8685847450e : SUCCESS in 9m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8299862dcc1e477883f17cedc8527571 : SUCCESS in 10m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4b0bffd84efc4119b3a94f47c6dc9b42 : SUCCESS in 14m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ce7ec50b55a04dcc83534caccd9f27e5 : SUCCESS in 14m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0caaadd3194a4586a463375fb58fe629 : SUCCESS in 1h 10m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5d3adc3e12346b88b36cbe24a15442d : SUCCESS in 1h 37m 12s","accounts_in_message":[],"_revision_number":1},{"id":"90f00c9c6f4d3efd0dbe403319efc3662c2761e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 18:03:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e0015b8fbc449ddb7982ccc11e83163 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2f50f25c86234f340f66373e8caffee2c54696e","revisions":{"e2f50f25c86234f340f66373e8caffee2c54696e":{"kind":"REWORK","_number":1,"created":"2020-06-15 14:20:56.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/99/735599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/735599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/735599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/735599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/735599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/735599/1"}}},"commit":{"parents":[{"commit":"c2f72e26a42eceb4b64f80d04db8253a4fbdb793","subject":"update constraint for cliff to new release 3.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c2f72e26a42eceb4b64f80d04db8253a4fbdb793"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-15 14:20:49.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-15 14:20:49.000000000","tz":-300},"subject":"update constraint for python-troveclient to new release 4.1.0","message":"update constraint for python-troveclient to new release 4.1.0\n\nChange-Id: I147047f7a13ac40acd5fa000be502f291648a6b9\nmeta:version: 4.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Commit: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Change-Id: I977241e8d0f63ebe04db12864e6881326cba1818\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2f50f25c86234f340f66373e8caffee2c54696e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e2f50f25c86234f340f66373e8caffee2c54696e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
