)]}'
{"id":"openstack%2Frequirements~735754","triplet_id":"openstack%2Frequirements~stable%2Ftrain~If4b4ed2c598969e30bc65b8384bc94e5304256f6","project":"openstack/requirements","branch":"stable/train","hashtags":[],"change_id":"If4b4ed2c598969e30bc65b8384bc94e5304256f6","subject":"Bump tap-as-a-service to 5.0.0","status":"MERGED","created":"2020-06-16 00:57:09.000000000","updated":"2020-06-18 06:45:28.000000000","submitted":"2020-06-18 06:45:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735754-1592462728409-bab3d51d","meta_rev_id":"698e81158e7bc96981d89a63eb9b009ec37bacc4","_number":735754,"virtual_id_number":735754,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 06:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-18 00:51:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-06-18 00:51:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 00:51:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-18 06:45:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3fb7cffafa4f0f731014d1e01451b8245cc8ae42","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-06-16 00:57:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58ff676355054ebae600b9801b550476f8b124c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 01:23:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9052d0141bdc413d9cf401d495d5659f : SUCCESS in 6m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0dfbff72746a45e79895500613bb6f16 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e11b7cacc29d4011b7d2d2fc95ff3923 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4fb13a72e994a92a892c2be180d2783 : SUCCESS in 8m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d54464e4a8994964b31aa32a79eeb01f : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8b6f36986eb24e51bc516f2e6226a000 : SUCCESS in 7m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/91180149ba1a4d619ff2362c56ad75b5 : FAILURE in 5m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/63cd7ccebb834105a32c6d7ac69e91f7 : FAILURE in 8m 27s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c0d543bf931147e49c945b1a2092e136 : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/823b59ded4ff4b45b9f22e6f25e0c74d : SUCCESS in 13m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3c9b114a242c45beae636163016f948d : SUCCESS in 4m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f7db2023d9cc4e438528478bb52564ba : SUCCESS in 9m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f9e824ff48c84f1aa85ad37af6a063a6 : SUCCESS in 9m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b4cf0e7a50b64251840fe98c9d099ec3 : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/356b36bcf5ae447b8b881f53d12238aa : SUCCESS in 10m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a71f9cf98fb4413bb391873fe85b3010 : SUCCESS in 19m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2e027d9fd94f4e8b848f9072a22e6e54 : SUCCESS in 12m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fd50b3d04b564c969c12ff7695758b04 : SUCCESS in 21m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b6af620c2ea5432eacc71fe3693d0b8c : SUCCESS in 23m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/427246901f56488f936d534c404d0fcb : SUCCESS in 15m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4f77d43564dd412787e1fecfe242da0f : SUCCESS in 12m 19s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a0ac768238e2449b96dff966a3dbdf58 : SUCCESS in 13m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8cf563f658ab4ab4847d9b1eb12f0bf7 : SUCCESS in 18m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/efb0a6720ff34eaab7e2816d72015c1a : SUCCESS in 5m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/579f3c96bf614323a886dc128930caf1 : SUCCESS in 16m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1334ebbcf2bc4732aecf408aeffebbc7 : SUCCESS in 10m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9337024e6d8043059907f20f7c3e2087 : FAILURE in 18m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/714fca94df7047909439b843d3dc9f44 : FAILURE in 9m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/26566cf0d0fe406686ce7bf111a693af : FAILURE in 18m 42s","accounts_in_message":[],"_revision_number":1},{"id":"91eca58d5642b8106a104ea3d2fc45dff085ac3f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 15:47:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9b1c1e2325b55e46073f27653bd722c48a806019","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:40:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14e155e73b8d47f89adbc1b2018668aa : SUCCESS in 10m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c84b2a2ac75e4a3f81fbae32c24e7342 : SUCCESS in 11m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05a608dc5e4d4413ba0768a9717b3b81 : SUCCESS in 10m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0303c6b137ce43ca8ff1628c274f09b8 : SUCCESS in 10m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e04aa3d474334f3894fff84892ae94e4 : SUCCESS in 10m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cff1ab8d6d364e66b61fc58172005924 : SUCCESS in 7m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9993266c5c544b96b1557df14921058e : SUCCESS in 10m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e51ac5f1b2064044b91fba05b791b4ce : SUCCESS in 9m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/050eb28eeaa240a88c99f920344a4b5d : SUCCESS in 9m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/87bf5c1c49d4490da209e90a30fc79c9 : SUCCESS in 12m 09s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a4eef719856c4be2bdd026521bdc2b84 : SUCCESS in 9m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ed4e55f3d25e4a12a0c3559f749103d3 : SUCCESS in 8m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5e7442bc252a4fafb88bce342621a0de : SUCCESS in 15m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/85098af2c98a4609951bbe5b93c947b8 : SUCCESS in 13m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/491824fa4cf244d581d6f4a7b63843c1 : SUCCESS in 10m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/b2e442a40d2649a8b72e401a4fda632b : SUCCESS in 27m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5409b3b69c354901a4e07d6f1ad0b510 : SUCCESS in 23m 44s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4ff82128a3f943578191b37ce66da3d2 : SUCCESS in 25m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eaee933c98c541af9678852a7696fdb4 : SUCCESS in 27m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/09a4117b1ab6425e8a94ca9354b138b0 : SUCCESS in 20m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4aed9b7c8dc0474591d295242a297153 : SUCCESS in 20m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/41089b1d77fc483490342d5bfa869027 : SUCCESS in 17m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b34d57adeb214b54b388b139e2c4d8af : SUCCESS in 23m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ea370cd3191a4e50bdeafad880fe2b6e : SUCCESS in 8m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cf573f2fc2164702a2045d8558847c4f : SUCCESS in 11m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/262bedbe6ab544b6b49b84614c32105a : SUCCESS in 14m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ae79cb0013c46058c5ff4df94027e47 : SUCCESS in 1h 14m 51s\n- tempest-full https://zuul.opendev.org/t/openstack/build/60e24c70333f4671b65cbbf404a4eb42 : SUCCESS in 1h 36m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a5f0470e450844e5a88db6bf6ed77687 : SUCCESS in 1h 06m 33s","accounts_in_message":[],"_revision_number":1},{"id":"976c8c0af8a7762e8ed83d3d1bf74c92f92603e2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-18 00:51:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fc085220b83e9fee1a6eec04ec6f1077e7ea9b8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 00:51:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a5c52e8de43f66d09b15aa75fcac517cb301090","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 06:45:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4596ab9ff13a4784af8c05d0472e2fde : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b333311916d4608a3c6a9f276732702 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76bdf04e31744fecaae318b625def111 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e11541c381ec4915ad7993a84c35e285 : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8bfde2bc4cc4cc483d15b0fabc6d29d : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/66ae486ce98f43cfb4094f6764e3546a : SUCCESS in 5m 46s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d7076b535faa42caa994a809c15608ee : SUCCESS in 8m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/39a506f0aba743e8895963279546ea19 : SUCCESS in 8m 13s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/40e18c584efc47ce8167e9941f365188 : SUCCESS in 10m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c768698082634975b8302690a9fc7ce0 : SUCCESS in 9m 22s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/56d4d1c3ba7c4e70b125743b7f78cb42 : SUCCESS in 5m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/614a4fda3b154d85b6f635e1f2adde69 : SUCCESS in 5m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9d54c804495c411db1ba48bdd2bae485 : SUCCESS in 11m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e500a020d99d4d9188f6b4690be96dc3 : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/91bb7ef4d6214daeb2facb5a0036e529 : SUCCESS in 7m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/3477a4e4b2864bd7b2944f27745ddae7 : SUCCESS in 14m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f39a648e7a5249d7be95eaf3847e6389 : SUCCESS in 13m 06s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/85749712afce44359034cfcc717c5b20 : SUCCESS in 25m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7a3b9854940447b5928824c279f0dd61 : SUCCESS in 24m 26s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/96f98881ce5a414d9a168e58256a4a5d : SUCCESS in 12m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3e6b49676f0146b39186bd74df7226a7 : SUCCESS in 12m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/31c91e18f74943d199cfaf7658ff60a6 : SUCCESS in 15m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/067e38b86ede4d66b931916bf716efe1 : SUCCESS in 17m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/383e45d405894b67959407821d0e68c9 : SUCCESS in 5m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/749431a432f449f68369c7afc5233f85 : SUCCESS in 8m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7885f697c75645b885ce50449368d0bb : SUCCESS in 7m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/efa2843acaf84083bb39010e27224cc2 : SUCCESS in 1h 10m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f77e2f3bccea413396e5c74a1c58f690 : SUCCESS in 1h 10m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ccf58ba8652b463f965e57266e156aed : SUCCESS in 1h 02m 10s","accounts_in_message":[],"_revision_number":1},{"id":"698e81158e7bc96981d89a63eb9b009ec37bacc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 06:45:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ca83a6cc6daf6ed43fe80d4a80e0b737ddf0fed","revisions":{"1ca83a6cc6daf6ed43fe80d4a80e0b737ddf0fed":{"kind":"REWORK","_number":1,"created":"2020-06-16 00:57:09.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/54/735754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/735754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/735754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/735754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/735754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/735754/1"}}},"commit":{"parents":[{"commit":"7de7a2aa6ad56630f1a1ff950d0c597e983ba37a","subject":"update constraint for horizon to new release 16.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7de7a2aa6ad56630f1a1ff950d0c597e983ba37a"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-06-16 00:56:27.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-06-16 00:56:30.000000000","tz":600},"subject":"Bump tap-as-a-service to 5.0.0","message":"Bump tap-as-a-service to 5.0.0\n\nnetworking-midonet needs version 5.0.0 in train release\n\nChange-Id: If4b4ed2c598969e30bc65b8384bc94e5304256f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ca83a6cc6daf6ed43fe80d4a80e0b737ddf0fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ca83a6cc6daf6ed43fe80d4a80e0b737ddf0fed"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
