)]}'
{"id":"openstack%2Frequirements~735817","triplet_id":"openstack%2Frequirements~master~Ic34b06e42757d848f366f9c3064e0c8f3a670894","project":"openstack/requirements","branch":"master","topic":"retire-neutron-fwaas","hashtags":[],"change_id":"Ic34b06e42757d848f366f9c3064e0c8f3a670894","subject":"Remove neutron-fwaas and neutron-fwaas-dashboard from requirements","status":"MERGED","created":"2020-06-16 09:05:03.000000000","updated":"2020-06-17 11:14:36.000000000","submitted":"2020-06-17 11:12:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735817-1592392363506-2f414a6f","meta_rev_id":"9e3e373816a1c57e71a8aa5c68731eede0dcef21","_number":735817,"virtual_id_number":735817,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 11:12:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-16 15:44:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-17 09:51:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-17 02:10:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-17 09:51:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 15:44:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-17 02:10:08.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-06-17 09:51:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-17 11:12:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c4fc3d1f6c3f0488706b2faa17adc89aad0b21d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-16 09:05:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2163ac793cece885def9e7a2114ad7401aaaa494","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-16 09:46:33.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d8987c6a014a21dd1f6c75e2089c3e532405d76e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 12:14:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4a1a7d0073e4b61a8d6cfd5d68c28bf : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e87cf3d8ed5d41db92641b53812ef730 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54e2035646c74cc2ba8c2cef0cc6e408 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83f393c4b3fe4284b5b3e5f8c56630f0 : SUCCESS in 8m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32c2597e212c4f7cb21599ff4604d06b : SUCCESS in 10m 32s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/4cdabf3cd09f4938afcaf0d945a2216d : FAILURE in 8m 13s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ec13cb3ac994f07b0eb0044e3136823 : SUCCESS in 58m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d1d170448d84ddd912a8b851df5337c : SUCCESS in 1h 22m 35s","accounts_in_message":[],"_revision_number":2},{"id":"2d2f6c38ddac9a8f5e8dd558ae7dc5ff69cf630d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-16 15:44:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0bf5a8bf22a4e7fd3613bfa6a72d700a41c609a3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-17 02:10:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"90ea39e7ae6e1d5479534d267451b464411a78fb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-17 09:51:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"91551b3db7e046e0ef8ecca0cf0b19776111da13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 09:52:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6fdc1ebc12700426ad175f808267a0b6cd8633af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 11:12:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8839b6ecdc549c58a31fef9b000a651 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8e5f34b473e47cf8d6c155d759c910a : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/436c866544364995bdd27d4de6bf6fe5 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db0a3d5b640a4e4398bec011f9e2803e : SUCCESS in 5m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e5553038e2b4149808f25f66cbabfa7 : SUCCESS in 4m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce3a24f3b7564755a6f2995f7cd449c7 : SUCCESS in 54m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2d6218dab054455a72471cd39a7ab65 : SUCCESS in 1h 13m 46s","accounts_in_message":[],"_revision_number":2},{"id":"df64ada513e32e266fbb1a730ddcd86bb7f71244","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 11:12:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cfa8dc40bc5360735400ca1e6d1409a7eddfba55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 11:14:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4644c768366445fcbf3ce3905da64a43 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"52b0966a22a05403a9547a59544b85f6f9a0ec1a","revisions":{"55b5f6c051580e4644af261c0cb53f9ec26d6bc7":{"kind":"REWORK","_number":1,"created":"2020-06-16 09:05:03.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/17/735817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/735817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/735817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/735817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/735817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/735817/1"}}},"commit":{"parents":[{"commit":"d1bce457f78d7410cc64d57c056e4142e83476a9","subject":"Merge \"Remove the Tricircle deliverables from req\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1bce457f78d7410cc64d57c056e4142e83476a9"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-16 09:03:12.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-16 09:03:12.000000000","tz":120},"subject":"Remove neutron-fwaas and neutron-fwaas-dashboard from requirements","message":"Remove neutron-fwaas and neutron-fwaas-dashboard from requirements\n\nAs was decided during the last 2 PTGs, neutron-fwaas and\nneutron-fwaas-dashboard projects were deprecated in Ussuri due to lack\nof maintainers.\nNow, as we still don\u0027t have any volunteers who would maintain those\nprojects, it\u0027s time to retire those projects.\n\nChange-Id: Ic34b06e42757d848f366f9c3064e0c8f3a670894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55b5f6c051580e4644af261c0cb53f9ec26d6bc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55b5f6c051580e4644af261c0cb53f9ec26d6bc7"}]},"branch":"refs/heads/master"},"52b0966a22a05403a9547a59544b85f6f9a0ec1a":{"kind":"REWORK","_number":2,"created":"2020-06-16 09:46:33.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/17/735817/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/735817/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/735817/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/735817/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/735817/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/735817/2"}}},"commit":{"parents":[{"commit":"d1bce457f78d7410cc64d57c056e4142e83476a9","subject":"Merge \"Remove the Tricircle deliverables from req\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1bce457f78d7410cc64d57c056e4142e83476a9"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-16 09:03:12.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-16 09:46:28.000000000","tz":0},"subject":"Remove neutron-fwaas and neutron-fwaas-dashboard from requirements","message":"Remove neutron-fwaas and neutron-fwaas-dashboard from requirements\n\nAs was decided during the last 2 PTGs, neutron-fwaas and\nneutron-fwaas-dashboard projects were deprecated in Ussuri due to lack\nof maintainers.\nNow, as we still don\u0027t have any volunteers who would maintain those\nprojects, it\u0027s time to retire those projects.\n\nChange-Id: Ic34b06e42757d848f366f9c3064e0c8f3a670894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52b0966a22a05403a9547a59544b85f6f9a0ec1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52b0966a22a05403a9547a59544b85f6f9a0ec1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
