)]}'
{"id":"openstack%2Frequirements~736030","triplet_id":"openstack%2Frequirements~master~Ie9c304546c058c895cc42d41ba766e034cdf6e0d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie9c304546c058c895cc42d41ba766e034cdf6e0d","subject":"update constraint for oslotest to new release 4.4.0","status":"ABANDONED","created":"2020-06-16 20:46:39.000000000","updated":"2020-07-29 16:18:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0c0f55a89fba155329279276e07bc9e7bbd4fe36","_number":736030,"virtual_id_number":736030,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-07-29 11:51:19.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-06-24 16:00:17.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2020-07-29 15:12:13.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-29 16:16:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 16:00:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-29 11:51:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-29 15:12:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-29 16:16:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e7a15498f31a53c00079c08f52b3249c56f11d44","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-16 20:46:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"359fb0a95f732281345fc4b890a676ef9678ddd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 21:39:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"9bcd0cf7e69db6dc08a2b315b6c5fb7dc0a5d632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 22:51:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c971829614bc4bffb4d12a845cf0162d : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc08ebea496d4783bb0ebb99c7ea062b : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd5b83dbaaed4f81b5b18cf2a8f5c561 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bb805560d434cbe8805b24e5a4071c7 : SUCCESS in 8m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8e0e15bf9ff4e7686ba89e33a7ba031 : SUCCESS in 6m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b8574f607b2142da8cb5e8ee3c2fdd7b : SUCCESS in 8m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db218f5ad15b43d5abe34c432fbd9a5a : SUCCESS in 10m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5a71034b101b4a6b8b75f0be6aedbcd1 : SUCCESS in 10m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eca605026548458988bb86128e0baafe : FAILURE in 12m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/da349da05cfb4534baec9c235db00043 : FAILURE in 6m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5e4cdc3eab454126baf1828285ae4468 : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/154e98bcf9a94152ad9f06a751d139fe : RETRY_LIMIT in 8m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/36ec5577794e4d5089946862272667a9 : SUCCESS in 18m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9d324b5ad2df4b24b6849de84989b364 : FAILURE in 8m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/020389f8fdbd4282953f8fa29e1f15a6 : FAILURE in 14m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e4cdd0b8ab340339f4d1060acf482dc : SUCCESS in 19m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/27b24caba2d74e2dbb8f1aad6bc0fe60 : SUCCESS in 8m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/242ffde832854a7e84084830d66226b1 : SUCCESS in 9m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9e389d7a083d489f9e9e1476991d2e6c : SUCCESS in 12m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dba1f3ca3f1743ac98bd12d7b71d498b : SUCCESS in 44m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd06cb166dad4c568a2fbde3dc2d9b0c : SUCCESS in 1h 27m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8bca899e1a0367bd197e0f733b0a269b821e884e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-16 22:53:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"97f4cde3b28947dd7e1c6f1a478f2c44eb9bc7a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:34:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"843a39311bd843c71f00b8514d3d5d6ed1946972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 00:17:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba9c5a6f56dd42dda60bd6bae1e0ad88 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/039cc3a171dd41789270a4f207136099 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57b17f4ad1294aa2bf706a5350f022d9 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8102d3d3c50f44cb821cc4bec1119e31 : SUCCESS in 6m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8dcbb59920044846837d420a3ff8b6c2 : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f146b1bd997c4d96926c3a099614f54c : SUCCESS in 6m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dff2f43cf4b143cfa3d5a01c4dd3c3c8 : SUCCESS in 9m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/68f292ea54bc49299a7ce7eafb5ec22b : SUCCESS in 10m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d1d2ff17545c41c48108d90071a88f0a : FAILURE in 11m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b5234cb64f3c48408a5ae97a42cd6a9d : FAILURE in 5m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0601ad6b4dec4c3ab99cc7108d3ad211 : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9262b8b0cbba40b8a83252b608f8c3e1 : SUCCESS in 7m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/99f96173fb0048f6ba43360b35f53449 : SUCCESS in 13m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dacd777fe086417493ed2b2c3c5ef92a : FAILURE in 8m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/527f47a430f04113b4629867909974b4 : FAILURE in 15m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/30b5a366898448e1b62e7d638105d9e2 : SUCCESS in 21m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c951e884253a449cbdc61ccc3244e2bd : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/777e35dd55104240b2edea2475c82ba4 : SUCCESS in 8m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d9094b42569b4086a23b95224bee5b68 : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a60877458d04d5489f0b24c4401f659 : SUCCESS in 1h 03m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/896f6f5dc9bc40298b38ee1657cf68ef : SUCCESS in 1h 18m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4dc20b013726f3ff07b9b8322c0bdbf778ee5a7a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 04:00:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"146edfce6acf77558e8d9219b01ee3e7786b410f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 04:43:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"933440a7459e7bbe5056b5631542b54488d54084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 05:07:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a27d2bbdf6742fdab941d667047b49a : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2789e4a6b8eb4061befe2ec580d7e1cc : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc8b7acdc4154eeebc3cf99420819602 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/801a775e682e4dee8fe258706b5c8cd8 : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e48e9b60b9a445fb798b232d72d3bc7 : SUCCESS in 6m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0390e8a9f81f412b8ada3dd8015ffed4 : SUCCESS in 3m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b3b68ee5047e45b58e9ebbb34de1846f : SUCCESS in 8m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b18c3799412248218a98f270f9c2baf5 : SUCCESS in 13m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/97f1aac1b08d4fdeb01140582748cd98 : FAILURE in 8m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/001aba2be735484b872c9bff45a32cf0 : FAILURE in 4m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d56c84b13977441e8b5a9321b01a2b0f : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e16270d627ff4ad4922e8833b2cb7fc3 : SUCCESS in 7m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7b5593e5141846838491259ba1aab65d : SUCCESS in 26m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/20c4a8947af34f9db133887103c28b14 : FAILURE in 4m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7661a7d794fd4c07bb6ba946836e6016 : FAILURE in 12m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e566f35c3ba454484737ee2ff6dbe31 : SUCCESS in 20m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cc1a0a37ddfc4448b8a220c064448882 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b23463fa2a443e690e53ceb6a3fafbc : SUCCESS in 7m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2cbbf36934f84375bf591fd90b4035ba : SUCCESS in 17m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c46446ef55a4465caeda6451b8d61d7c : SUCCESS in 52m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/548c6e45ca1f4433b0d921040874f0c6 : SUCCESS in 1h 07m 12s","accounts_in_message":[],"_revision_number":1},{"id":"93a3ce8159265a716f321756720cfdce84034646","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 05:34:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"735307cbdcbdbff63f808b66bf73480c446be3af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 05:50:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"6221f5ac966e60bd27f08d0e6fcba793943f62a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 07:03:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce822ba1f8dd4a109bcef3c17625f78f : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d2da9539c849fdb50a2d282bf55488 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59718b8f69b741d0a2fc27dce92d54fd : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d640d5bf71e41c4b6b292e2edf2a125 : SUCCESS in 7m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a8ff9cc685b04104955417f7f8028890 : SUCCESS in 5m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/51208e86dcb94fac916d0e9622c2d6e9 : SUCCESS in 6m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ccc4821a387047548cd2a4299a072b2b : SUCCESS in 10m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/058baf3caa0f46d8b91aa6c8ed5badbd : SUCCESS in 15m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e8a46f79993e44aea2ef9193263b1d80 : FAILURE in 12m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0d61ea04200f4f28a57714595ae3eee5 : FAILURE in 7m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0ebcdd84925d4107a5f633ced2779703 : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8607b0211c294aa894d8315b0de355c7 : RETRY_LIMIT in 4m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/79c0af02904c47ac81ff5c0be0799b07 : SUCCESS in 31m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dcb066c75bb14677ba7c94a65026180e : FAILURE in 7m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/65fd70069e384f2ea46d66accbba6513 : FAILURE in 12m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/96e5bcadd90549b2a22ac286c3f1ce10 : SUCCESS in 21m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d16aae6e0abc42769def29739cc0b4b5 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c66cc464718a421187b85ff6c5f2d257 : SUCCESS in 9m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/82a2dabb88e344688617ca0a45bfb1f0 : SUCCESS in 10m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c6688323e9a4b919d023f22d5c872f4 : SUCCESS in 57m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f750fd723294af681dfcf8a143b7c0c : SUCCESS in 1h 28m 49s","accounts_in_message":[],"_revision_number":1},{"id":"da05295ffaf23f41874ac9abe0ebcda6c30460c7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-17 09:49:40.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bb79e69e684d39bea3815e091d76e226fa66433a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 09:50:03.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"ec6ceb94661d06c87eda80657a888042b04fc78e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 09:50:07.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"53d8f74bfe7a296d10fad5968b9112050ceebfb0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-17 09:51:07.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"67efd56c02f2d253adc77e815134380a62c9685d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 09:51:30.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"93e60426507d04041a00f00dbc387f2c98fd22f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 09:51:32.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"fd8c34bb1b350f01a5f21ead5bd4e2efd7c791cf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-17 09:54:27.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"fc665172e2c958c3a773162c0051992ed391a8d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 10:24:10.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"98c01ca8310b70347fdd6db5f552c55eb43eae7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 11:34:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a62b41ebbb94cb485914abe9c108c3e : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72be10db66fa498eaf94178955d009f9 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a817d08aa68a4a52b6274c356b982a02 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e677fe5754be48b1948eacf6b32025ff : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1cb5219706af432fae90ecf28b7018ae : SUCCESS in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9d9c56a61a5a41489e6b833b72c93ed0 : SUCCESS in 5m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a59ddc6b375644a382f21cb2ea7d2736 : SUCCESS in 10m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/40c045ee7e5d48e9b3112240fd23e23a : SUCCESS in 11m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/96996c22a1b94baca4de27d5b8dd946b : SUCCESS in 13m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b7f530f216c8408995deefadea9660b4 : SUCCESS in 6m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ee182e40e0114487aaf2e74a04866dde : SUCCESS in 12m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bc8b34f52f1548aaa5316ac7782edcdb : SUCCESS in 7m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eb6f8a1dcdf24a3689b57c0a117513dc : SUCCESS in 17m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f77df40735c24e50af3b81e1198cc61d : SUCCESS in 41m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f4b8d54882d04053848ff85637300558 : FAILURE in 16m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7d4f0d42a8354a44a573bceb9f4abc78 : SUCCESS in 18m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6bf1822d308341ec805084a71621f9ff : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c8cd6b8c549846509e824dfa55a40696 : SUCCESS in 10m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/14cc381bada240778d51f5fffb699a7d : SUCCESS in 17m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f56805415ab4e559f49073a3884a7f1 : SUCCESS in 52m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e725091ac54748d4a06fb6a31db4c019 : SUCCESS in 1h 22m 30s","accounts_in_message":[],"_revision_number":4},{"id":"22578b8c7d5e251f59b1544f8fd907d421eabb67","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 15:47:33.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"011c8532a5af4a6b33487b99f54c69fa325f01e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 17:13:10.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"3bb915dd869b11a924d06e81ea08644e519127d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 23:34:45.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/588db3780ad94a4094920069b1b13b0c : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfa38bd03a774926b933f45a26b9d030 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c730ab79132449dc8dc05dd95aa207e6 : SUCCESS in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/465602819e7a49c48ddad894c9db4c5f : SUCCESS in 8m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bdf6e87eac6d4dc0b2d58d58ec6aeb8d : SUCCESS in 10m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/59beac0bd9a64717b495391cc0caf3ca : SUCCESS in 7m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e9a5e6d2a45d48ed82863c7c3998c80f : SUCCESS in 10m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cc35f57507ee467998fd820333ac3fae : SUCCESS in 13m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9552734fb2434ad08b674d3882f596da : SUCCESS in 10m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3b2da780862346319eb01d3257cda04c : SUCCESS in 9m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1cbdab22808c48c783a1bdeddaa30a46 : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ffffdc0b17d44489a3c1318fd15e9e7e : SUCCESS in 12m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e3c7de8e29d24404b01b05e93b53d814 : SUCCESS in 14m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/237e01f31dd248a2be848c98f12699d6 : SUCCESS in 28m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/00ca9899d2444b7eb6bd51f31050ac4d : FAILURE in 17m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/05e14a0be6614fb593a4184c9d889c56 : SUCCESS in 20m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2111bfae34f049539c626f8585b71514 : SUCCESS in 9m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a9e02cb004a4a26b9f873d7f4948d43 : SUCCESS in 11m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/13ef8e5e075945ab8514de364610e174 : SUCCESS in 12m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cdec65ce64143d2bdee7f15ac7b0fd6 : SUCCESS in 1h 00m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/879ed68a9332450bb5ceaa24728c1343 : SUCCESS in 1h 28m 04s","accounts_in_message":[],"_revision_number":4},{"id":"b008da1035855860eb936b720c70c1e8ec2799b7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-18 20:19:01.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a0638b8293e1077b43cb0affa70d9c5e936a556e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 20:51:13.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"cd400986beeb863382bf3782f4c82b7498f8e6f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 21:55:06.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d8efc8b84d042a0ad87d77bffe789b4 : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15eaba5e593f4fdf8a563ad09c64f774 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3b704b570e74e778f80043cd728a930 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37dfc4216cdf4f919bad8a0f36c58bee : SUCCESS in 8m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84a5710b81244ddfa88be8891025a82c : SUCCESS in 7m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/83a05fac476b44b88169ca2c4577b1a4 : SUCCESS in 8m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/52d369afc1b042a9b19b627ffb435a73 : SUCCESS in 11m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4daefda03f3b4a719f0604238fd4740f : SUCCESS in 15m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e62a618219364103a146db3a3f21de78 : SUCCESS in 14m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/31403e0807424f219a0ceacfc8047fef : SUCCESS in 9m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a1a5062f9f4a4f8c98f8cbb13ec24773 : SUCCESS in 12m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e656dfa095ee4538a0a63f4fc217dc3f : SUCCESS in 8m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d0ef030448944d56a791be30fe09ec84 : SUCCESS in 21m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/424ec8123bcc4bf58c193ca663958a9b : SUCCESS in 24m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/71a0f676be87468abba6db2d42ba2304 : FAILURE in 14m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/83ef2c2479114790855603896454250b : SUCCESS in 22m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d0486828d8bb433483389c52508c5ce8 : SUCCESS in 8m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bdb4491fe9484f5b8f5f35ed3f0fa887 : SUCCESS in 11m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e68b895aae0949c9995ec11f9f04f3d3 : SUCCESS in 12m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5ece1bc01ce4ee0b912bebb60cf59c5 : SUCCESS in 1h 00m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9677f69840c640ef9daa5273562a19c6 : SUCCESS in 1h 34m 13s","accounts_in_message":[],"_revision_number":4},{"id":"9f378e71249859fde320f54eb9caebf0c23fc6a6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 16:00:17.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b52a54883f20abb16f89459d373959b14fcb9e45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 16:23:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/65707b2aef4245898c887f15e04b7bf8 : SUCCESS in 19m 11s","accounts_in_message":[],"_revision_number":4},{"id":"753e0cbdb5462af86c5b46864bef18abef0e10bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 21:51:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb26eec59ed446cf98edc5e9cdffa4bc : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/077b5d76f453479f8800522403806bda : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a04ec70cb39148ee9a6aa7b302ba430e : SUCCESS in 9m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b93be0e48b8741fa94aeecf1455b08f2 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08b0d807e86b4dda9995129caa19a0c5 : SUCCESS in 10m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cde5187b24a943f197005671741fa732 : SUCCESS in 7m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/78ae5fb0b2164088959e4763552cd860 : SUCCESS in 10m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e236dd5bee6b43779aad113c9a444aa9 : SUCCESS in 16m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/02ea4d7a2a4d4eb895fa3a2c53c7d983 : SUCCESS in 12m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1684517e1b0749f18430ed2ff31bd2b6 : SUCCESS in 7m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/826ac671a38e47328cea1d77696f750a : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3dd8b1088f7646af96ef469c55349488 : SUCCESS in 8m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c4b517fbcbdb4f328685279fa9260450 : SUCCESS in 16m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bea0b25f0bd94b5fad1439354ae5c64a : SUCCESS in 30m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d1b576321c6744a5b82bb1f7ef7d5056 : FAILURE in 15m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/748ccdc005d04c31bd1590a973748da7 : SUCCESS in 18m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/694857d0162e45888bdc912b65d81e7e : SUCCESS in 7m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89b280fe9973478fbc8626fe144e687e : SUCCESS in 10m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d1b81420b6c442ff91bf620fd98c92c1 : SUCCESS in 14m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9f8d55a2574453c9ab2ec917f74830f : SUCCESS in 1h 08m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/205ea63021ad46a6a2a8792645ceed67 : SUCCESS in 1h 40m 58s","accounts_in_message":[],"_revision_number":4},{"id":"83a267f3b5f5684efa78e087a3e54f344d0b86ca","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-15 20:37:01.000000000","message":"Patch Set 4:\n\nLooks like this will need https://review.opendev.org/#/c/714676/ to merge before we can update.","accounts_in_message":[],"_revision_number":4},{"id":"078359052e08db5b1d7628a6e00e0e9687c9ae06","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-29 11:51:19.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"be0c33a914a2227f6968dee8d9cd9afdd97f6129","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 12:57:15.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"88b9473669fd50485eb8b8ee50a58b2969778893","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-29 15:12:13.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"858fb92e9515acba829db72ef5a70ff85ebda4cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 15:26:40.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"0766946215967f685af8478eeb7f9418c67f9ad1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 16:16:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91339240a2c64bfbb5fe29aa8572206d : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a612e6d980a440fea43ab1fb90417db9 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c22cfdfdb5c4d9db2579140c8399bbc : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe7426f12b4748029478cced736050f4 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/955bf058fac94278a0ff4d4f6e69fe3e : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b37142fa87ec47828b8144743e8edd0b : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c8ce502655f24412b236a0cf9fe407da : SUCCESS in 12m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1a6409d00d1544c98fa1f832224eb438 : SUCCESS in 12m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/637ae70634644c48b7496d80968565b2 : SUCCESS in 9m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7dfdb1c294ac41948ceabd370a022e57 : SUCCESS in 5m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/270a9c4fbc284cdd934ce53c0fabe4c0 : SUCCESS in 6m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3696dd73490545efac3f1d4d08d94820 : SUCCESS in 5m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0678263afe45400291c0e3724dd2e181 : SUCCESS in 7m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8cb7ec070147491f846cf2465833181c : SUCCESS in 43m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/efcd3fa8e83c4b78b6ecc9fd25442eb9 : SUCCESS in 18m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ef26ef0756fe4687a33bbb98bb554143 : SUCCESS in 14m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f8d815a17a744a94844b1e26692eeec5 : SUCCESS in 19m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/41064b0e5b9446a3a8a56041a67f0db8 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc881ce5533a456e94dd59c8a2841be7 : SUCCESS in 7m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a0cc9ca83d0644e9843db25072132501 : SUCCESS in 12m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76ccb48ddb5743af91a95a4adfd13674 : SUCCESS in 1h 08m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25db646b9a80446395e6d95c5428e88e : SUCCESS in 54m 52s","accounts_in_message":[],"_revision_number":4},{"id":"666f80ac2d072ed41b3ea11ecc94f4b644a3105b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-29 16:18:54.000000000","message":"Abandoned\n\nHas now been picked up by nightly updates.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ead49ad71c6d6835a619551c9100e60738b8c309","revisions":{"80585e3feda1cf56550c5d08da7f4b8e6f2eac9d":{"kind":"REWORK","_number":1,"created":"2020-06-16 20:46:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/736030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/736030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/736030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/736030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/736030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/736030/1"}}},"commit":{"parents":[{"commit":"64af35b5bdce2a5ff829d136dd2cf4ede019c5d4","subject":"Merge \"Update projects.txt for rename\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64af35b5bdce2a5ff829d136dd2cf4ede019c5d4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-16 20:46:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-16 20:46:37.000000000","tz":0},"subject":"update constraint for oslotest to new release 4.4.0","message":"update constraint for oslotest to new release 4.4.0\n\nChange-Id: Ie9c304546c058c895cc42d41ba766e034cdf6e0d\nmeta:version: 4.4.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I5f2d9e775e166f33dc122a6addad0e48602b55e3\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80585e3feda1cf56550c5d08da7f4b8e6f2eac9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80585e3feda1cf56550c5d08da7f4b8e6f2eac9d"}]},"branch":"refs/heads/master"},"4e7be8df86529e3b0ff9621a44a57c823557e06e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-17 09:49:40.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/30/736030/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/736030/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/736030/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/736030/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/736030/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/736030/2"}}},"commit":{"parents":[{"commit":"64af35b5bdce2a5ff829d136dd2cf4ede019c5d4","subject":"Merge \"Update projects.txt for rename\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64af35b5bdce2a5ff829d136dd2cf4ede019c5d4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-16 20:46:37.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-17 09:49:38.000000000","tz":0},"subject":"update constraint for oslotest to new release 4.4.0","message":"update constraint for oslotest to new release 4.4.0\n\nDepends-on: https://review.opendev.org/736143\nDepends-on: https://review.opendev.org/736146\nDepends-on: https://review.opendev.org/714676\nDepends-on: https://review.opendev.org/734442\n\nChange-Id: Ie9c304546c058c895cc42d41ba766e034cdf6e0d\nmeta:version: 4.4.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I5f2d9e775e166f33dc122a6addad0e48602b55e3\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e7be8df86529e3b0ff9621a44a57c823557e06e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4e7be8df86529e3b0ff9621a44a57c823557e06e"}]},"branch":"refs/heads/master"},"a2a3043698d6b02538b30fca652dc3f5a726f7a3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-17 09:51:07.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/30/736030/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/736030/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/736030/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/736030/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/736030/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/736030/3"}}},"commit":{"parents":[{"commit":"6541ca39cae14cd5c77b6e6071b342602f5f80d3","subject":"Merge \"update constraint for oslo.concurrency to new release 4.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6541ca39cae14cd5c77b6e6071b342602f5f80d3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-16 20:46:37.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-17 09:51:07.000000000","tz":0},"subject":"update constraint for oslotest to new release 4.4.0","message":"update constraint for oslotest to new release 4.4.0\n\nDepends-on: https://review.opendev.org/736143\nDepends-on: https://review.opendev.org/736146\nDepends-on: https://review.opendev.org/714676\nDepends-on: https://review.opendev.org/734442\n\nChange-Id: Ie9c304546c058c895cc42d41ba766e034cdf6e0d\nmeta:version: 4.4.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I5f2d9e775e166f33dc122a6addad0e48602b55e3\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2a3043698d6b02538b30fca652dc3f5a726f7a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2a3043698d6b02538b30fca652dc3f5a726f7a3"}]},"branch":"refs/heads/master"},"ead49ad71c6d6835a619551c9100e60738b8c309":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-06-17 09:54:27.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/30/736030/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/736030/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/736030/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/736030/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/736030/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/736030/4"}}},"commit":{"parents":[{"commit":"6541ca39cae14cd5c77b6e6071b342602f5f80d3","subject":"Merge \"update constraint for oslo.concurrency to new release 4.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6541ca39cae14cd5c77b6e6071b342602f5f80d3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-16 20:46:37.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-17 09:54:25.000000000","tz":0},"subject":"update constraint for oslotest to new release 4.4.0","message":"update constraint for oslotest to new release 4.4.0\n\nDepends-on: https://review.opendev.org/736143\nDepends-on: https://review.opendev.org/736146\nDepends-on: https://review.opendev.org/734442\n\nChange-Id: Ie9c304546c058c895cc42d41ba766e034cdf6e0d\nmeta:version: 4.4.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I5f2d9e775e166f33dc122a6addad0e48602b55e3\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ead49ad71c6d6835a619551c9100e60738b8c309"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ead49ad71c6d6835a619551c9100e60738b8c309"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
