)]}'
{"id":"openstack%2Frequirements~736037","triplet_id":"openstack%2Frequirements~master~I68fae187bb9f5aac377ed37ab097c4a135682fa3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I68fae187bb9f5aac377ed37ab097c4a135682fa3","subject":"update constraint for oslo.concurrency to new release 4.1.1","status":"MERGED","created":"2020-06-16 20:54:55.000000000","updated":"2020-06-17 05:28:14.000000000","submitted":"2020-06-17 05:26:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736037-1592371604495-2f414a6f","meta_rev_id":"a8935930f6d16c32bca7d6c2bcff2df9da142796","_number":736037,"virtual_id_number":736037,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 05:26:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-17 03:59:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-06-17 03:59:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 03:59:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-17 05:26:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"214b3546e6a991993435999cefb4a640961f6a87","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-16 20:54:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"016f1c2c957dea79137fea8ab746a5cec705950f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 21:42:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"59b4c2bd8c2112789ddfb4e512b11a300a217cf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 22:52:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6230a0cb8e454471aab5832b2b260aae : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80fd60aa7520468ea035f5d7781a2a65 : SUCCESS in 8m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30cac70d1a554be797481e43df5a860b : SUCCESS in 9m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a94d53301014f83b934b67d8c38e818 : SUCCESS in 7m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/312a568a4e774081b18249091b12ae65 : SUCCESS in 7m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ca6515dfb5664922952f4d0927257b98 : SUCCESS in 7m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/750330e20ba847b3a03c100db483aa42 : SUCCESS in 11m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8b8d4280044141eb91eda618195ed3ec : SUCCESS in 12m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dbf7e964e447496983d62b617be08512 : SUCCESS in 11m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5c17f62a433044f7b1906f457ebf97cf : SUCCESS in 7m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/67ad401bc1264065868ed9e3a1a035e3 : SUCCESS in 12m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63ef76f9045242968709f0d957c2c4bc : SUCCESS in 11m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8c3f34f92e4f4d0fbecc0d0a4dc7034b : SUCCESS in 23m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c79149ddd49943328baab65e6ee2ab58 : SUCCESS in 26m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cac84250876140639830817a35b38349 : SUCCESS in 15m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5eb1ed3f23e74fb88f0563df159c6772 : SUCCESS in 15m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0cd6517f09284991a309480dc0f2449a : SUCCESS in 9m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/26c4734e7c1f4454ad9fd42eecfcee03 : SUCCESS in 11m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1fd2985d7b6843da8b7d2dbaa6319a4b : SUCCESS in 14m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7432adbe748d41f3948c0aab82acb0d0 : SUCCESS in 52m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0390053d5f1f4e52b3b5d12ca6012007 : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a9f8291ae91c8a9a05c705fe887e6b6222ed7f81","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-17 03:59:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f48a139499d99949e8b789ca0ebaac930da419e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 04:00:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6f8e18a196453a575a2eb97c9a122585600b168b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 05:26:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9312abba5ce139b12220c3bd092fb82968ce651b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 05:26:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15ea692ca3034b88b368f4e213d26338 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ccac09c236542feb7590bc6f0c7d268 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6fc86ddc2894f0f96a57e13ad201e9c : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba902620b1904ab3a5d3c46cbccdd735 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e12159339dc425e9d59fd5f40c5b521 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2d75d4d766134be7b33716b4596822a6 : SUCCESS in 3m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0a8fa40566f34395826e9547a89bc201 : SUCCESS in 9m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f36f7ee714464144bca31dc95bc3ae24 : SUCCESS in 10m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0b2a85f9f6b64dbfb2809f059642d249 : SUCCESS in 8m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8b0faaecfdac452eae2bf2358a89060d : SUCCESS in 4m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1a7b98eaf0c7437b9af6370dd2016c39 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/208ff40393804ed3bce8b64994d85926 : SUCCESS in 11m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4070f014d23a41f0a1d3807da466d9be : SUCCESS in 13m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8061cdc0ed5e40028723034a0d015218 : SUCCESS in 34m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a993b2769bac4531ad5f93f0c2bba0cf : SUCCESS in 14m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6016621a06714d5784d83b452b7bb8ac : SUCCESS in 19m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6339477fc5054327b4da8199bcbf6bae : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f92c70bef23441a81d18ffaace1278f : SUCCESS in 6m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/357234078c634867a0bcb1b4972caaff : SUCCESS in 18m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1942100d5882439d8bc891b3210fe98e : SUCCESS in 18m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed8283d37f5c48aeb09d9767586a083c : SUCCESS in 50m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d46df931c424d3eab74c5015e9d4f1f : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1645ee15ee943a5b3a70b66ccb9c8f9c25ae2721","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 05:28:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aa313527d8c4d868aa09559d424732a : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e15a01425c38e4b955daec806872075d4ab69747","revisions":{"e15a01425c38e4b955daec806872075d4ab69747":{"kind":"REWORK","_number":1,"created":"2020-06-16 20:54:55.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/37/736037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/736037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/736037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/736037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/736037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/736037/1"}}},"commit":{"parents":[{"commit":"64af35b5bdce2a5ff829d136dd2cf4ede019c5d4","subject":"Merge \"Update projects.txt for rename\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64af35b5bdce2a5ff829d136dd2cf4ede019c5d4"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-16 20:54:53.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-16 20:54:53.000000000","tz":-300},"subject":"update constraint for oslo.concurrency to new release 4.1.1","message":"update constraint for oslo.concurrency to new release 4.1.1\n\nChange-Id: I68fae187bb9f5aac377ed37ab097c4a135682fa3\nmeta:version: 4.1.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I5f2d9e775e166f33dc122a6addad0e48602b55e3\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e15a01425c38e4b955daec806872075d4ab69747"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e15a01425c38e4b955daec806872075d4ab69747"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
