)]}'
{"id":"openstack%2Frequirements~736382","triplet_id":"openstack%2Frequirements~master~Id4969156383545eb94462677e3321b89c1160b48","project":"openstack/requirements","branch":"master","topic":"ensure-pip","hashtags":[],"change_id":"Id4969156383545eb94462677e3321b89c1160b48","subject":"Ensure pip is installed locally","status":"MERGED","created":"2020-06-17 21:21:38.000000000","updated":"2020-06-19 15:21:33.000000000","submitted":"2020-06-18 07:50:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736382-1592466645090-87c47c09","meta_rev_id":"1ab6e353e61ac2990b9635b6e3ce64bd5006989d","_number":736382,"virtual_id_number":736382,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 07:50:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-18 02:10:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-06-18 02:10:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 02:10:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-18 07:50:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aee3c4d6dffb3149ef8420874a133594fa86857f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-17 21:21:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90476bf33a349ccc1621e489354e45728d8b43f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 02:00:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b217352ba6cd406db5acaf0258adc1b8 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e07d0facc9554329a18a2dd34adde5b9 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d3b3f1d3fe84aecb597f3e3b2b033d5 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93a8a6d373fe4b9483f021694903cbfc : SUCCESS in 6m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/999c9763e6bf462c8afe2fa8cb93440f : SUCCESS in 5m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60c4b8599ae74971aadbb469ffdfa533 : SUCCESS in 51m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b46ee9ea12d431e960c312d6422ea8b : SUCCESS in 1h 43m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7bb66d336802738d980b5a0319a185cd5a775b80","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-18 02:10:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46f52110b95f5943a0b27db71ddb959a9f34724d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 02:10:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"759a1a20bdac5b65313c6d6e0dff554f1358c6ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 07:50:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2030b786979481d9088bac5ab36b0a6 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9721e73d075e4a94a2a142c859266fe1 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72e4f04d2ea7401f996519dcd3a58f3a : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8e990bec68e4ecc914b4bac719c1a10 : SUCCESS in 6m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fb78dac0aad245e5b016499440da0853 : SUCCESS in 6m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb7df7663ca748cbb238a83543a8227f : SUCCESS in 53m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51273fdb973047158c82497a7352fd81 : SUCCESS in 1h 19m 23s","accounts_in_message":[],"_revision_number":1},{"id":"7cf4734709a93b9bf6ef4c8364abed473e163d32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 07:50:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f639a4bfa7fcf1983a9107578fec5d880c77cbca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 07:54:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef609742d32b482c80fda1df4a570899 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"919c2c51d767710c7e50f1046d670c78965ec009","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-19 15:21:33.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit d303fe350d74d2d40aa5a591ee9d372a529d10be","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74a5bed896893ecd3bdedb100439f773af7c7f5e","revisions":{"74a5bed896893ecd3bdedb100439f773af7c7f5e":{"kind":"REWORK","_number":1,"created":"2020-06-17 21:21:38.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/82/736382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/736382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/736382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/736382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/736382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/736382/1"}}},"commit":{"parents":[{"commit":"1b6e378b4644eae14d130ffc83df99d384a89f42","subject":"Merge \"Remove grenade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b6e378b4644eae14d130ffc83df99d384a89f42"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-17 21:19:06.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-17 21:19:06.000000000","tz":-300},"subject":"Ensure pip is installed locally","message":"Ensure pip is installed locally\n\nRecent changes to the base nodepool images removed the system-wide\ninstallation of pip and virtualenv. As a result of this, some\nrequirements-check jobs are failing with an error that pip could not be\nfound.\n\nThis adds the ensure-pip role as part of pre_tasks in our\nrequirements-check playbook.\n\nChange-Id: Id4969156383545eb94462677e3321b89c1160b48\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74a5bed896893ecd3bdedb100439f773af7c7f5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74a5bed896893ecd3bdedb100439f773af7c7f5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
