)]}'
{"id":"openstack%2Frequirements~736782","triplet_id":"openstack%2Frequirements~master~Ifb677e8839b736164e7fe9bf53cb0856bf5457e3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifb677e8839b736164e7fe9bf53cb0856bf5457e3","subject":"update constraint for python-watcherclient to new release 3.1.0","status":"MERGED","created":"2020-06-18 17:16:04.000000000","updated":"2020-07-16 00:14:04.000000000","submitted":"2020-07-16 00:12:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736782-1594858347469-bfbdc47d","meta_rev_id":"f0fb48880d2d6307a53bcf15e083346704645c2c","_number":736782,"virtual_id_number":736782,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-16 00:12:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-24 20:30:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-06-24 20:30:47.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-24 20:30:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-24 20:30:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 15:59:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-24 16:00:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-24 20:30:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-16 00:12:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af02f5650b82500df17d32b6918182d33755106c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-18 17:16:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"405313eb6d71df62e3009d17dc7fd6399f49c2a4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-18 17:41:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1a04c7d73388b556bfbb41b3768f8abfe5e09f22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 17:45:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"01243b61749d148ec66ff044d18c13ea98c162bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 19:28:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5edd3b54591465e9e6da8bdfbbced48 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/689a3ab7c7974c3a9b98370c06357522 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a20d028ef45e47209207b87057821054 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97231db588e543079c21c98057f85032 : SUCCESS in 6m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/51df567469034c089fc9f036384a225b : SUCCESS in 6m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/21d61591cb094a148b2f7f0fcfc1fe18 : SUCCESS in 5m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9d237c4b5a54c3cadc7e2838cb75b32 : SUCCESS in 9m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bb92c59af5134dfda8f783e505cad3af : SUCCESS in 13m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8a9f1d46ba064e758ed64b2b6cbea502 : SUCCESS in 10m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f000089581ec43f19bbab72c1c5f7946 : SUCCESS in 6m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7f53e4e953c14fc1a92822760084ac58 : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a7ff054ff4f543e7bce1cea6a65a5103 : SUCCESS in 10m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1bc7682b6e07485dbb4d7e61f44b213e : SUCCESS in 12m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9b74b0b93f48420ca20e8e711f8d469e : SUCCESS in 30m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b5d1956ff1f8468ba133399f0c17475d : SUCCESS in 18m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ff3fa46c16e342049d1d39d2c4f315e3 : SUCCESS in 18m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b43ae6dd21394c1cbc7b0ce4999babba : SUCCESS in 8m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8eb105f19b9745a587670fd71403b27b : SUCCESS in 9m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e9dfa6b1668e4589a1becdb33833da71 : SUCCESS in 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4a90a1ae5934f06812605dbf08dd15c : SUCCESS in 51m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ae16b9c88554dfd88a06bed2d3e33e8 : SUCCESS in 1h 11m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5860b774ed40c42034005af49d9c2e1c3305f3ff","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-18 19:30:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"33cc53f11277500a8aa23b893423cb977b50cebe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 19:31:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52a3e83f99075302f1b8078bb46da09a9962525e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 00:13:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/348334e47f364ba3ac2c7aa73fd0f376 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/878dc9df1dc6476180c108981cee4b3a : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ba15501722246bf97d9d0a70afd209c : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96f468359a9c40aba39a79580fac669a : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a25603c60f594f2db6ec6772b5984b0d : SUCCESS in 11m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e56d8c453bc43e1bb151319ca50baa3 : SUCCESS in 10m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/76e757ca58164342b33d269d5bd24831 : SUCCESS in 28m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/50d4616521ae4eb28572bc1b91f6f931 : SUCCESS in 28m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ecfaa4bf1711463eb3c2437d4ac74dac : SUCCESS in 26m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/79b894bef6d249eb9bb73b71a5e5e805 : SUCCESS in 14m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/401ffcd2df9649598e622aad1b1cd864 : SUCCESS in 17m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/219f087348a04188ba9954fdeada598f : RETRY_LIMIT in 5m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5adcc6a109ae4cc296fd4d71443db360 : FAILURE in 53m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f935964cf084456081884fff7f047145 : TIMED_OUT in 41m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bedbf3fc8015468889c4d77fc6a068b0 : SUCCESS in 22m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24dd095144e547c6866a6fcd4f214526 : SUCCESS in 37m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c3fb5a03ec364d429b31df2122ed4828 : SUCCESS in 6m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c677eb6990c64e20bd60d0e5c8c030ac : SUCCESS in 16m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c9173ad68d0d473b876a8d2f384f98c7 : SUCCESS in 18m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a5a84c4f98154af7976ca5830881c461 : SUCCESS in 20m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de7b7552b4b946e29298c219dd21dd3a : SUCCESS in 57m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a52c8f645b5145ff86f59df67a0597b2 : SUCCESS in 1h 21m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8e3012178a3213591b63c98ff1a23848a81bb0c5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-19 02:39:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a92380a51d48528dad8354dde0e92ba60095baf1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 02:58:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"413fdb486b07138c593516f5f60edc5f363f4fae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 04:11:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af2657ee367f463a8a79c642955a6985 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f35e7f4ed814c1da9c9a4ed88532cf5 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da98f9cc8f724fafa995c29f008e3b21 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad7edc2310de453d98ad48b151fce00b : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b067706b82645eab31bf8881a5701f3 : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e2af7a933d24269abb06a024b2b8f15 : SUCCESS in 6m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0527e3d53d8a4170b733d55359c2c8c3 : SUCCESS in 12m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/10e8f65ef08642998bae760f67c27c6f : SUCCESS in 10m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9112a9ab5a734209905a222e3a34c680 : SUCCESS in 11m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b78e355d83144418a28e324c0af37681 : SUCCESS in 8m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/80c017ad5c1b4000932ac22dc60bcf37 : SUCCESS in 12m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/80a4b321d2d44f38982245d7f306a504 : SUCCESS in 9m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c870ece6054947c38b333d8583d80943 : SUCCESS in 22m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0aa5209a670d4af8bdfb07180f49d278 : SUCCESS in 27m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7f90e8271f0d4163ad042494016bfecf : SUCCESS in 19m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc266c11a5904b088a04a68a99fd0006 : SUCCESS in 18m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0df9c6c3d25a49c6a8491080c2a397cc : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/332727fa736549b595c445806dbac17c : SUCCESS in 9m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/97d8deb7f9854a4498ba0cd0bfadfe73 : SUCCESS in 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4eb8d5f43d7c411b8998553b06f88bf5 : SUCCESS in 58m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8fd995a78524521bc780ca6ac008b74 : SUCCESS in 1h 29m 26s","accounts_in_message":[],"_revision_number":1},{"id":"edb3d60bb30c2560e936d5a35218d964f14c09e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 04:11:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d4344bf217448d9f6c3b2bd6b1bfb72d150d0bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 05:28:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab102b0ef551406fa6f65f6b4f751fa9 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77f583d9cc4045f793a0be9a706e3c98 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a6584ff5f33492ab31a3ec6768fdab6 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f541673987fa4c9289f72c6af3bcb064 : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a79a0e9a3af4dc580f0a91751aa433a : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4090b58e0fc34cff89e4009109500706 : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e25f36b5f2a049108b80f8d502c53461 : SUCCESS in 8m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/54e43bdfbc90431297d1db5f4cf80d0b : SUCCESS in 9m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4a5e9a9b51ba483aa9fdc24a740fbb0e : SUCCESS in 9m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dc38ef51d6ab4b79bbf60de6ab559e5f : SUCCESS in 5m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1c3efac50b084c3e8c444ceac4b99c78 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/94ad569b055b43ecbd36211e8a78c932 : RETRY_LIMIT in 4m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9c094fb164dd47eebc11922b5520ff8a : SUCCESS in 18m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/90d1ada55c014cd0b70e74bf1198296a : SUCCESS in 23m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5009156de0414301bd0e341c0df30d17 : SUCCESS in 14m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/feb7fc32b6524cc988a05f819ef263ae : SUCCESS in 17m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9437c9b9fcd1460a97e4336d5be3ad56 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f10e708c0ff44bab9d8e52aa06e4d71 : SUCCESS in 7m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/94e1522c9cee48b290d623af77b1b2c0 : SUCCESS in 12m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/24bc62678c0e4dabafba2590a5c594d3 : SUCCESS in 14m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7c47dafcc2d4f8b8fb8c3f2f8b61811 : SUCCESS in 37m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/225c12f509634abeba292fcf9f49957e : SUCCESS in 1h 16m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0b9257cb485c78ebaa845d24c1be84a97b13ca88","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-22 15:59:41.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e7eb5b404ad1d655b3ec8842de44f523975fc206","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:18:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/92bef79e99724388b61a9616a3c10d0c : SUCCESS in 14m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7ae62362c8c738b74cb2fc1c58493221c247b580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 20:13:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ee03be42e404958a99001f61d85e947 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b207f059c5542c2b8a291913091ebbd : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac656d3090ba42ecbbd85e121c798fd3 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03283a46f1734a2bb2faa699a3b8bbae : SUCCESS in 6m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/761dc79703fe41a390a354bea28d05d7 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a47185e39b9741228a3ef0a7f60d5a67 : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c46448a3c0df4ca3926d1adf4bd8da70 : SUCCESS in 11m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1ed60768db794790bc08f768f78d05b5 : SUCCESS in 10m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f7ce0dca0ba74242a42f8b0c0dba0036 : SUCCESS in 12m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2498ed2388df4f80b2391055053d31c4 : SUCCESS in 12m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/92ef8cd00b294811a0000eff37776353 : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2547bf07d73b4c8ea3db3dbca44f26c8 : RETRY_LIMIT in 6m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7a74b9a4e82b41129ffce56fa0d54a98 : SUCCESS in 14m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/75abd06427f941298a090807b1a2c146 : SUCCESS in 30m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e00379926dbd445f8e34d622b32e1294 : SUCCESS in 18m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/87aadb272ba84e90ab7778643b22e87f : SUCCESS in 18m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2d04313b758140969b50cd12badbf5a3 : SUCCESS in 7m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ef87638f1334a6fb6c3d960bbf42ebf : SUCCESS in 9m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b262b5847b6045b7b4a215fb3a9198cd : SUCCESS in 13m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c55ef8e0be65456fa1d0789eb6bbdb1b : SUCCESS in 1h 13m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e7942709be2466fb10d726b8e088355 : SUCCESS in 1h 39m 59s","accounts_in_message":[],"_revision_number":1},{"id":"53842a89cd31174020768c52a80c1c976b849c49","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 23:34:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"62a37056fd5591aba039686597096268d642e329","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 23:52:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2be84c32f50b4774b85f70502294b9d3 : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":2},{"id":"85d519a886d6b26d77c0a3356086742a114ab4d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 00:59:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4f0c7fb2a84e3fb5e8cdc6bc0b0d96 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97640b830008415abe2d45c0dc587d5f : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1003805e1e9c476c9f20a7d707a99124 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1848d16a817048ce8bc2a45bb955fdbb : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf2005a3aa4a45249d2da571a11da960 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b7be4ce6172e4f1ba1c93abe50aa676c : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9856a2bcf9b04a7a879baee20a15ced3 : SUCCESS in 8m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1bbe2ecb441e470c8c22c98a99a241ad : SUCCESS in 9m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/85d2e7eceb614d3a977b9d910aeb997a : SUCCESS in 8m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c33d622b3c354590bf24f0fb5ed687ae : SUCCESS in 5m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b6d893f000c042f6923f62185e4cc48d : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da6e568110764015a4480f4dfa98b027 : SUCCESS in 7m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7eda12d101684d1e98c1e3e8b1682a43 : SUCCESS in 13m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1df0b1106a244715b12cc870cdaf3fe9 : SUCCESS in 27m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4bf77602715d4f3db29d6e911928b015 : SUCCESS in 13m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b5961509cf704fd9bfdec56a5bfa955c : SUCCESS in 18m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/89f9baf0db2d40109c314d4165dc9012 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ed516174d7042fbb6ee0453ec3f9365 : SUCCESS in 6m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bcac6820ed6b4551881cdf95c1c96e4b : SUCCESS in 10m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7df51ab7cc18482abf473327f7386a4c : SUCCESS in 47m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc51fc736df549509013226958433765 : SUCCESS in 1h 20m 42s","accounts_in_message":[],"_revision_number":2},{"id":"a62a56add61e5cd96a9502ee95e7389728b4b5fe","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-23 01:03:42.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"64a8db33b85d0920308bcb9b7144f6a50f8d594e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 01:04:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a5d2ee6df77caf8dc07a9dbe09e49f5b4a22ec01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 02:19:36.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"7c08bf0738189dcdb12cb506619c542da1414e5d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 16:00:42.000000000","message":"Patch Set 2: -Code-Review\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a0887b605e84e987ec2a0bb0f7aac0b933150863","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 16:20:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5b7c95b537b24e7aa81c22763ca80a9d : SUCCESS in 15m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9522f11cf5412eaae93fce0d92c27df23ae24521","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 20:30:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2506dbd9c41f81b778c967efd587cf7e1c497f2f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 20:30:49.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"36c0205ed6810a42bce96148c6d75b8998f4c1b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 20:53:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e241d081f19d4d51bc360567a39fa8af : SUCCESS in 14m 46s","accounts_in_message":[],"_revision_number":3},{"id":"5a8e8c919e7b48e11947101b1ecf2866dadc7aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 23:24:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0065a01a4a24489784c869be43e84403 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cff72c1b23b4944a816116c02d341c1 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/480d613854914d8daa3c3606d73d69c3 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d98959c735484d0c8b665f7b06f746a4 : SUCCESS in 5m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b176d6a6974c40d2ac6e3b962f7a1ce0 : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/388c3da320264ea3acd846b12fa5ced5 : SUCCESS in 6m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6facdc20e6e44a0b8362cd5d78a6009b : SUCCESS in 12m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/491fc95a83624d3b88ecc0b4e940ee7d : SUCCESS in 12m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e6f148af8b9d4cc78062fe5bdd4e0482 : SUCCESS in 9m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3f7f6281bcad4e439dfcbb36622faf2f : SUCCESS in 7m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/94b23d28c43743fbb98d1906c2e51f71 : SUCCESS in 11m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2818c161b3c040c0b90a2f346742ba0e : SUCCESS in 10m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7e33eeb26c6d4b979b7c9a951b77789e : SUCCESS in 20m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c59a411e38814fe886013335b4187a92 : SUCCESS in 28m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3fab41079f5349a1a5b80e520aa6eef5 : SUCCESS in 15m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/80ab86192e3e4c66882a1955cab1afe1 : SUCCESS in 17m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d6311d90d7cd456a9a60e1a2d08ff391 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53f2c8a863f84d82adcc18c1dcd69c66 : SUCCESS in 7m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/71ba02005c60493fa932a28f822d3cfa : SUCCESS in 10m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71a40bcdfbcf4f96af1c69b3652ddc22 : SUCCESS in 57m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56e281d6e8a046c39e691fffc81704e0 : FAILURE in 1h 04m 03s","accounts_in_message":[],"_revision_number":3},{"id":"55e9d30447c2c0f8ac277bdfb7c8e363f2f8e215","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-15 20:35:25.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4c2eaaf056fefe8725c6997a679f21dc7bad69ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 21:14:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1025da77ae6a4193be7434c19b3df4dd : SUCCESS in 34m 52s","accounts_in_message":[],"_revision_number":3},{"id":"c9acdf65b23f395fad6127786fcd63ae289a818b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 22:46:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4859372b2ab2458db0ac26af622b175b : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bce8015e889b41a884471bdff0e7aaf8 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a52acb658e0461385c5c4b7bd2df7c7 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78d2dd5f1de048f7a62b556fc6729084 : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/77a9285991244fac9839724f465c7b08 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bdbe5cd91eb9498bba064e6b1b0c2d4a : SUCCESS in 3m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/26746c1f92bb4b01b778a5c193841d99 : SUCCESS in 8m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1788466c98534db6ba5887b1045186f5 : SUCCESS in 10m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3dfcd8b7985b4b40876660ae2cca22a7 : SUCCESS in 10m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e2abb24dbbd14f20a611789f5db0139d : SUCCESS in 5m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/df6aad42005649929c3b88186bc34735 : SUCCESS in 12m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9081d020a8034e6d835eaee1091c0cc8 : SUCCESS in 7m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/55c4a7f080cf453a88620d5bcb3b6b2b : SUCCESS in 7m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b88dff4a8ec34863a44f04f0da058aeb : SUCCESS in 15m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/45fbfc2a0082482bba810627834ffbc5 : SUCCESS in 25m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f0bfc44099f74fb3a30b28568cf12413 : SUCCESS in 14m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd6143717a11420d8c48b2ff82f58dde : SUCCESS in 17m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/35fef4dd4b7f4158b2076147d6ff496b : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ab305bb5b35d4fd1beb37373f1003024 : SUCCESS in 7m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/04b553ef108c4eef9128f209d8c2822a : SUCCESS in 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d580e021f0cb4b5c8257fc8c5917f896 : SUCCESS in 39m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5fc565a0e87e4753be0a9a6c3e63f314 : SUCCESS in 1h 15m 42s","accounts_in_message":[],"_revision_number":3},{"id":"4311636bddec10a892e48b25eb3db0966caafb40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 22:46:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"92fb298325ecae7ccb72592135e955bc88b63d6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 00:12:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/588ada7087054fcb8d17c3018fd30a6e : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b333967d336497ab319cf3a50cb98e1 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f158744bbeef4067bb638b0c10e81681 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f17178ee730b4068a71fd4185dffa02e : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2f23eb96fe144198d841e920bcb37ec : SUCCESS in 5m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc9740f71949461f81cb3e2911db0dfe : SUCCESS in 5m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0ee2943da6c4420b8a18b4f98f88c6d8 : SUCCESS in 7m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e11a17873385476db4f0317740aa0426 : SUCCESS in 8m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1cd5b087c19f425982abb876fb4b9470 : SUCCESS in 9m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d1bedd440386438f995d7be42b7d2a88 : SUCCESS in 6m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1d14383186ee4ef9977daf57d4ce58b8 : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c8af6a35890e4d188cdbd153c9d594d4 : SUCCESS in 7m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1a1d69b152904a08a3b92fbbf3b6f2ea : SUCCESS in 7m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/38782ddc28384cefbf42b3af135f497a : SUCCESS in 15m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8a081d28ada94c91bcf310b7290702f8 : SUCCESS in 33m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9e61686410e347049828a28b2d6647d7 : SUCCESS in 11m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9697eb1ea8a64809b8e5dff218965183 : SUCCESS in 17m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5b2adc8b146e4600b272ff0fc3cb62f0 : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f008d9bceb24b46b4b62b91e018a0e5 : SUCCESS in 6m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a9edb1c8753049ef977c04ea14f766ce : SUCCESS in 10m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bb5a0bf79419439b9359e3187646e4e8 : SUCCESS in 8m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65fc92c37be44ce0b1b8f6c9cbc52f84 : SUCCESS in 1h 04m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9cc371a70b644419cc38d3af35ddb75 : SUCCESS in 1h 18m 00s","accounts_in_message":[],"_revision_number":3},{"id":"f6523bc8d6003b83cb1c5a26d2de86b3daf6ab8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 00:12:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f0982e79b9a87aa710fc7e8fd0329f67ee0975b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 00:14:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef2e6bf260d4c5e9bc399039a39c3aa : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6da3a399716319b136b5595f2dc75b8532f21272","revisions":{"ea8c078b20435885ceebb94d1328343855552b60":{"kind":"REWORK","_number":1,"created":"2020-06-18 17:16:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/736782/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/736782/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/736782/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/736782/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/736782/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/736782/1"}}},"commit":{"parents":[{"commit":"69340b77185d22ef9b22598a0955b0896150694c","subject":"Merge \"Remove mox3 lib from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69340b77185d22ef9b22598a0955b0896150694c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-18 17:16:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-18 17:16:02.000000000","tz":0},"subject":"update constraint for python-watcherclient to new release 3.1.0","message":"update constraint for python-watcherclient to new release 3.1.0\n\nChange-Id: Ifb677e8839b736164e7fe9bf53cb0856bf5457e3\nmeta:version: 3.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I3d60d4af030824a6b5f7064ca7e846b3cfc6d481\nmeta:release:Code-Review+1: licanwei \u003cli.canwei2@zte.com.cn\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea8c078b20435885ceebb94d1328343855552b60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea8c078b20435885ceebb94d1328343855552b60"}]},"branch":"refs/heads/master"},"de548b7f00c2112fcc473518204d91b866689add":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-22 23:34:26.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/82/736782/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/736782/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/736782/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/736782/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/736782/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/736782/2"}}},"commit":{"parents":[{"commit":"5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc","subject":"Switch node tests to inherit nodejs-run-test-browser.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-18 17:16:02.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-22 23:34:26.000000000","tz":0},"subject":"update constraint for python-watcherclient to new release 3.1.0","message":"update constraint for python-watcherclient to new release 3.1.0\n\nChange-Id: Ifb677e8839b736164e7fe9bf53cb0856bf5457e3\nmeta:version: 3.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I3d60d4af030824a6b5f7064ca7e846b3cfc6d481\nmeta:release:Code-Review+1: licanwei \u003cli.canwei2@zte.com.cn\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de548b7f00c2112fcc473518204d91b866689add"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de548b7f00c2112fcc473518204d91b866689add"}]},"branch":"refs/heads/master"},"6da3a399716319b136b5595f2dc75b8532f21272":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-24 20:30:47.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/82/736782/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/736782/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/736782/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/736782/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/736782/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/736782/3"}}},"commit":{"parents":[{"commit":"c905979a82a51eb48d1cacaf370e178b703f2d4f","subject":"Merge \"Remove sphinxmark from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c905979a82a51eb48d1cacaf370e178b703f2d4f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-18 17:16:02.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-24 20:30:47.000000000","tz":0},"subject":"update constraint for python-watcherclient to new release 3.1.0","message":"update constraint for python-watcherclient to new release 3.1.0\n\nChange-Id: Ifb677e8839b736164e7fe9bf53cb0856bf5457e3\nmeta:version: 3.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I3d60d4af030824a6b5f7064ca7e846b3cfc6d481\nmeta:release:Code-Review+1: licanwei \u003cli.canwei2@zte.com.cn\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6da3a399716319b136b5595f2dc75b8532f21272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6da3a399716319b136b5595f2dc75b8532f21272"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
