)]}'
{"id":"openstack%2Frequirements~736784","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ia5200e7807253b8d05b5847c9a7a6ebfa813e67c","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Ia5200e7807253b8d05b5847c9a7a6ebfa813e67c","subject":"update constraint for os-brick to new release 2.8.6","status":"MERGED","created":"2020-06-18 17:17:14.000000000","updated":"2020-06-25 20:47:27.000000000","submitted":"2020-06-25 20:47:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736784-1593118047534-a308ddd4","meta_rev_id":"bcfe7e1198ed87bfe984b5dae96f6b9598bdd333","_number":736784,"virtual_id_number":736784,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 20:47:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-24 20:02:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-06-24 20:02:15.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-24 20:02:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-06-24 20:02:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-24 20:07:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 17:23:30.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-22 16:00:34.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-24 16:00:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-24 20:07:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-25 20:47:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c50e44cf32dbda6cee1052562f10b6611e175e44","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-18 17:17:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a1245e4af1374954fdce768b244036c68ec9168","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-18 17:23:30.000000000","message":"Patch Set 1: Code-Review+1\n\nThis fixes a regression from 2.8.5, which in turned contained a fix for a security issue (OSSN-0086).","accounts_in_message":[],"_revision_number":1},{"id":"05b033db31ccd82c1ace14102d646294c9f29110","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-18 17:41:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc3dad9fe4025d5e3ff70fca8e1ee8ec88f4bc8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 20:04:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3d996a45ed04f82b0295ef56b4650b8 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8fdfb3a5d7b5458db2004819d78e9ba8 : SUCCESS in 8m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ee69e3d9e4324bd0b2b514ab97d8cc47 : SUCCESS in 8m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ded5e34cb8ac48179333e79734a5412b : SUCCESS in 8m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03b886f6e3274676ace539c3d72db1d2 : SUCCESS in 7m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc3b1c44887c46938b835928a7ca8d1c : SUCCESS in 5m 45s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ae901859c2c247eea3da13c56596db5c : SUCCESS in 17m 51s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/aecb925f9bd14fa8ad6eeb3d60dd485f : SUCCESS in 17m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2ba9c2830fb445c48d40eb62271aa6e1 : SUCCESS in 12m 24s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fe60a8ded1474924ac67a99e40bdc0fd : SUCCESS in 13m 23s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/a40f811627504ef4bd545a4ba8cf1857 : SUCCESS in 19m 59s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7d07b429e8104e6ca1826963ca7f853e : SUCCESS in 8m 50s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/f09cc462e7494ac282daa900606a663e : SUCCESS in 8m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9be4758a69264b6d804d812d0a002891 : SUCCESS in 16m 14s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/8b76a5a66afe40208322225e248f8ad4 : SUCCESS in 13m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1669421a13ca4198bd3a5833d80d4d8f : SUCCESS in 7m 50s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/749bae0d6d6444aa865eb7bd3ed54124 : SUCCESS in 23m 00s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/8778f797738f4be18ac7a79e2085c6e9 : SUCCESS in 28m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/46106060c2c8408f82dc26aefa7ecc78 : SUCCESS in 21m 54s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/610f1d2c36ce486d8e72c8710b5158f9 : SUCCESS in 30m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8ce95c4ded74461f9609c6e5c9acbb99 : SUCCESS in 12m 09s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/ee4854f30aac4ab79704a3182f767d77 : SUCCESS in 13m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/50724db2318f478f898dd0c09e2ffa16 : SUCCESS in 16m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/addc61df65864623aa7c555e9b8329bd : SUCCESS in 11m 13s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/5c74a5d92ecb4a318015b13dea66b248 : SUCCESS in 9m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0bc44b68235a466ea2806bb7af92ea05 : FAILURE in 18m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/35b5149a063840ddaee22aff95ffdd46 : SUCCESS in 1h 44m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ccc2585021984a20852a41475c837432 : FAILURE in 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9ed02d51d1389aa57d55e12d4a598d1ca69d2876","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-18 20:47:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f27a2a1c6970509492e2dee7d5e0991224322634","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 22:22:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/946be573b45e4658afc854833888509a : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/83b31c46daa64b15bc892d2edf12afae : SUCCESS in 7m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/106db948ebc84e1599f779283c124906 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d72a32559dec4e178e1045d69a680c4f : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/21210b302cc448e7851fb360d7285d9e : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/264434d4f1284e66ba57e0c5bc91af33 : SUCCESS in 4m 51s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/13d23c4fb2394c63be51110f9bfb95c3 : SUCCESS in 11m 08s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/875691b7825f47a5a45627008c2b6743 : SUCCESS in 8m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6cc53546faa54a6b8a80a3a4a4b4fd87 : SUCCESS in 11m 17s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ed42e6077a8446d68fcd1b7d4920ec0f : SUCCESS in 9m 42s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/81569e27050c4ea2bab3563e7ce92822 : SUCCESS in 14m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/f101ac8d3a9a4cddb6585493ba6ccd7d : SUCCESS in 5m 59s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/b490bf4db4f14d819cab06786629b6c3 : SUCCESS in 5m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6d25f0c8cbb241158bdfafb805772aae : SUCCESS in 9m 10s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1dbaf9c4cee844dcb7f3e38c028c69d3 : SUCCESS in 17m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a18248edaa4468ab2f198e0359bb43d : SUCCESS in 9m 58s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/efe8897c3ec44af6b49e9004954fad86 : SUCCESS in 24m 35s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/e5af14203b3a4e2ba9167827f6025b2b : SUCCESS in 22m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9757b2d11acb410588397887fedaada1 : SUCCESS in 26m 00s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4b2ac0dc92fc4a8e888e51f092c95bdd : SUCCESS in 24m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1d03cfe6e3ba425e948f4f20953b25b6 : SUCCESS in 15m 43s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/9e91c435e76249fbbf53a6be02eb9566 : SUCCESS in 12m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/77683bf3cc29496f97eed6f45bade0ac : SUCCESS in 21m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c1237ea607184934bb2bbcd0cbe2deb5 : SUCCESS in 11m 37s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/58a6bf6e428345c08b683b68064226e6 : SUCCESS in 11m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d84324d2cec4872a8a826363e68670d : FAILURE in 21m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9644e1682b4b46c182ddef7d9399ef19 : SUCCESS in 1h 29m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5b3095f84e104ddaaaa7618a064458cf : FAILURE in 19m 00s","accounts_in_message":[],"_revision_number":1},{"id":"22ad59f95902b0ae3f09ca4d3f80cbd866ccba78","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-19 02:39:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"98bf293895b63da7ca763485ab076385d163c417","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 04:15:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92dc7b8c543344f99be942fe71e525a3 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f1ec9c2975f4fd0bbb65e9486c7c0bb : SUCCESS in 5m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/407ec06135294432a93753305b9c0bc5 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a87a358e0904e91a7e2179d61be4a24 : SUCCESS in 6m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/146911a6a8354865a52810ecc565f63f : SUCCESS in 6m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/93d2f77328934d5ba9912a943baa71f4 : SUCCESS in 5m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8bd13538bae448fe928041c0e9bb1be9 : SUCCESS in 10m 09s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/1ac6557414db422ea0f4e039ab800726 : SUCCESS in 11m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6e213fc6ecc14ab4b19cf9f34f659233 : SUCCESS in 8m 54s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fb715367177b4e13b1a8b1fc73436d66 : SUCCESS in 11m 12s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/8ce35fa0411c4eddb2149c6ec2407ccf : SUCCESS in 12m 33s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cef4192561e047208b253e694bc40cd0 : SUCCESS in 6m 29s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/4445ceefbd364919ae4ae11e4f78cbac : SUCCESS in 8m 18s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/03b588bb5c434c048693ad84664d9510 : SUCCESS in 12m 56s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/098addf4835744978f0e04a61c96e62e : SUCCESS in 13m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2c34ad13610c44f7bf6b3cb2728baad9 : SUCCESS in 7m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5fe0338df0624eaf9c0dda4130c0e51e : SUCCESS in 23m 00s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/a83c8be8a644465e874d600a267443f4 : SUCCESS in 25m 34s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/258da342b5b24297b76c88cc43c054d8 : SUCCESS in 22m 44s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/19af6dfd5ce04c739fa2163e3a55c807 : SUCCESS in 25m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2e1c100a1a954e9b85caf431b7e8ff84 : SUCCESS in 14m 11s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/6abff8be442b439199e92405d4f01604 : SUCCESS in 18m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/115a78bfb5704dd2b0008ce463f6c22a : SUCCESS in 18m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e6e02226e084413399dc0a2f462411bc : SUCCESS in 12m 49s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/d1234f0d7bd04429950722996df859b1 : SUCCESS in 9m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bfe417cb5b804fe8b54b2074b22d0eac : FAILURE in 27m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/91dacd07aa934841ae5621e6f342e9ae : SUCCESS in 1h 25m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/55bcd01c67d14996a1b8fd8fc87777bb : FAILURE in 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c2a02f1c56eac88d0573b0faf6d2a60b4de87fc6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-19 13:12:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ed395be3b7f64f64e727e950a35ad7f905f3a898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 14:38:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ff9e39fd99b4967b896c217629930e5 : SUCCESS in 7m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86120d85efd648de82d23e6b3a6de36d : SUCCESS in 7m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0decf9b27bd04f52bafbf3fb91ba07ef : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38d968d61795402184319a0507389876 : SUCCESS in 6m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/524d1d62f2b8412a9d5a7257c01cee17 : SUCCESS in 6m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/60251a6820cb41cd9c576e42f9f3d479 : SUCCESS in 4m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c5634c5fbd99438ca0210e5bfca89aec : SUCCESS in 10m 41s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/c9aa801a198e45f1900a384395859670 : SUCCESS in 9m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/be8ab95e1886401ca63bb1744212bc7f : SUCCESS in 9m 37s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/910a329c6e57404d83bd320d188e0218 : SUCCESS in 13m 05s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/1a9afc5a5981473c95961b55359da0a8 : SUCCESS in 10m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ff67d1727a4947f2b0d81d5690c44d28 : SUCCESS in 8m 05s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/e19343cc01224896b7977c1467c1f169 : SUCCESS in 7m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/76ab25a0df05448b8ec4f56383d92e1b : SUCCESS in 13m 01s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/741279d8062b422cabb6a9ea75b7187a : SUCCESS in 12m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b87902e5e045424cbb2f74ba756fad46 : RETRY_LIMIT in 6m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ed20e5bb4ee04e75a2f7193e31d16076 : SUCCESS in 25m 26s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/455008c21b264d6b9da1d3cebc6b0139 : SUCCESS in 37m 56s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9deba036cdbe43bea54a15e205cd8073 : SUCCESS in 23m 16s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/5961b0b00ab9426e9a94738942c26911 : SUCCESS in 28m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3aedcf1b15b141feade9181b5f3e113e : SUCCESS in 16m 48s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/e30c9bddcca1436d9881c07205ec9528 : SUCCESS in 14m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d07fbcdb98074f108327a5893b452558 : SUCCESS in 17m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7a08fbcc22cf4325afdcbc1fbab36791 : SUCCESS in 12m 31s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/cb17f44b297b451fbc4cb8666caa8a2c : SUCCESS in 10m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba9d0cf99edd413387b528b4efcd848c : FAILURE in 26m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6dc0112f02df45ab8642a242907d58c3 : SUCCESS in 1h 21m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/713b07d904df4260b4921aa294b703bf : FAILURE in 19m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b714997ee6210d7ca9a681926b8770ea0c0209e3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-22 16:00:34.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f36e5f7a50caa6bce31dc5516cb16080e444cbf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 20:17:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0874fcd538c74453b0a46f549e2fe7d7 : SUCCESS in 8m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/03c5840b69a34f8f988afc18d8dfe776 : SUCCESS in 7m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/acac842580f744f8b3a9060bcfe87efc : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0305196a62d144d3b489ebb589d64e75 : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07e68058699f44309469453122efd724 : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f589689902347d1bf82f5d6a77bee66 : SUCCESS in 4m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2f532647fa69499988c331700d88b6d9 : SUCCESS in 11m 58s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/30f851016c2a40928806aebab66fa0b4 : SUCCESS in 11m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9df2a4b2d52044aca2e6b85853a6149f : SUCCESS in 15m 53s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7e04c535fb014d218234119004e3d9eb : SUCCESS in 11m 36s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/25378c919d7d4398ab08674d0dc7fcd9 : SUCCESS in 13m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/2f382030354a4c02ba4a925f5e5adf7d : SUCCESS in 9m 34s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/b2c3af3c14974e568e406fbe9e5b88e9 : SUCCESS in 9m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/eb29baad6184439aaec0a73065b32dfd : SUCCESS in 11m 36s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/bb1e46034729413faedd6995c31416be : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2c15b230fdae41a58c9817e8fa736d57 : RETRY_LIMIT in 5m 35s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/72f9da2eb637427cb17a0b7073ba43bd : SUCCESS in 26m 46s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/4e0905c2524442c2be157156d1e00d3a : SUCCESS in 38m 47s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/30cca7cc91714dcf8d19596ba38c8962 : SUCCESS in 31m 16s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/870d668f81174962a703dd34b99710c5 : SUCCESS in 34m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5318845bb4634cac8e5ab41074d9f693 : SUCCESS in 25m 23s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/df79aad363a44aadb7dc3d3818e5323b : SUCCESS in 25m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/06d8ebc7f83c4dc9b95d9f2af4a17fb3 : SUCCESS in 23m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bebccbb364b1442cbb298f7a6ac06572 : SUCCESS in 9m 24s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/3efe2a05a49b49cba149b1bb252bb426 : SUCCESS in 9m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0d365d05c844c619f36ab8f41e2225c : FAILURE in 22m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/373a4266248842fbb4145eeb54afc279 : SUCCESS in 1h 30m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ae974f6215bb4d828584acc5835ec7c8 : FAILURE in 22m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1905cf1383e80443a20a5ae738fa809236217285","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 15:12:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4ebd30cd5cc797a9f1303d9671958dbed842942b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 16:00:44.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ee47f2a410da0ed9b6517f08f276ef971e5a8ccc","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-24 20:02:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2a092469976b0a769224420ceb2e0ec7de99523a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-24 20:02:37.000000000","message":"Patch Set 2:\n\nRebased to pick up https://review.opendev.org/#/c/737405/","accounts_in_message":[],"_revision_number":2},{"id":"409d2254fbc145ddc90bbf15b3f1d622cb5c0062","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 20:07:53.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2a24a53a7f73ecec2fb989dbd834883c6acf0b39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 23:36:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c94d35f3a76b4330889f8807132beb33 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2766d0b78da6413699490921beb5f371 : SUCCESS in 6m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1b254573a3b84ab3aa51b0815509b37e : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5be41be7e2044d138848771dd64034d8 : SUCCESS in 9m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9517b2990c284a619e70adf04566b7d0 : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/823b8e9b6449410589ca81f7ff7cfc31 : SUCCESS in 4m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/905bae9dff5f4b0db45eed29838859c8 : SUCCESS in 13m 26s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/417b99f89f3a4fd295ac3bf6597e062f : SUCCESS in 10m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5116c992745a43149c624e5d48d2976e : SUCCESS in 12m 03s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6c79e31a7b854207a8b0335f6db0cb23 : SUCCESS in 10m 51s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/07d6fa0f94ab4b3399e480c37644b660 : SUCCESS in 15m 10s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/307387fec4c54c5db20927568ec015c0 : SUCCESS in 5m 35s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/379bfeb50f224eaabd84a7bf7c6059be : SUCCESS in 6m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4ca07704dcd54f9facdd1b52415ffaae : SUCCESS in 8m 09s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/913fed5b937e49669f58fc54fb211a61 : SUCCESS in 12m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a7c354d23d564075a0443c747dc84629 : RETRY_LIMIT in 4m 44s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/bb2f0b2c82504089a86b88fc066ee0e6 : SUCCESS in 27m 00s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/9073826191fa4941a32991f893cee32b : SUCCESS in 32m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/29ebd2438b2b402fa44c66a2376af284 : SUCCESS in 21m 57s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/5fb30db8fb724091b627e22b19df3ef8 : SUCCESS in 27m 00s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/bf53bca0947e46ebb96d221f3a8bfd40 : SUCCESS in 12m 58s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/e82f54870574414693107ac168acbbe0 : SUCCESS in 16m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de55af08d42c4b9f963d554695a2e766 : SUCCESS in 19m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d96731bf07084ec383cc304555fd23b9 : SUCCESS in 8m 12s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/8ac7bbaab0744132bcbee2166e898055 : SUCCESS in 15m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/69fd0678387542e2a80fdd794b5da154 : FAILURE in 20m 51s (non-voting)\n- tempest-full https://zuul.opendev.org/t/openstack/build/ada788939db947e2a96858e5e56f93af : SUCCESS in 1h 43m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3bd4c640c4f94186bab9391210cc4b16 : FAILURE in 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"faa91b313d3e7a8859c38f95cbdc45ead4955429","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-25 03:03:37.000000000","message":"Patch Set 2:\n\ncross-horizon-npm is failing with:\nFailed at the horizon@0.0.0 postinstall script \u0027if [ ! -d .tox/npm ]; then tox -e npm --notest; fi\u0027.\n(says it can\u0027t find tox)","accounts_in_message":[],"_revision_number":2},{"id":"7ec08a262f67e48a928704e338205e604c1ab28d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-25 11:05:22.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2b35cd9cd13efdcc6838322aa045f36b96ef0eba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 15:08:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d38d4570d40a42f48f21509f9e85707b : SUCCESS in 7m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3556774bb9ce409e9f34a6cfd9c71a0b : SUCCESS in 4m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/536ee72c71834840a1f676c500d1d03f : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/907ef06a04644230ad442a88f3a0e882 : SUCCESS in 6m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e63b9c020b44ba682ec9c33de53e7a2 : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae755499dbe448528609f4d2c6289090 : SUCCESS in 5m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/02ca1c41af1049eab1088e9c6c887026 : SUCCESS in 9m 27s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/0fdce7df254042b9a0cd5c4cc6468af1 : SUCCESS in 11m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dcef33d295cf4eb6bf2f5797740f32e3 : SUCCESS in 9m 06s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/eb6f5fe1a30a4693af33f8095d7a4eb8 : SUCCESS in 11m 17s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/295318f533904c24b141f7c99b3b46e9 : SUCCESS in 13m 09s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/12f04772162b4f1fb760576824872cf1 : SUCCESS in 5m 29s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/e139b88b74dd4670b00b0301c909a001 : SUCCESS in 7m 16s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9fac9bacd6ea457d8542df850487f347 : SUCCESS in 8m 26s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/6991a7db78a24c65b52d315efea3161f : SUCCESS in 8m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/face0b90bb53437bb2fbe77e6cc2ec33 : SUCCESS in 7m 49s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/73619ab0fccd48338541c6cce6675a5e : SUCCESS in 14m 47s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/f086e33a5264421583d8466d74fa3e62 : SUCCESS in 30m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4b3058851e6443b6b9c22336518b4935 : SUCCESS in 22m 27s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/a0faa9ca422e4530ad3b0c791408a3a2 : SUCCESS in 22m 42s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8e1791f04d7744f6ba7157ddd76ada8f : SUCCESS in 11m 58s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/d746bdda9f6f4767a8b2663571d85e52 : SUCCESS in 13m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/790afed9563d421a9ea06c2b57a9ef4a : SUCCESS in 15m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e785827363bf45fb8fca1f441a7575a6 : SUCCESS in 9m 57s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/fd5606dcef25402db472c2846e67b013 : SUCCESS in 8m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2f2ac6f25004c5eb6c3af232898c940 : FAILURE in 22m 51s (non-voting)\n- tempest-full https://zuul.opendev.org/t/openstack/build/dc1ed5ac30fc4c8e9457934a3a7154b5 : SUCCESS in 1h 37m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5bcfa522f0e24acab8a0a6e1a8b78631 : FAILURE in 20m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b480b9e53dd71e58e5c0b63327a509c13fe456a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 15:14:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b7bc735a4221cf189adbf1c82d87b809524c0c13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 20:47:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e854d9dae74908982bbd78fe5cc42e : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a4a133ab40be4fa0b7881dbd908871d6 : SUCCESS in 4m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c0fec8978328494890f47a72d18afeba : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a301fb52f2d4de8a18c3576c1c7d7cf : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e175172c1f6f4d1ba59d95606c92e989 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/daf1ded859c7431fa59b285afaeefca1 : SUCCESS in 4m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/77e28820da4e47db9aacd6a177ffff2c : SUCCESS in 14m 48s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/1ed3a2149b7241b39323a93d1fdbe6b7 : SUCCESS in 8m 57s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c6724a236eb9435791dbcc1d8551f94d : SUCCESS in 9m 18s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/c30ef3184e784668a3a0d1d22e300fe7 : SUCCESS in 10m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/77dd5a6c44d6450c80d6eada0177c61d : SUCCESS in 5m 46s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/c1d8f334a8394bd3bce74cb610aa397d : SUCCESS in 5m 56s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e7353b92b9a9402c80bd628ac226e6cd : SUCCESS in 12m 36s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1d8df1096fd6446cbd615d5abb383320 : SUCCESS in 12m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65fd0727a55c4fc7984475cc629577be : SUCCESS in 7m 15s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a5c57cdbf5aa47bbbd625950b210a86c : SUCCESS in 24m 39s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/bb20d5542ec847dabf752a2059fbba07 : SUCCESS in 31m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/cec03e589a934a9da42d426cc727aff8 : SUCCESS in 24m 21s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4f3a2d38efcc4209940cfb57b5df958f : SUCCESS in 29m 50s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/53a146437d3d4fd0b3299a0af2938c20 : SUCCESS in 13m 06s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/5baee6699c36480dbbdb7f35138ece28 : SUCCESS in 14m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0872ace3276e4ee4afcbfc25331197a1 : SUCCESS in 19m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e8a2313ee68f4194bd8ac15583a5c6e9 : SUCCESS in 11m 03s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/3b06e486efe8418a8c6f3da5ad5ee3dd : SUCCESS in 9m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c603e36148044a46a2a4845726443217 : SUCCESS in 1h 31m 08s","accounts_in_message":[],"_revision_number":2},{"id":"4f2c260cef9fd763323257982b4d3f9343734d66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 20:47:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"72d1b5f86d40ff072b2e1c72ce1c84666b65daff","revisions":{"dcade7ce01b444e15073df4c383deff9b091ea1f":{"kind":"REWORK","_number":1,"created":"2020-06-18 17:17:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/736784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/736784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/736784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/736784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/736784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/736784/1"}}},"commit":{"parents":[{"commit":"d091e1ac3e1790d7633bd5e4592fd3e2a923ec6c","subject":"update constraint for oslo.log to new release 3.42.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d091e1ac3e1790d7633bd5e4592fd3e2a923ec6c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-18 17:17:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-18 17:17:12.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.8.6","message":"update constraint for os-brick to new release 2.8.6\n\nChange-Id: Ia5200e7807253b8d05b5847c9a7a6ebfa813e67c\nmeta:version: 2.8.6\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: I1b6f0f4be314193c319f185c6977a8c06ccba17c\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Ivan Kolodyazhny \u003ce0ne@e0ne.info\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dcade7ce01b444e15073df4c383deff9b091ea1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dcade7ce01b444e15073df4c383deff9b091ea1f"}]},"branch":"refs/heads/stable/stein"},"72d1b5f86d40ff072b2e1c72ce1c84666b65daff":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-24 20:02:15.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/84/736784/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/736784/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/736784/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/736784/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/736784/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/736784/2"}}},"commit":{"parents":[{"commit":"9764824ef92eaf6637edb5cecdfb4030cdf6e075","subject":"Make greande jobs n-v for EM and oldest stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9764824ef92eaf6637edb5cecdfb4030cdf6e075"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-18 17:17:12.000000000","tz":0},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-24 20:02:15.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.8.6","message":"update constraint for os-brick to new release 2.8.6\n\nChange-Id: Ia5200e7807253b8d05b5847c9a7a6ebfa813e67c\nmeta:version: 2.8.6\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: I1b6f0f4be314193c319f185c6977a8c06ccba17c\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Ivan Kolodyazhny \u003ce0ne@e0ne.info\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72d1b5f86d40ff072b2e1c72ce1c84666b65daff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72d1b5f86d40ff072b2e1c72ce1c84666b65daff"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
