)]}'
{"id":"openstack%2Frequirements~736795","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I3c52abda6351c72f91d0bd3b1db6c572a541786b","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I3c52abda6351c72f91d0bd3b1db6c572a541786b","subject":"update constraint for os-brick to new release 2.10.4","status":"MERGED","created":"2020-06-18 18:22:12.000000000","updated":"2020-06-19 01:00:18.000000000","submitted":"2020-06-19 01:00:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736795-1592528418503-92f6c52c","meta_rev_id":"46663fc23d12ec43d024168478f35bb7489a78d2","_number":736795,"virtual_id_number":736795,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 01:00:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-18 18:29:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-18 20:46:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-18 19:42:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-18 20:46:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 18:29:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-18 19:42:15.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-18 20:46:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-19 01:00:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"22f9a55dcd4933aa258d90110a51b8ad470b8ecd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-18 18:22:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3f7ce992531014c55640be8b0a4a475b3641e30","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-18 18:29:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7c9e5e7b2ae014cee52deb1dcf241ce4c9fde68d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-18 19:42:15.000000000","message":"Patch Set 1: Code-Review+1\n\nThis fixes a regression in 2.10.3, which in turn contained a fix for a security issue (OSSN-0086).","accounts_in_message":[],"_revision_number":1},{"id":"8d433462a55d483aefb8d7d584fa85c98ca4b225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 20:23:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96d982289b7c4087894e5209947ce0de : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30e30aa5f8004ac7818d1fbec88033e7 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eef06685d2cb423d96c5a7acea4e5858 : SUCCESS in 7m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/992727c17a454d61a22eaa0636c046bb : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fc292fa93ba241df8fcad30528f79e29 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8905a147bebc4c4ca269c3542aaeafb6 : SUCCESS in 5m 10s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5479da8b08de4b9398b864b563767f54 : SUCCESS in 9m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aaf14b2ca602460fad0513b92b7fc6d1 : SUCCESS in 10m 28s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6a1325eaa4ae4e879ed422f5e3abf043 : SUCCESS in 12m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f3b34a7f2d734ada8b9b63938ba1dbb3 : SUCCESS in 16m 23s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/18a2bfd1262446539a389ec2e7b14007 : SUCCESS in 9m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/659bc3af2af648d39b74e64c9e77be47 : SUCCESS in 5m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7b137ce1b67f42e69b3a583d8a526b33 : SUCCESS in 14m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5fba3f13e8fe45ec843de08db10e4c6a : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df64401350e648cca7a6cd7671584bba : SUCCESS in 13m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/060320a9db964b868b0036cbce62566d : SUCCESS in 18m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/15fb514e5dd644bf867c6a6d4baaf124 : SUCCESS in 15m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b6ee4512ed8b47bcb907160ad1447e82 : SUCCESS in 25m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8cd1dbb0e117431d9f1babf65d84b375 : SUCCESS in 24m 29s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6ac6068a945b44ce9bd19391ee0d5937 : SUCCESS in 14m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d4e8a27a6bff4f0e81e3291cd3e3b5f5 : SUCCESS in 13m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/34f9ec24b4714ae89560b7d6d2853250 : SUCCESS in 11m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/16d57e54f7414d9381185d0efc02fe27 : SUCCESS in 16m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1cee8882a8f142b79812d50d4f27c855 : SUCCESS in 9m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bd9e03cc091f4f08988df177356f5977 : SUCCESS in 10m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5eb62f8a90dd4bc699aa7cb4f361496a : SUCCESS in 11m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/37aa7e00f10c406b9c0e942d1077196f : SUCCESS in 1h 27m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/03bb0ab696424b1fbf39fac78d08e81b : SUCCESS in 1h 30m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8fc09022002b4e01ae9c60d5683100f6 : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a4fd15ea2dd8f2713880dd05e901c5e1c28b4b01","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-18 20:46:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f04efd137f7233e168310c0d50fa10a11b6995d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 20:47:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1078bf00d525db15ac1636a93acb66ad426f1d20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 01:00:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fca7634693ec4504b749c361b3712925 : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0df47b21b2214393abb88ac23e5e9304 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/daf7c024f4464a10bc812ef59fed5b22 : SUCCESS in 7m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be249bbe6f1f4e5093bb55482c5eae4b : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c73f5a7cbf3c4d319aabc13fcf9b6d52 : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4ec5873a8044dddb7cbaecef305cb96 : SUCCESS in 5m 05s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/84a716fe4412488780cfbaf1888bf5e5 : SUCCESS in 8m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4451f3dcd3d041b99e0d9087e43f474e : SUCCESS in 8m 41s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0eba3dd4af6943b184df2cb57e135466 : SUCCESS in 11m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/378c929334574564b730bfa8b7f6e91e : SUCCESS in 9m 10s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8bfbab3c67d14f86a453159a50f73f4e : SUCCESS in 5m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c55297d219024667aeffb919ff41fe0b : SUCCESS in 5m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d6b1114b5d254f28923b09e6e62245ab : SUCCESS in 11m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cc1b19bbda2e4bc4ac961640fa20a3d7 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7ef20ba904bd418d9b8d8242f50b24cc : SUCCESS in 7m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f496c484b0c0431aa52b3142c59ad841 : SUCCESS in 36m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ef645a15666a4c8c84874e4f21a2acb0 : SUCCESS in 12m 21s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a3f794068fe3474aae1e4e60572afb44 : SUCCESS in 27m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8c770e12bec34dc19c6cc783849a095d : SUCCESS in 24m 21s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/acf0bdcc34b44816bb8017beaebaa9fb : SUCCESS in 22m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/27e3f2b627a546b39cebb08233432ab3 : SUCCESS in 13m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/87f8d6cb90714e8b9e7d7c81412a5672 : SUCCESS in 14m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19d0120903ae4d0a8ae1e003e3139ed4 : SUCCESS in 24m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7b7ef6ebcc7048e5be53f039aebc905a : SUCCESS in 5m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3212a23cf82c42a9acc050aeb16c7b8c : SUCCESS in 19m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d95c7ee920204a7bb0d11ae4e10e5777 : SUCCESS in 9m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4d466d538824b64862f5af440b28160 : SUCCESS in 1h 11m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fffbbeafb81941628db9adf976df8951 : SUCCESS in 1h 43m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/97294d3cc7a0442b9e9cef59e33a9218 : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9edca11e806fe6c070bb34058a3f5d6ecaa3f7fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 01:00:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7af855b650760f0b25096fe629169341f4b76094","revisions":{"7af855b650760f0b25096fe629169341f4b76094":{"kind":"REWORK","_number":1,"created":"2020-06-18 18:22:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/736795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/736795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/736795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/736795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/736795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/736795/1"}}},"commit":{"parents":[{"commit":"1ca83a6cc6daf6ed43fe80d4a80e0b737ddf0fed","subject":"Bump tap-as-a-service to 5.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ca83a6cc6daf6ed43fe80d4a80e0b737ddf0fed"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-18 18:22:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-18 18:22:10.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.10.4","message":"update constraint for os-brick to new release 2.10.4\n\nChange-Id: I3c52abda6351c72f91d0bd3b1db6c572a541786b\nmeta:version: 2.10.4\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: I6c3e06a07950f81bd59d56bf7f27df185fd252cf\nmeta:release:Code-Review+1: Ivan Kolodyazhny \u003ce0ne@e0ne.info\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7af855b650760f0b25096fe629169341f4b76094"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7af855b650760f0b25096fe629169341f4b76094"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
