)]}'
{"id":"openstack%2Frequirements~737030","triplet_id":"openstack%2Frequirements~stable%2Fussuri~Id4969156383545eb94462677e3321b89c1160b48","project":"openstack/requirements","branch":"stable/ussuri","topic":"ensure-pip","hashtags":[],"change_id":"Id4969156383545eb94462677e3321b89c1160b48","subject":"Ensure pip is installed locally","status":"MERGED","created":"2020-06-19 15:21:33.000000000","updated":"2020-06-23 03:40:38.000000000","submitted":"2020-06-23 03:40:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737030-1592883638454-e9b972b9","meta_rev_id":"c9d743a78e0e135ae9347e581fea4370f5f5db25","_number":737030,"virtual_id_number":737030,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-22 18:21:00.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-23 03:40:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-23 01:17:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-19 15:25:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-06-19 15:32:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-06-22 04:37:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-23 01:17:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 15:25:39.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-19 15:32:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-22 04:37:55.000000000","updated_by":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"reviewer":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"state":"REVIEWER"},{"updated":"2020-06-23 01:17:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-23 03:40:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5f94f73dd785813a408de43d7a8b9468ea0dfe4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-19 15:21:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1a33081f6e5a3998611a51043779c1f8f1dbdfee","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-19 15:25:39.000000000","message":"Patch Set 1: Code-Review+1\n\nPlease!","accounts_in_message":[],"_revision_number":1},{"id":"22f6e65dbbc25a570e06d2b8830d3d7563bfe336","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-19 15:30:04.000000000","message":"Topic changed from ensure-pip-stable/ussuri to ensure-pip","accounts_in_message":[],"_revision_number":1},{"id":"fe8497b449966f5d9ba60e610ff739ca228f4c02","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-19 15:32:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2e7bd1a6d94a065629d0569ba83da80239718a08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 17:40:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5186534ca44043608238549195723096 : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b88699f430349a38980363fc897c99a : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aef6ea0707f14644bec2e153c2713d2e : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/034b9067643946c5bfd5252971a86a43 : SUCCESS in 4m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38e2bf4f790c4519bd3a29386761276d : SUCCESS in 56m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4081e2c40cd14fe594e656c708cef374 : FAILURE in 1h 16m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f34efdd783a04b62af61f3f8f6535adb : SUCCESS in 1h 26m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/79a22e5522b6433d9fba1083ba2447d8 : SUCCESS in 1h 17m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7da6f9c6fded6dc577052ad988d75c1ccca678e7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-19 17:45:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ee287ff86a7e0e02e598f7564ee0d4cecd108f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 20:34:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81c428eb66e5445697215b29fd83d247 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5897f493b5de480fa1ee43cf91bd8298 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37c8135505d14e2ea482d3ffe4720465 : SUCCESS in 5m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d4f6d45190bf43eca3329521400b6ac3 : SUCCESS in 7m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7c51f40f49e4ece888471cc3965a0e9 : SUCCESS in 44m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/50257663788d4ac68a3e106c974fbbb4 : FAILURE in 1h 15m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1ad41af25d44442be8dcf773e251e7b : SUCCESS in 1h 31m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/a680ee4a7ac848cd817405a2ecfc9088 : SUCCESS in 1h 16m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e1a589f01cde39adf10f733eb39f0a5d9e80cc13","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2020-06-22 04:37:55.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9af5414444c7143de76f0e6de2a45ff584285fc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 06:08:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b718e4a15dd146d98f49bfc72205cba7 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/586887e66591463cb33faa845a5bc431 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/796af5dec69e442cad8c6fdf620a88f8 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7b31513a4d64652be736495b8062c36 : SUCCESS in 5m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5363e96788ff4daf947557a7ac5e789e : SUCCESS in 45m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5f88352b13647e7b671a420290ed870 : FAILURE in 1h 03m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2c5ad05d4a8433f946cd8fdc5fb6dd6 : SUCCESS in 1h 29m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/4ef72e867c8c4776a9d3d5f35584d143 : SUCCESS in 1h 10m 36s","accounts_in_message":[],"_revision_number":1},{"id":"052a284c9e4aacc8f015759531e9597c2ca820f5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 18:21:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"894598c976749253b903be1cbf3122de2447bded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 20:55:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0ae0b2429ff4ffb88f2aa2e634c2a26 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c9448c8008d4a2d8c4eb5b22b50a3ff : SUCCESS in 8m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8097cfe3102458cb6de01c4a59bd413 : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/532e88a9384f4c6089ab2dea47bd979d : SUCCESS in 4m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39c5c9e641ae458c93948543689e54ec : SUCCESS in 51m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/287f1c0d303c4549b536bc4e56a082d4 : SUCCESS in 1h 24m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1df5ef5b06334208bd469c5a7bb24a26 : SUCCESS in 1h 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/18f59380fa3c42cdb4e651061d7b1638 : SUCCESS in 1h 08m 40s","accounts_in_message":[],"_revision_number":1},{"id":"3e0ec0383eb4efb659eff6a4f76d3c3d4c7f6488","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-23 01:17:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce0d1d6e0c19a06d1d658e3b57c5850db19a2787","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 01:17:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d1bf6688ae141dcd6d19ec53873b94187fbd4b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 03:40:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c9d743a78e0e135ae9347e581fea4370f5f5db25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 03:40:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b8054eb18a426ebe185d02d879dbaa : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ed0a4d3ace649b5af0a580030d501dd : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/303f37f3e0ea41d38f32ecfe3643099b : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3469c1b6b65e4104969af8701a0d4d04 : SUCCESS in 3m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/927078363f3944b4b82c167bb8035edc : SUCCESS in 55m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d5647b7482c4353a75eeb88b3c9613c : SUCCESS in 1h 21m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78e32816f4c1463b8d54cb35a3dc08e1 : SUCCESS in 1h 19m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/dcdef1f5a40745dcb0c97362a3d76d2b : SUCCESS in 1h 16m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d303fe350d74d2d40aa5a591ee9d372a529d10be","revisions":{"d303fe350d74d2d40aa5a591ee9d372a529d10be":{"kind":"REWORK","_number":1,"created":"2020-06-19 15:21:33.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/30/737030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/737030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/737030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/737030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/737030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/737030/1"}}},"commit":{"parents":[{"commit":"d700518ecb589c52460cfab4eec318b824fac5f5","subject":"update constraint for oslo.log to new release 4.1.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d700518ecb589c52460cfab4eec318b824fac5f5"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-17 21:19:06.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-19 15:21:33.000000000","tz":0},"subject":"Ensure pip is installed locally","message":"Ensure pip is installed locally\n\nRecent changes to the base nodepool images removed the system-wide\ninstallation of pip and virtualenv. As a result of this, some\nrequirements-check jobs are failing with an error that pip could not be\nfound.\n\nThis adds the ensure-pip role as part of pre_tasks in our\nrequirements-check playbook.\n\nChange-Id: Id4969156383545eb94462677e3321b89c1160b48\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 74a5bed896893ecd3bdedb100439f773af7c7f5e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d303fe350d74d2d40aa5a591ee9d372a529d10be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d303fe350d74d2d40aa5a591ee9d372a529d10be"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
