)]}'
{"id":"openstack%2Frequirements~737033","triplet_id":"openstack%2Frequirements~stable%2Fstein~Id4969156383545eb94462677e3321b89c1160b48","project":"openstack/requirements","branch":"stable/stein","topic":"ensure-pip","hashtags":[],"change_id":"Id4969156383545eb94462677e3321b89c1160b48","subject":"Ensure pip is installed locally","status":"MERGED","created":"2020-06-19 15:24:59.000000000","updated":"2020-06-25 08:26:03.000000000","submitted":"2020-06-25 08:26:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737033-1593073563960-f0b752bd","meta_rev_id":"f7b0a8b7783f2bf9f538680232d5e5f5f9c15e31","_number":737033,"virtual_id_number":737033,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 08:26:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-24 20:34:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-24 18:39:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-06-24 18:39:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-06-24 18:39:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-24 20:34:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 15:26:48.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-19 15:33:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-24 15:58:43.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-06-24 16:00:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-25 08:26:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73c00d2e0b0fe5461d5c1f524040e6681c1840e7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-19 15:24:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc6cad27c627887d71a1a350d50d998f4cb61b45","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-19 15:26:48.000000000","message":"Patch Set 1: Code-Review+1\n\nI\u0027m hitting this with cinder: https://review.opendev.org/#/c/736833/","accounts_in_message":[],"_revision_number":1},{"id":"654245b4593981bba25a3321d03d903c6a067bb6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-19 15:33:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e8701af0c7c6876d45c94abd6b305faf82a1cd06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 17:33:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d67ae2b1ee964e6aa344f5f1f4138228 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/28d347f1a6c04581853d5063092944cd : SUCCESS in 5m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9a675cfab0ef493386f3a93f187f9210 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d92bddf845614dc796f8336296aae923 : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1627698fa86a4d12b33fb97e430e4416 : SUCCESS in 7m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b028e13fd6df45a5b368541c6bacb7dd : FAILURE in 22m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/859526b107b346b4bcf7111e3168ea51 : SUCCESS in 1h 21m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/70bd2743eb5f4bf6bc46972528357789 : FAILURE in 17m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c50d131e33d0b6bcc0b4c7cdd81b6b79c031d5da","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 23:36:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c1e03d864dec0774c3f8f5e51ca3b884e569f4e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 00:55:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c337854427b42079244f9ec892af406 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54241028cee24714b4fc16545cf73659 : SUCCESS in 6m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7ae94aea77604c26bf6276300d3292af : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79162d7d5eae4bd38accfae4300628e0 : SUCCESS in 6m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2a175d5a2dc478b8bc6c5c7660e08a8 : SUCCESS in 5m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/62dfe3c24eb04e6498983aeb1a1b059f : FAILURE in 19m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/648f37a32e3f4033877f5f61aa5d0d96 : SUCCESS in 1h 16m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/79336da18e0d45f6937debfc15bb18bd : FAILURE in 15m 56s","accounts_in_message":[],"_revision_number":1},{"id":"325f7cd87fb7bfcd6c4997a48114cd4462dec76b","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-23 20:58:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fa2d48765b294e4f91676d0f0ff9282437a4b460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 01:33:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a176d3e6a0b5416ab60640d6ea68bece : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/626d340d60814e45bc6fc90022dad36c : SUCCESS in 4m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4bb197e9189d4181b77f5775224b30ca : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de83d9c5bdbd4c789cf4e2b126975fc8 : SUCCESS in 4m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a129054690784b6089d61e01b1248bb6 : SUCCESS in 4m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0eb040e2e064cc089b78bcd8c5d2aff : FAILURE in 12m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3f89b0527c1f4cf79de4655b1f46162c : SUCCESS in 1h 14m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e6d58b32b0484286b42a9935c661f495 : FAILURE in 17m 22s","accounts_in_message":[],"_revision_number":1},{"id":"4334ac2e3bfd3654e9fefae2abf4986941941f40","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-06-24 15:58:40.000000000","message":"Patch Set 1:\n\nrecheck\n\nhttps://review.opendev.org/#/c/737405/ just merged","accounts_in_message":[],"_revision_number":1},{"id":"52b47fc9129a377c52f723b1b1bdd9a2c289f644","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-06-24 15:58:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8491ce7713c1daf28aa422caabce5411c152c921","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 16:00:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1b5230081bd723b70133d3909d23ac013006d7a5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 18:39:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e24722f76304bbde5eae1e9e4921d2a283a3b48c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 20:34:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0c95784e2e23363f000e6f92e9917ae11b91274f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 23:05:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c52595021554daeb6a906ea390264fd : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2bdd2bfcd9bb4e888bce97cc8f1a34de : SUCCESS in 6m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cff764717cc14bc38b3253dfdff91903 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2956a3d8456e4406aaaf35a08abcf3e4 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de5e0374cc4c4e3886bb714af8f43320 : SUCCESS in 6m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4511cbf3ae744339bb638043be37a192 : FAILURE in 18m 29s (non-voting)\n- tempest-full https://zuul.opendev.org/t/openstack/build/5db7af21177a49d09ced5a56f9222778 : SUCCESS in 1h 25m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e0cbdb66fc4d4498863870948c979fbf : FAILURE in 15m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3ce5d0c05f271b483d3287366a900e842079454d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 23:05:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e543141552fd9306fb6dfc86c5bd6c4bbb2ec58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:26:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/035883a5a79640e98499ebca27f5cd6c : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82dc13347cec496383d06d9f1ef00168 : SUCCESS in 8m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6f7a8ba81b854176a90ff6ebffb2ec0b : SUCCESS in 9m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd6f6c6a184d483c9f685139c1eba31b : SUCCESS in 8m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/438034983a0143babf0c03c4befaef91 : SUCCESS in 8m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/13910b85eed54a6db4f0d79dcbe6e191 : SUCCESS in 1h 22m 17s","accounts_in_message":[],"_revision_number":2},{"id":"6f0c1234033cd6d348279fb9673c1fb3887b5ebb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:26:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b8aee25c8e66dd84bb28dace86a12bcc940ceeca","revisions":{"0f51382318eff21f9d5a02375bd55b3f805fcc62":{"kind":"REWORK","_number":1,"created":"2020-06-19 15:24:59.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/33/737033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/737033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/737033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/737033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/737033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/737033/1"}}},"commit":{"parents":[{"commit":"d091e1ac3e1790d7633bd5e4592fd3e2a923ec6c","subject":"update constraint for oslo.log to new release 3.42.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d091e1ac3e1790d7633bd5e4592fd3e2a923ec6c"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-17 21:19:06.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-19 15:24:51.000000000","tz":-300},"subject":"Ensure pip is installed locally","message":"Ensure pip is installed locally\n\nRecent changes to the base nodepool images removed the system-wide\ninstallation of pip and virtualenv. As a result of this, some\nrequirements-check jobs are failing with an error that pip could not be\nfound.\n\nThis adds the ensure-pip role as part of pre_tasks in our\nrequirements-check playbook.\n\nChange-Id: Id4969156383545eb94462677e3321b89c1160b48\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 74a5bed896893ecd3bdedb100439f773af7c7f5e)\n(cherry picked from commit d303fe350d74d2d40aa5a591ee9d372a529d10be)\n(cherry picked from commit d43e0a585719930516fd82504d84e3867512c7f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f51382318eff21f9d5a02375bd55b3f805fcc62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f51382318eff21f9d5a02375bd55b3f805fcc62"}]},"branch":"refs/heads/stable/stein"},"b8aee25c8e66dd84bb28dace86a12bcc940ceeca":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-24 18:39:06.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/33/737033/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/737033/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/737033/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/737033/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/737033/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/737033/2"}}},"commit":{"parents":[{"commit":"9764824ef92eaf6637edb5cecdfb4030cdf6e075","subject":"Make greande jobs n-v for EM and oldest stable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9764824ef92eaf6637edb5cecdfb4030cdf6e075"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-17 21:19:06.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-24 18:39:06.000000000","tz":0},"subject":"Ensure pip is installed locally","message":"Ensure pip is installed locally\n\nRecent changes to the base nodepool images removed the system-wide\ninstallation of pip and virtualenv. As a result of this, some\nrequirements-check jobs are failing with an error that pip could not be\nfound.\n\nThis adds the ensure-pip role as part of pre_tasks in our\nrequirements-check playbook.\n\nChange-Id: Id4969156383545eb94462677e3321b89c1160b48\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 74a5bed896893ecd3bdedb100439f773af7c7f5e)\n(cherry picked from commit d303fe350d74d2d40aa5a591ee9d372a529d10be)\n(cherry picked from commit d43e0a585719930516fd82504d84e3867512c7f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b8aee25c8e66dd84bb28dace86a12bcc940ceeca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b8aee25c8e66dd84bb28dace86a12bcc940ceeca"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
