)]}'
{"id":"openstack%2Frequirements~737036","triplet_id":"openstack%2Frequirements~stable%2Fqueens~Id4969156383545eb94462677e3321b89c1160b48","project":"openstack/requirements","branch":"stable/queens","topic":"ensure-pip","hashtags":[],"change_id":"Id4969156383545eb94462677e3321b89c1160b48","subject":"Ensure pip is installed locally","status":"MERGED","created":"2020-06-19 15:26:54.000000000","updated":"2020-08-05 22:33:29.000000000","submitted":"2020-08-05 22:33:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737036-1596666809708-8e31e638","meta_rev_id":"cfca00e38de78962b1a1f913b590a52e1e739e6b","_number":737036,"virtual_id_number":737036,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-05 22:33:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-08-05 20:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-08-05 20:57:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 15:34:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-05 20:57:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-05 22:33:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"799098a468963443d7f44d4e90b101a461c62427","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-19 15:26:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a3c121c230c31a69e4e65c582c4afde97469897","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-19 15:34:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ad83a286d68e0f96a0275bcea1ef3a85dc0104b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 16:43:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa52a6ead73a4c8ab97a5c01e1d18da9 : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1af5b21ae0d644c9b54601521aa8ee17 : SUCCESS in 5m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/95fe544e868f4cc2b456e6082aa68f00 : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45a86f0a07434a18af7cd42e76ff4d7e : SUCCESS in 6m 55s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bebcfd7b58c148638684cfcdfd652780 : FAILURE in 17m 55s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/1b076fcb89bf4bc4a323037a50e5b1d9 : FAILURE in 21m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/824bddb4607848c490ccb8a9ddfd7444 : FAILURE in 19m 23s","accounts_in_message":[],"_revision_number":1},{"id":"cdefedd1b4246f0ad7522fc4fdd658d728b34de1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 16:00:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7864117d8584edc87103dd341867e397a5d6ca58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 20:29:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddc635e5eefa422396bdd9c4d8be77cf : SUCCESS in 6m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8e8f8e44e09416aafe728fe6254b10a : SUCCESS in 9m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4fd5e3161df64cf6b738cb4617e24b99 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c3fc6499c894ce2b68e7502b73cd112 : SUCCESS in 12m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a94dbab56a16464faeb45f1ad7b94b7a : FAILURE in 18m 53s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/773d4dc1beef41cc88edc844e22518f6 : FAILURE in 16m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9338aed734a84d98bb9d527ee39cd405 : FAILURE in 16m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9c6c440c5020407a88a49fa544e7ab40226c8a8f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-12 00:13:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c8ccf87893e77ca73da1097fd17b000865e3a7bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 01:30:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f273a082ac1462f908cf1eb5284301f : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b076fc817ba43208af495461ebe093f : SUCCESS in 4m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1381cdd7895d4a0b80af872f08912815 : FAILURE in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2202ae6d2def4a8ba99b7324e27040d7 : SUCCESS in 4m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/89b5f282581e4f95b9d2929461f19ab4 : SUCCESS in 1h 00m 47s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/64999c3ec3c94f809cfb24f702503241 : SUCCESS in 1h 14m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e14ab32d3894450381534ab78c290c53 : SUCCESS in 52m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7dbe7fd154528677cf87f263040ea23c5d3be0d7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-12 02:29:17.000000000","message":"Patch Set 1:\n\nI gentoo got rid of py35 a while ago, gonna be harder to test","accounts_in_message":[],"_revision_number":1},{"id":"3fc96aa7fa7674df7dfe68989570f6c8993a0a90","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-15 21:09:07.000000000","message":"Patch Set 1:\n\nrecheck\n\nError is something I think is related to setuptools that I could have sworn was fixed. At least, the error looks really familiar and I\u0027m pretty sure it was fixed elsewhere.","accounts_in_message":[],"_revision_number":1},{"id":"dd56a7e914272fb81710e8ff87174dcfc69d2858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 23:00:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85a04b16f9a24dbca4bde9810fe3fb41 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d227d1cd3b84f2bbc43058c80285b27 : SUCCESS in 7m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9d3e72afcbe344e7bac56acad44bb2b9 : FAILURE in 6m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fcf90c0abf0746fa9c3e133ddb3bc06f : SUCCESS in 6m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/80a290928a964069af2f8990c89368fe : SUCCESS in 1h 14m 15s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/8e48078f7f474ea396061473fcc9fcca : SUCCESS in 1h 14m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c2e883a1dc4344c8b768df87a02fc5bd : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":1},{"id":"15e39e7479435d41f00038d0a71acd8021d8128b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-17 10:21:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"81353c633986b9f2b78cb54d3d21cde53ca400b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 12:26:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e29436c96e46889cf85b2fe6fde07c : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5a80c43bf5a42448206cc45e92c7376 : SUCCESS in 5m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ecf69b3b3dda4b2f8f8ea372d2b1d253 : FAILURE in 5m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/51c4a864534849719269c63116a3a903 : SUCCESS in 5m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea09f987e19c439e83c3987b3f2ef08e : SUCCESS in 1h 04m 42s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/7e7ab934726345f393c0e0a4d7619751 : SUCCESS in 1h 38m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b21008ead9584168a58db261dafdfbde : SUCCESS in 1h 03m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c6f5ed01f628cd89ff06f061dd459b76d8ede5eb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-29 11:50:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7f30b3751d18e3ca3d66c420cb3c79c8e938472c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 15:28:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c22cb8f3517465e879c146126f3b377 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf0edd2314964885bc3e5fbfb623656e : SUCCESS in 4m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/abbb934a2fe54d49aa1e7207c87e0929 : FAILURE in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a2fd74f37d4421394c3c3d51b1395e2 : SUCCESS in 4m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e6615f53904845fea95ea24ac1eb7bdd : SUCCESS in 58m 48s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/497c5a77546b4ee0b3d1c1f84dcc2e71 : SUCCESS in 1h 35m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3a9dda984f874f08b85420ed6216f57f : SUCCESS in 52m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8c88baf78d21a0439060871c1fce98f9147754d6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-05 18:43:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ac7317c1ae467d8ae123a20686c673c91b53a5c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 20:56:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8b9d9bd32934254adbc252397895505 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bd6e5c79e2f48a3956c9694f9a156e8 : SUCCESS in 5m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4aaaf4392c114eac9ced93d67dd3c844 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5674b1f7d86041e091cd5d1646078108 : SUCCESS in 3m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5eb5765986e64a3cb822e22a1eb1bc9e : SUCCESS in 1h 07m 01s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/11da20230fff4c5a80003b4058f5f3ef : SUCCESS in 1h 39m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b937b080262c41ccafa7a8ea3dfbdd01 : SUCCESS in 1h 16m 01s","accounts_in_message":[],"_revision_number":2},{"id":"c31ba13a3b3b25eb4140f7f0a58613a9f8ffdc0b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-05 20:57:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b4ea5f37a3021cea79f8bd6480baea26d06ddc82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 20:58:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b4a7bff9a9cf38544673cadb18731e9b7a30cc41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 22:33:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"38e3f7f648fd09a54065aca2d4e7222adabe6fda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 22:33:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2be6e6c6cb54e86b5a003de51cc8d19 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e15657a40969422c8136070344d73745 : SUCCESS in 5m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/83ea6a1a3e1b4e24a88eedbb07104c79 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac107eed4c15434bbae7fe1a4b9bb8e3 : SUCCESS in 4m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a7a0d1a75b54b92a948dfc1b32b3a9a : SUCCESS in 1h 32m 07s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/d066911e799444708dad694d46f64991 : SUCCESS in 1h 14m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/99114addb46a4ebf8a4cc3a66627ace0 : SUCCESS in 46m 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bc7833bb37d89a0973cef88621ba7ade3be574a1","revisions":{"e24a412836f4eff58821ceba517916dc13397b3e":{"kind":"REWORK","_number":1,"created":"2020-06-19 15:26:54.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/737036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/737036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/737036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/737036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/737036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/737036/1"}}},"commit":{"parents":[{"commit":"035ca011130f04e6ca6d53eabd473fcb0d9a1aa8","subject":"Pin diskimage-builder \u003c 3.0.0 on python2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/035ca011130f04e6ca6d53eabd473fcb0d9a1aa8"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-17 21:19:06.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-19 15:26:49.000000000","tz":-300},"subject":"Ensure pip is installed locally","message":"Ensure pip is installed locally\n\nRecent changes to the base nodepool images removed the system-wide\ninstallation of pip and virtualenv. As a result of this, some\nrequirements-check jobs are failing with an error that pip could not be\nfound.\n\nThis adds the ensure-pip role as part of pre_tasks in our\nrequirements-check playbook.\n\nChange-Id: Id4969156383545eb94462677e3321b89c1160b48\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 74a5bed896893ecd3bdedb100439f773af7c7f5e)\n(cherry picked from commit d303fe350d74d2d40aa5a591ee9d372a529d10be)\n(cherry picked from commit d43e0a585719930516fd82504d84e3867512c7f6)\n(cherry picked from commit 0f51382318eff21f9d5a02375bd55b3f805fcc62)\n(cherry picked from commit aa3018d58e739951b1cc6697e2a207a73739976f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e24a412836f4eff58821ceba517916dc13397b3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e24a412836f4eff58821ceba517916dc13397b3e"}]},"branch":"refs/heads/stable/queens"},"bc7833bb37d89a0973cef88621ba7ade3be574a1":{"kind":"REWORK","_number":2,"created":"2020-08-05 18:43:18.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/737036/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/737036/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/737036/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/737036/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/737036/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/737036/2"}}},"commit":{"parents":[{"commit":"035ca011130f04e6ca6d53eabd473fcb0d9a1aa8","subject":"Pin diskimage-builder \u003c 3.0.0 on python2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/035ca011130f04e6ca6d53eabd473fcb0d9a1aa8"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-17 21:19:06.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-05 18:42:15.000000000","tz":-300},"subject":"Ensure pip is installed locally","message":"Ensure pip is installed locally\n\nRecent changes to the base nodepool images removed the system-wide\ninstallation of pip and virtualenv. As a result of this, some\nrequirements-check jobs are failing with an error that pip could not be\nfound.\n\nThis adds the ensure-pip role as part of pre_tasks in our\nrequirements-check playbook.\n\nAlso necessary to squash:\n\nFix failure with setuptools \u003e\u003d 49.0.0\n(cherry-picked from commit d30e19237f8eb8be9e1a736f1d4503c8cfcd74b2)\n\nChange-Id: Id4969156383545eb94462677e3321b89c1160b48\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 74a5bed896893ecd3bdedb100439f773af7c7f5e)\n(cherry picked from commit d303fe350d74d2d40aa5a591ee9d372a529d10be)\n(cherry picked from commit d43e0a585719930516fd82504d84e3867512c7f6)\n(cherry picked from commit 0f51382318eff21f9d5a02375bd55b3f805fcc62)\n(cherry picked from commit aa3018d58e739951b1cc6697e2a207a73739976f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc7833bb37d89a0973cef88621ba7ade3be574a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc7833bb37d89a0973cef88621ba7ade3be574a1"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
