)]}'
{"id":"openstack%2Frequirements~737307","triplet_id":"openstack%2Frequirements~master~I79350c8f66c0229768af64631542b5c1b5d48af9","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I79350c8f66c0229768af64631542b5c1b5d48af9","subject":"Constrain kombu in global requirements","status":"ABANDONED","created":"2020-06-22 16:03:56.000000000","updated":"2020-07-01 13:55:06.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b7a3791d159dbe89ab5646ee3a4f1662ef93f9eb","_number":737307,"virtual_id_number":737307,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"date":"2020-07-01 13:54:46.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-24 00:38:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2020-06-24 17:13:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-23 19:49:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 11:14:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-23 19:49:17.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-06-24 00:38:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-24 17:13:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4d84101ea4378e12f616dee8c1f591ef83a5f78c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-22 16:03:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e2c9ab997bed9e1da5042ca6c767ca91117be76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 20:15:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ca2253bb7da49f1b6f91adde5c16b1f : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/450f9a6cad934960b61af4898ac8d312 : SUCCESS in 10m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f557a2954204bd7ad17e5373ba01494 : SUCCESS in 9m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6e2edff3be34c158d438834583ba296 : SUCCESS in 11m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07e30b7c76f74245aeae57b42e29f639 : FAILURE in 5m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f07a5edfa3a446f5a1cec51617059692 : SUCCESS in 54m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ec227343f8d4bc386818c64603ff918 : SUCCESS in 1h 22m 50s","accounts_in_message":[],"_revision_number":1},{"id":"945b6b5290946fe040ce616c362c7255d778a13b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-23 08:22:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8f1d9374569388aaed34ae66e781d421b15d950","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 08:37:22.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a60bc540abd14c23b47ea72985d7dc8b : FAILURE in 11m 09s","accounts_in_message":[],"_revision_number":2},{"id":"41de6984be742b1c0c3996c85a02c6f6b10298fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 09:56:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dfa28ed1b6943d89411308f002477e2 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1fa0341435545e791252d9fd94922fc : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edfaccd555484841954e25cfa17af1da : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25be5e1592094fee84f944f5fe25eab1 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9221b6fe0a8a4190b5e891bc3b9382b2 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b021790952a4ef08329e91874f79151 : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2b904ab8cb414892bbfe411e479ea82c : FAILURE in 10m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/40bc613207ec43fca9149d492403ea3f : FAILURE in 11m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f99fe28431cb451497e7cfeba5ae3b1f : SUCCESS in 10m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3411c221a2c343828afaab8b7edf686f : SUCCESS in 5m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/08abede13ab449a789279d5d6ac0ebe6 : SUCCESS in 12m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f722808981a14caba2fe6cbe717b056b : SUCCESS in 9m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/50436f26502d4fcd94557a0a45979adc : SUCCESS in 13m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1a27ee8ffff9445890046dfdbee5caf3 : SUCCESS in 25m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ae1b578417fa43d4a7c48106c35559e8 : SUCCESS in 14m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a76cf8dfeca54b7b845981a5b5697a30 : SUCCESS in 20m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0a1d118653404756b4bb451a0585b611 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d109d7457bb242d8a66050bbdf52ec47 : SUCCESS in 7m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ec13eb4c45b548b9a6bd6239747abfb5 : SUCCESS in 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cf278ca85d3473bab5b6562836e036b : SUCCESS in 52m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69f11b7203b34b70891ef71a688aee08 : SUCCESS in 1h 28m 53s","accounts_in_message":[],"_revision_number":2},{"id":"be5662036198dd50fc24b2cafe27a7e094436582","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-23 10:05:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b1191e815a8f76b3390ab3e5511e9c3468dfbef2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-23 10:06:37.000000000","message":"Patch Set 2: Code-Review+2\n\nWaiting for CI recheck","accounts_in_message":[],"_revision_number":2},{"id":"d326c29527c41a88ec8b48eb3418dbcd5f237a0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 10:28:18.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3fcdde38314446beaf3d888cf4ef411c : FAILURE in 16m 31s","accounts_in_message":[],"_revision_number":2},{"id":"6abba5751b44e9770193bf220a0fd236de453fcf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-23 11:14:45.000000000","message":"Patch Set 2: Code-Review+1\n\nHmmm... CI complains about amqp and kombu versions","accounts_in_message":[],"_revision_number":2},{"id":"1223683f49359121f861ca17e9fa63c3fa7c35e2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-23 11:21:05.000000000","message":"Patch Set 2:\n\n\u003e Hmmm... CI complains about amqp and kombu versions\n\nI think it\u0027s related to kombu\u0027s requirements [1] and our upper-constraints which is equal to `amqp\u003d\u003d\u003d2.6.0` but I\u0027m not 100 percent sure.\n\nI guess we should adapt py-amqp\u0027s version too.\n\n[1] https://github.com/celery/kombu/blob/619754f130d2fc138fa6e228ac43aad52c0a1fa2/requirements/default.txt#L2","accounts_in_message":[],"_revision_number":2},{"id":"be172a0bf9922d431b20edfe9075731c796fab87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 13:09:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3346eace94814a8aab4758345f782ce6 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d3ce57f1ab34e11b2d8ac5c0ab2a9fe : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88076469393646088d906400969d8b3e : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e447d36b3a2841e4a3cef7d3b4c45bd2 : SUCCESS in 6m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a3d54935f9c4177a5ff867067b47559 : SUCCESS in 8m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8915c059e34b4bd691af3d96138d1c62 : SUCCESS in 6m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/368579747ee04c2d8bd9cbbcce5225de : FAILURE in 9m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d42824f5eed484587a7929713c5b15d : FAILURE in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fc1bbfeb9cfa48a8a50ea28b504f2030 : SUCCESS in 11m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5125c12367114f3892a348d83dbb4f2f : SUCCESS in 7m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a8bb81f7a08e4a9c8592e0efe36406c7 : SUCCESS in 12m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc031609acb04e3d97361dfde6bb5ba6 : SUCCESS in 9m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/76f60fba977d47debcf47830895f000f : SUCCESS in 16m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/64a01d82518045dd978195b921ef1cd9 : SUCCESS in 32m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/89d82d2ba6b14d2086f637b3e135b64f : SUCCESS in 19m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/00ac740dc67d419f9a05ea886fa65fd5 : SUCCESS in 22m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/00682d13ecb74a57a27129d39c46b5f5 : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7defbbc651c646fd82234e1bf21d36e1 : SUCCESS in 8m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/155c63ddd39842169b504519f2200ee3 : SUCCESS in 14m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0b7b4befc494e38ad3f5b7b41648517 : SUCCESS in 56m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6946720c02047808a7efc6eb094e9e1 : POST_FAILURE in 1h 25m 20s","accounts_in_message":[],"_revision_number":2},{"id":"018c85ea55e16b454ab556aa86cef47d1451a3a7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-23 13:32:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"13e37df02353e0db77a9a894ea7937d64f39ef36","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-23 13:34:18.000000000","message":"Patch Set 3:\n\n\u003e \u003e Hmmm... CI complains about amqp and kombu versions\n \u003e \n \u003e I think it\u0027s related to kombu\u0027s requirements [1] and our\n \u003e upper-constraints which is equal to `amqp\u003d\u003d\u003d2.6.0` but I\u0027m not 100\n \u003e percent sure.\n \u003e \n \u003e I guess we should adapt py-amqp\u0027s version too.\n \u003e \n \u003e [1] https://github.com/celery/kombu/blob/619754f130d2fc138fa6e228ac43aad52c0a1fa2/requirements/default.txt#L2\n\nYes, kombu 4.6.8 requires amqp\u003c2.6 :(\nI\u0027ve updated the patch with that constraint.","accounts_in_message":[],"_revision_number":3},{"id":"b615434cd5784806077f7b8d63682ca2d3f56ef3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 14:09:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/189f782d5f3a466e8fd2f98e031c8a4c : SUCCESS in 15m 07s","accounts_in_message":[],"_revision_number":3},{"id":"b8b694849661ff0fc2f79858ca19663a5c713721","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-06-23 17:29:07.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"002198d4d1f6d9ef5118b5ed6c1889161030544b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 18:16:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40c21c90fb7b4768b215735d7b87deb7 : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec9cf75b78ea4cf094785f3af3f7b5ab : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40633cea873346bfa9c1143f5205c30e : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d41332641f84266a1e118f40f82a8a7 : SUCCESS in 7m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9095009354d044e19f10b153c698ec97 : SUCCESS in 6m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e755ecff0ad64f2b86ec2d810735715b : SUCCESS in 3m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d0cf62f5eb7943a8a18152085c21d85f : SUCCESS in 8m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/30947558dcfb40cdb440872d8e6ee545 : SUCCESS in 11m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c20000313e05481fa5010e511f8ee81c : SUCCESS in 11m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3805861e00a14fec977f18c4aeb2575e : SUCCESS in 7m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/289e96e4fbf14e90b944d2f363a21b09 : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/518f8847223b41db9b18c9730725abcc : SUCCESS in 9m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ae20180645e54a0990726844bdd9f889 : SUCCESS in 19m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cc4fc3e828f447b485f0a1a3cdccb549 : SUCCESS in 25m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b173380cf927463aacb13580d90e08e8 : SUCCESS in 17m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20236e6a49f64318bba9523fb51b6204 : SUCCESS in 17m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0ecbf64f746f44378ccd674431d4f370 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8bffec755db04c4995e2b4b3b4ffa7d1 : SUCCESS in 7m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/788a6b41472448aa87bf05f6dbc081c9 : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79dd2c94dc584f248331397ce94243ab : SUCCESS in 58m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c07cb00aa76a43ac91857965fe7b4d5c : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":3},{"id":"5512f2244227ca879076f376ee50263898fe7e70","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-23 18:56:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"527f988873655721423608dcb149acccaf7999bd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-23 18:58:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dfc7fad2a1eada6bfd36744fbccc64e549cdbdd7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-23 18:58:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"43fb6289979ff2400f1ff2f7fcfb98688dbe5d49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 19:16:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/27a51ee583a840ad89726fa541b82e91 : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":5},{"id":"9e7f577fee943e083f3184ecf64e0577928c0819","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-06-23 19:49:17.000000000","message":"Patch Set 5: Code-Review+1\n\nThanks!","accounts_in_message":[],"_revision_number":5},{"id":"87e519edcbccc3f1f9eda20c13dbe8dad9720218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 00:38:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7443870f12947e4aa6e82d033005103 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1534469371bc4df8ad782b16551e95a9 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b78da0bdb4e4f88b14e5ff86a40d6c0 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dabe656476b494aa0299a1a7803de50 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/493806b8fd6c484fbd4e010b473c8163 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a212d00b4b3442bdbbedf4a5f5006614 : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bdced8cfff954e04b0ba8a15462cc5af : SUCCESS in 7m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/61ba26d149674687b0a5a0653051e309 : SUCCESS in 10m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/581f552f26f240b1a43d326c0fecc83e : SUCCESS in 10m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/96747b6831f241d599d440ecc7615611 : SUCCESS in 5m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9979954ef8d9447d9abbb23b16448954 : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9abaa23eb1d94cd58dac59c1d8bf8da4 : SUCCESS in 10m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4c68c1e6529d4e4fa414ee4c2c06f154 : SUCCESS in 12m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/419d66f9d06a4fb6b0ff92cbefd1f6b9 : SUCCESS in 27m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6aa5055797044fcf91c049fa75c1b729 : SUCCESS in 12m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fb4bc290ebfe4a0db773684972e98d08 : SUCCESS in 20m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/91dc194d72154e59a9a57d24f36521bf : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/122ee8a0f64b43f98a2d190f37845631 : SUCCESS in 6m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6a1f859ea7ac4ef195ff231e58353e31 : SUCCESS in 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b825bfb64f734e2b9018b6f8c6a3364a : SUCCESS in 53m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0f6fe84402441a5a34677d9446b517d : SUCCESS in 1h 12m 55s","accounts_in_message":[],"_revision_number":5},{"id":"c66ed9dbac662e9aa5a8c5dcc3cc0ca0807b1c45","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 17:13:03.000000000","message":"Patch Set 5: Code-Review-1\n\nis this an acknowledged bug upstream? will 4.6.10 fix it?  If so, perhaps excluding the one bad version would be better","accounts_in_message":[],"_revision_number":5},{"id":"8be3418c02d47ce801d7ac8fb14cf52b5e1c9eea","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-06-24 17:30:14.000000000","message":"Patch Set 5:\n\n\u003e is this an acknowledged bug upstream? will 4.6.10 fix it?  If so,\n \u003e perhaps excluding the one bad version would be better\n\nWell, I did not identify any upstream bug in kombu, but oslo.messaging unit tests times out with 4.6.9, 4.6.10, and even with the most recent 4.6.11 (released 10 hours ago).","accounts_in_message":[],"_revision_number":5},{"id":"4040045a8849c60ab74239b6537342cc7c529ef4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 18:46:32.000000000","message":"Patch Set 5:\n\nIt sounds like this is something that oslo.messaging will have to adapt to.  Is that planned?","accounts_in_message":[],"_revision_number":5},{"id":"3bcf5fa407131908eb49c12b1fdac8b57647cff0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-01 13:54:46.000000000","message":"Patch Set 5:\n\n\u003e It sounds like this is something that oslo.messaging will have to\n \u003e adapt to.  Is that planned?\n\nThanks Matthew, you are right, and it\u0027s confirmed from oslo team, too, that this should be fixed inside oslo.messaging.","accounts_in_message":[],"_revision_number":5},{"id":"b7a3791d159dbe89ab5646ee3a4f1662ef93f9eb","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-01 13:55:06.000000000","message":"Abandoned\n\nAs this should be fixed in oslo.messaging","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"92ac507c1209ac80aea81cf175900bc73f6c3ec9","revisions":{"223c766fe927cf1c4d0397c6c54e95ef92002770":{"kind":"REWORK","_number":1,"created":"2020-06-22 16:03:56.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/07/737307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/737307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/737307/1"}}},"commit":{"parents":[{"commit":"d9bc8c5e8bf44b1024ada46f6cbc1f5132422c09","subject":"Merge \"Remove mox3 from projects.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9bc8c5e8bf44b1024ada46f6cbc1f5132422c09"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-22 15:56:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-22 15:56:25.000000000","tz":120},"subject":"Constrain kombu in global requirements","message":"Constrain kombu in global requirements\n\noslo.messaging unit tests started to time out when kombu version 4.6.9\nwas released, so let\u0027s exclude kombu 4.6.9 and newer versions until the\nissue is fixed.\n\nThe patch that causing the unit tests to fail is:\n\nhttps://github.com/celery/kombu/pull/1193/commits/398aa5b8cd1fe1fc189ab35a2cb0bececf51bf37\n\nChange-Id: I79350c8f66c0229768af64631542b5c1b5d48af9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/223c766fe927cf1c4d0397c6c54e95ef92002770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/223c766fe927cf1c4d0397c6c54e95ef92002770"}]},"branch":"refs/heads/master"},"a9925865d1a5fac5d3154569bad1d29de1f32cc5":{"kind":"REWORK","_number":2,"created":"2020-06-23 08:22:20.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/07/737307/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/737307/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/737307/2"}}},"commit":{"parents":[{"commit":"d9bc8c5e8bf44b1024ada46f6cbc1f5132422c09","subject":"Merge \"Remove mox3 from projects.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9bc8c5e8bf44b1024ada46f6cbc1f5132422c09"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-22 15:56:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-23 08:22:02.000000000","tz":120},"subject":"Constrain kombu in global requirements","message":"Constrain kombu in global requirements\n\noslo.messaging unit tests started to time out when kombu version 4.6.9\nwas released, so let\u0027s exclude kombu 4.6.9 and newer versions until the\nissue is fixed.\n\nThe patch that causing the unit tests to fail is:\n\nhttps://github.com/celery/kombu/pull/1193/commits/398aa5b8cd1fe1fc189ab35a2cb0bececf51bf37\n\nChange-Id: I79350c8f66c0229768af64631542b5c1b5d48af9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9925865d1a5fac5d3154569bad1d29de1f32cc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9925865d1a5fac5d3154569bad1d29de1f32cc5"}]},"branch":"refs/heads/master"},"6d6ee0529218dca0cb817a6a22e4d3e615d384d4":{"kind":"REWORK","_number":3,"created":"2020-06-23 13:32:29.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/07/737307/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/737307/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/737307/3"}}},"commit":{"parents":[{"commit":"d9bc8c5e8bf44b1024ada46f6cbc1f5132422c09","subject":"Merge \"Remove mox3 from projects.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9bc8c5e8bf44b1024ada46f6cbc1f5132422c09"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-22 15:56:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-23 13:30:11.000000000","tz":120},"subject":"Constrain kombu in global requirements","message":"Constrain kombu in global requirements\n\noslo.messaging unit tests started to time out when kombu version 4.6.9\nwas released, so let\u0027s exclude kombu 4.6.9 and newer versions until the\nissue is fixed.\n\nThe patch that causing the unit tests to fail is:\n\nhttps://github.com/celery/kombu/pull/1193/commits/398aa5b8cd1fe1fc189ab35a2cb0bececf51bf37\n\nNOTE(elod.illes): amqp is constrained due to kombu 4.6.8 constrain\namqp to less than 2.6.0\n\nChange-Id: I79350c8f66c0229768af64631542b5c1b5d48af9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d6ee0529218dca0cb817a6a22e4d3e615d384d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d6ee0529218dca0cb817a6a22e4d3e615d384d4"}]},"branch":"refs/heads/master"},"acf99baeb52d60726644e626509a11a0e423bf19":{"kind":"REWORK","_number":4,"created":"2020-06-23 18:56:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/07/737307/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/737307/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/737307/4"}}},"commit":{"parents":[{"commit":"d9bc8c5e8bf44b1024ada46f6cbc1f5132422c09","subject":"Merge \"Remove mox3 from projects.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9bc8c5e8bf44b1024ada46f6cbc1f5132422c09"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-22 15:56:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-23 18:56:12.000000000","tz":120},"subject":"Constrain kombu in global requirements","message":"Constrain kombu in global requirements\n\noslo.messaging unit tests started to time out when kombu version 4.6.9\nwas released, so let\u0027s exclude kombu 4.6.9 and newer versions until the\nissue is fixed.\n\nThe patch that causing the unit tests to fail is:\n\nhttps://github.com/celery/kombu/pull/1193/commits/398aa5b8cd1fe1fc189ab35a2cb0bececf51bf37\n\nNOTE(elod.illes): amqp is constrained due to kombu 4.6.8 constrain\namqp to less than 2.6.0\n\nChange-Id: I79350c8f66c0229768af64631542b5c1b5d48af9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acf99baeb52d60726644e626509a11a0e423bf19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acf99baeb52d60726644e626509a11a0e423bf19"}]},"branch":"refs/heads/master"},"92ac507c1209ac80aea81cf175900bc73f6c3ec9":{"kind":"REWORK","_number":5,"created":"2020-06-23 18:58:00.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/07/737307/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/737307/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/737307/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/737307/5"}}},"commit":{"parents":[{"commit":"d9bc8c5e8bf44b1024ada46f6cbc1f5132422c09","subject":"Merge \"Remove mox3 from projects.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9bc8c5e8bf44b1024ada46f6cbc1f5132422c09"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-22 15:56:25.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-06-23 18:57:45.000000000","tz":120},"subject":"Constrain kombu in global requirements","message":"Constrain kombu in global requirements\n\noslo.messaging unit tests started to time out when kombu version 4.6.9\nwas released, so let\u0027s exclude kombu 4.6.9 and newer versions until the\nissue is fixed.\n\nThe patch that causing the unit tests to fail is:\n\nhttps://github.com/celery/kombu/pull/1193/commits/398aa5b8cd1fe1fc189ab35a2cb0bececf51bf37\n\nNOTE(elod.illes): amqp is constrained due to kombu 4.6.8 constrain\namqp to less than 2.6.0\n\nChange-Id: I79350c8f66c0229768af64631542b5c1b5d48af9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92ac507c1209ac80aea81cf175900bc73f6c3ec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92ac507c1209ac80aea81cf175900bc73f6c3ec9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
