)]}'
{"id":"openstack%2Frequirements~737405","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ic7989902c0a4b297e214173e93759ff56611e8c5","project":"openstack/requirements","branch":"stable/stein","topic":"grenade-em-nv","hashtags":[],"change_id":"Ic7989902c0a4b297e214173e93759ff56611e8c5","subject":"Make greande jobs n-v for EM and oldest stable","status":"MERGED","created":"2020-06-23 01:40:14.000000000","updated":"2020-06-24 14:50:58.000000000","submitted":"2020-06-24 14:50:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737405-1593010258458-07421745","meta_rev_id":"b9ded224d2c1efa54b38505f4213ce95a44924b8","_number":737405,"virtual_id_number":737405,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-24 14:50:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-23 03:04:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-24 12:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-24 12:27:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-24 12:34:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 03:04:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-24 12:27:17.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-24 12:34:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-24 14:50:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"874487aff246766609f7416e7132793a0523ad20","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-23 01:40:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a75f96859ee8dec488ff0925ae8aaef836e3f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 02:43:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cc00e19387a47288b36ed8b8e796bfe : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c03d73a1e2e641a2807d10b100de22e2 : SUCCESS in 3m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8bb929db61e14a3582933b47e6193d85 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e4919c1c88242c28ebee15a7d2f4723 : SUCCESS in 3m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ebfbaa12da8477a924eab68d61e8f8f : SUCCESS in 7m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/95af31c437304fb7a2f1a1c0e2e1c508 : FAILURE in 11m 51s (non-voting)\n- tempest-full https://zuul.opendev.org/t/openstack/build/7de24b34a8fc4aa4868a9ae5a0ac5e79 : SUCCESS in 1h 02m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/eeb138bd2a5f4c309b9f9d707e239e38 : FAILURE in 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7135f570fb618c1064ba0bbd93aa172ab1c99ada","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-23 03:04:45.000000000","message":"Patch Set 1: Code-Review+2\n\nsgtm","accounts_in_message":[],"_revision_number":1},{"id":"cd92b22b1e9c221db6721a50410bc6a5096482f9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-24 12:27:17.000000000","message":"Patch Set 1: Code-Review+1\n\nI think this is blocking https://review.opendev.org/#/c/736784/, which in turn is blocking https://review.opendev.org/#/c/736805/","accounts_in_message":[],"_revision_number":1},{"id":"5e8c446b2fbae8f5d532153d95dc32d9bed96545","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 12:34:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b428bb27257bad7a8b2aa20dd3e95abdf3d797fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 12:35:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"698d713a59f5d9f4d2497ad7d2702be7c5c6eeb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 14:50:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a7fa8b5c8b99519d1558b6c3529ec234ccd66978","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 14:50:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/985d8025730542d8b8821b5e8d832c85 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe820cac5ba24decac71dad5860483bf : SUCCESS in 6m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dfcd8753facd4198a731a423b4609a64 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e28735881e043beac49c52505055648 : SUCCESS in 6m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ea5a51d9f74452b835916bf8bc0735c : SUCCESS in 5m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c0896670c64d479e8775b4a2c224f44a : SUCCESS in 1h 25m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9764824ef92eaf6637edb5cecdfb4030cdf6e075","revisions":{"9764824ef92eaf6637edb5cecdfb4030cdf6e075":{"kind":"REWORK","_number":1,"created":"2020-06-23 01:40:14.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/05/737405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/737405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/737405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/737405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/737405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/737405/1"}}},"commit":{"parents":[{"commit":"d091e1ac3e1790d7633bd5e4592fd3e2a923ec6c","subject":"update constraint for oslo.log to new release 3.42.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d091e1ac3e1790d7633bd5e4592fd3e2a923ec6c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-23 01:39:30.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-23 01:39:30.000000000","tz":-300},"subject":"Make greande jobs n-v for EM and oldest stable","message":"Make greande jobs n-v for EM and oldest stable\n\nAs discussed in ML thread[1], we are going to\nmake grenade jobs as non voting for all EM stable and\noldest stable. grenade jobs are failing not and it might take\ntime to fix those if we are able to fix. Once it jobs are\nworking depends on project team, they can bring them back to\nvoting or keep non-voting.\n\nIf those jobs are failing consistently and no one is fixing them\nthen removing those n-v jobs in future also fine.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015499.html\n\nStableOnly\n\nChange-Id: Ic7989902c0a4b297e214173e93759ff56611e8c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9764824ef92eaf6637edb5cecdfb4030cdf6e075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9764824ef92eaf6637edb5cecdfb4030cdf6e075"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
