)]}'
{"id":"openstack%2Frequirements~737505","triplet_id":"openstack%2Frequirements~master~Ic1ab7927a5a863a099055b46d0cb2a6f99551472","project":"openstack/requirements","branch":"master","topic":"neutron-lib-release","hashtags":[],"change_id":"Ic1ab7927a5a863a099055b46d0cb2a6f99551472","subject":"Update constraint for neutron-lib to new release 2.4.0","status":"MERGED","created":"2020-06-23 12:00:58.000000000","updated":"2020-06-23 21:19:03.000000000","submitted":"2020-06-23 21:17:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737505-1592947054436-47516c0e","meta_rev_id":"6d0d6f4c60278f4c122747b623a15fac0e1e7a41","_number":737505,"virtual_id_number":737505,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-23 21:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-23 12:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-23 12:13:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-06-23 16:54:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-23 12:16:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-06-23 16:54:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 12:09:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-23 12:13:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-06-23 12:16:11.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-06-23 16:54:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-23 21:17:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e2f60d10b04bd09f0f519a7198f12ee270ecac9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-06-23 12:00:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a4ea7937863b1807fd2ffe5c7cc63298932fe4b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-23 12:09:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b36fbfab997651b86330de5ccb6b820fc701cd8a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-23 12:13:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6d67ce861336fc58e588e0e2d7a92a05df4e0da2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-06-23 12:16:11.000000000","message":"Patch Set 1: Code-Review+1\n\nFrom IRC chat about missing u-c bump:\nIt *should* have been automatic, but looks like the job to do that silently failed due to missing setuptools in the zuul gate image.\nThere\u0027s also a daily job that runs that is suppose to pick up all requirements updates, openstack and otherwise, but it appears that one is failing to.\nSo I guess for the time being, anything needed can get manually proposed and we will work on fixing the automation for that.","accounts_in_message":[],"_revision_number":1},{"id":"7bb45632824c16fea6ecbce16996756933149a04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 12:19:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/09f596d8406a45ed925c0c0111af5f5b : SUCCESS in 13m 54s","accounts_in_message":[],"_revision_number":1},{"id":"db88e1df9f0826f311dd046d3a9e8149c65cfa80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 15:38:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/902286a9c75c435c99619cb635cfff19 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3c5842eafde4b8f8e2df172895fd5d1 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0f288d4ce5746069d88415e44b69685 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaa9e088be25424da67c0265e8e66ec6 : SUCCESS in 6m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/627b9c3b5b1b4c748a2f4e5a23a3c49a : SUCCESS in 7m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/292814cdea8a423aa432eab3a7ed7875 : SUCCESS in 7m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bb87260e2a9c4b43b48d9061dfa6c0b5 : SUCCESS in 11m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/583529d02daa4186a868030fb7b49d10 : SUCCESS in 11m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e9b4af90f7ed49578dbd2e2efdc2cc56 : SUCCESS in 12m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c0bb59c80228468bac00b38cf915372b : SUCCESS in 6m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3679491bb1f3449dafd48db60e615a2b : SUCCESS in 11m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ae19e137f9124e1e9a1432d70196738c : SUCCESS in 10m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/687cb617316d4c6ea24a33649078e188 : SUCCESS in 16m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7c6ae257ad524eb6a0e3abda41195c7e : SUCCESS in 27m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/79d446341f924b9da44cca1efc20a4c3 : SUCCESS in 16m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2f68fd8fe43f450293c1302a4ecbb0cf : SUCCESS in 16m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/445825355b6e401d9f84aa2d0184479e : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a38bf5918f124352bf34654532603ffb : SUCCESS in 8m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/681f5765759b4fceaeb5f426e41e66bd : SUCCESS in 17m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe4802ddedbb4bb4b2ea9c9704a8ebe5 : SUCCESS in 56m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb9f2f142a094720addeadba7f8edd0f : SUCCESS in 1h 14m 34s","accounts_in_message":[],"_revision_number":1},{"id":"dd9fdc74c85eef38a4594ab986c9a9dd9b8ad558","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-23 16:54:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"92b1ba0bd327bab41f587b25d4fbf38f852fd0eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 16:55:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5b20dfb8d01b281668b87114dbb7f073dbe7d3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 21:17:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8ff54cf9e1f416a9ef02277e314401a : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ead324273264c428f17c0e000c89512 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/437dae3611b14a6387e0dc5e26c2f9a9 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69c2991516684183a80f624f3789c53f : SUCCESS in 6m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d79fce0b21345fcab250436b092412e : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/09788f6e94af424db455b88a3e19063f : SUCCESS in 4m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c49bd38e2d9e42ef8f59df98617d0f4f : SUCCESS in 11m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2033ada333264f089d6e2b0543e1176e : SUCCESS in 12m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0989c0b35ecc4278ad690a8c1db1bc6d : SUCCESS in 10m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/884324a0e3a74538b592030f0aadc082 : SUCCESS in 6m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d0cc87d579034b23aee2306efd52aa88 : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf4f9e3d53014049a25bc045837be3ca : SUCCESS in 12m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/867e1bbdc39a4d2cba58f466e3ac13eb : SUCCESS in 11m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/525c9f8c85a848fca41998c23ce5ddc0 : SUCCESS in 31m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fb00a13b856c416cbd560f0c780c03df : SUCCESS in 17m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/efa0bdf3748b40efb735948ed3ca6130 : SUCCESS in 25m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/782199dd8c0243fe8942dd12f54dec09 : SUCCESS in 6m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9bfa8e0106794865b18408be73f6aa52 : SUCCESS in 7m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9d0f0f55b42f441da653269458e81973 : SUCCESS in 11m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ccd645923dc74ea1b53d42baebf7045c : SUCCESS in 10m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b838ad814274cc3b546e712e4402f02 : SUCCESS in 57m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ad572bc249c47219bcff85f6072532d : SUCCESS in 1h 24m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0aaf04e78bb1f0062f72d0dc34772c4dfcf1ae85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 21:17:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c00ed78a49bf3153566aa15c1ca29867946b4664","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 21:19:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/175ea6e509c243308856fb8142fd657f : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0559be261ee86bdef6a97c802f53a263057b844a","revisions":{"0559be261ee86bdef6a97c802f53a263057b844a":{"kind":"REWORK","_number":1,"created":"2020-06-23 12:00:58.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/05/737505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/737505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/737505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/737505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/737505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/737505/1"}}},"commit":{"parents":[{"commit":"5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc","subject":"Switch node tests to inherit nodejs-run-test-browser.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-06-23 12:00:26.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-06-23 12:00:26.000000000","tz":540},"subject":"Update constraint for neutron-lib to new release 2.4.0","message":"Update constraint for neutron-lib to new release 2.4.0\n\nneutron-lib 2.4.0 was releaesd in https://review.opendev.org/#/c/735687/\nbut somehow we cannot find a patch to update upper-constraints for it.\nI am manually proposing it.\n\nChange-Id: Ic1ab7927a5a863a099055b46d0cb2a6f99551472\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0559be261ee86bdef6a97c802f53a263057b844a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0559be261ee86bdef6a97c802f53a263057b844a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
