)]}'
{"id":"openstack%2Frequirements~737561","triplet_id":"openstack%2Frequirements~master~Ifb445e3baca10fea27aad96f34db5b416e8c6457","project":"openstack/requirements","branch":"master","topic":"dragonflow","hashtags":[],"change_id":"Ifb445e3baca10fea27aad96f34db5b416e8c6457","subject":"Remove dragonflow from projects.txt","status":"MERGED","created":"2020-06-23 15:20:33.000000000","updated":"2020-06-24 02:20:33.000000000","submitted":"2020-06-24 02:19:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737561-1592965147229-3f73efbb","meta_rev_id":"88dc116f8921675aad353ec16d4fa3696d1323ec","_number":737561,"virtual_id_number":737561,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-24 02:19:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-23 21:20:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-06-23 21:20:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 21:20:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-24 02:19:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da74f571c5e1c0cc54e79b478954d3ae171bfa1a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-23 15:20:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db14dfdd2712fc676afc67975f7e1037cc5670a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 20:08:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be4d9929f99a4cb2b3f4c875e3d8b7aa : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d743db35db1a4c85ae71b04016b6c09a : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb300a8f724c499894a23df2ebcfb719 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d01b4f3b5e5143948894010016a680c3 : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/27dc45eae965450ca9a29ebdf567923b : SUCCESS in 7m 15s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/cb9d90c8bb6a423b8e23c14b8e176fd4 : FAILURE in 5m 33s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e825ec700444573bd9700f7f26fb287 : SUCCESS in 50m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1098d804716b4cfa84796487bfbf465b : SUCCESS in 1h 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8beed1640c66f53e8cd5294409f2ba9bdfda3328","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-23 21:20:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"488cb92f349d331a16961f35fccc6a02eff41279","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 21:21:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e5b66aa10370be705c7d72f1ba97593c970f35a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 02:19:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d92d08e8c87401eb39210900f74d83c : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9640760fa0e435998a3ecf59c8e7680 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6897ae0f48640c3968784e4e9191c8e : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d8292412f7446f296b1575fa941a8b1 : SUCCESS in 6m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/873e8715a6b94e65b242c2ec6e2cebc9 : SUCCESS in 4m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60924b4aa8804f9aa7f4443afdb19bfc : SUCCESS in 55m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8b68a9065b34ac59185ce3290eec5ce : SUCCESS in 1h 31m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8df96146f797261949e27eaa99b40eafe44775b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 02:19:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2108f992ccdab0455d0c497cf2943b206552e9ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 02:20:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f050f81d26d342c299bb492a59d02470 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"25d7527a5b354cc077c4ed687afec71c443a11e1","revisions":{"25d7527a5b354cc077c4ed687afec71c443a11e1":{"kind":"REWORK","_number":1,"created":"2020-06-23 15:20:33.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/61/737561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/737561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/737561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/737561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/737561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/737561/1"}}},"commit":{"parents":[{"commit":"5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc","subject":"Switch node tests to inherit nodejs-run-test-browser.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-23 15:18:41.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-23 15:18:41.000000000","tz":-300},"subject":"Remove dragonflow from projects.txt","message":"Remove dragonflow from projects.txt\n\nDragonflow is being retired [0]. This is step 1 from the retirement\nprocess [1].\n\n[0] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015600.html\n[1] https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nChange-Id: Ifb445e3baca10fea27aad96f34db5b416e8c6457\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/25d7527a5b354cc077c4ed687afec71c443a11e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/25d7527a5b354cc077c4ed687afec71c443a11e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
