)]}'
{"id":"openstack%2Frequirements~737584","triplet_id":"openstack%2Frequirements~stable%2Frocky~I9c8e226166358c7d4e3b5cd57512ac08497aee02","project":"openstack/requirements","branch":"stable/rocky","hashtags":[],"change_id":"I9c8e226166358c7d4e3b5cd57512ac08497aee02","subject":"Update constraints for eventlet to v0.21.0","status":"ABANDONED","created":"2020-06-23 17:02:34.000000000","updated":"2020-06-23 22:34:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a5214ce56ce267bf9c8e3649310b8a70e0fd46cb","_number":737584,"virtual_id_number":737584,"owner":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-23 21:53:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 21:53:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bbdf5858c9f5a2db411003b3a15087fa6c3067ca","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-06-23 17:02:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6329483f400cd56effb9055b81e96174b6dc612","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-06-23 17:19:50.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a232548010b846d2ee497d1263b97b3b90123d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 21:53:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/740280462cac42bdbe4b977afad6df43 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/302d1c75ae484466a533a4a26e36a0a4 : SUCCESS in 6m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e7a7f3526af648bbba1c3a4defe1df4d : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fbdf7b2cea9a41569f26b952f5b753a3 : FAILURE in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d840fcfaec844131ae1600f33922bcb7 : SUCCESS in 4m 45s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/314ff3e898a540c0bf7398027a36c30d : FAILURE in 9m 50s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/9ddb7ebaf671443099e925a6bb78114a : FAILURE in 9m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/71ff3b15e55845f09fbee629f4903a33 : FAILURE in 8m 09s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c6d48ffa46c24c6198a07db6831d98a3 : FAILURE in 6m 08s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/72d30d6809b247a88a077ccd000d6796 : SUCCESS in 12m 27s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a255aba44c6d435aa68123db6b2ace4c : SUCCESS in 5m 39s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2a6eb43f4b914cd08aba2317331eac10 : SUCCESS in 5m 58s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d6f56b3d85e64adab40a82b6e1cfc9d2 : SUCCESS in 12m 04s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/df287f77904242e894ca8fa8642d224b : SUCCESS in 10m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/25b0da6a01134f9795a19871050f5f3b : SUCCESS in 17m 04s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/490b3fb96c8e4ed08a5abff03e561156 : SUCCESS in 14m 42s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/24b57b80ec4c40fbab4c65dfd8d69465 : SUCCESS in 26m 37s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/858d085a5a7d4f8da78d171c5f040fca : SUCCESS in 26m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c829fbb6593c47b7bf58f9e6a8f7a9fb : FAILURE in 5m 36s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/fd665848b9954ba89e1c31944fa54a05 : SUCCESS in 18m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e95098e84594aef8bb16f23665faea1 : FAILURE in 8m 03s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/fc25e67d54d54d27a8f9e47f0f832bce : SUCCESS in 6m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/be53ee5cef8d41a584766689cb038a8a : SUCCESS in 12m 53s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/1c0cdf1b42e54088a3f3db18918ecd17 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a79d651369054d6b952f13ccf9f0f54b : SUCCESS in 4m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/668dc037a3514524a36003e5ffa693db : FAILURE in 16m 54s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/f7b8536b84b5464abc9c0decf09c7325 : FAILURE in 16m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fbe6292d31eb48a6844a5690e928a8f0 : FAILURE in 17m 21s","accounts_in_message":[],"_revision_number":2},{"id":"a5214ce56ce267bf9c8e3649310b8a70e0fd46cb","author":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"date":"2020-06-23 22:34:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"391fc6ebc3acfdbefb84e0cf78791b5145aee798","revisions":{"c8ee7668021dcd89ada1f495e134b66c90b5f95b":{"kind":"REWORK","_number":1,"created":"2020-06-23 17:02:34.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/84/737584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/737584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/737584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/737584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/737584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/737584/1"}}},"commit":{"parents":[{"commit":"e70b99f0d05adefd12b16f4f8882a9140a21e220","subject":"Pin diskimage-builder \u003c 3.0.0 on python2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e70b99f0d05adefd12b16f4f8882a9140a21e220"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-06-23 17:01:57.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-06-23 17:01:57.000000000","tz":-300},"subject":"Update constraints for eventlet to v0.21.0","message":"Update constraints for eventlet to v0.21.0\n\nEventlet v0.20.0 is affected by [0] causing neutron-rpc-server to fail\nwith SSL enabled. The issue is fixed in v0.21.0\n\n[0] https://github.com/eventlet/eventlet/issues/371\n\nCloses-Bug: #1884808\nChange-Id: I9c8e226166358c7d4e3b5cd57512ac08497aee02\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8ee7668021dcd89ada1f495e134b66c90b5f95b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8ee7668021dcd89ada1f495e134b66c90b5f95b"}]},"branch":"refs/heads/stable/rocky"},"391fc6ebc3acfdbefb84e0cf78791b5145aee798":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-23 17:19:50.000000000","uploader":{"_account_id":8863,"name":"Andrii Ostapenko","email":"anost1986@gmail.com","username":"aostapenko"},"ref":"refs/changes/84/737584/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/737584/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/737584/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/737584/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/737584/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/737584/2"}}},"commit":{"parents":[{"commit":"e70b99f0d05adefd12b16f4f8882a9140a21e220","subject":"Pin diskimage-builder \u003c 3.0.0 on python2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e70b99f0d05adefd12b16f4f8882a9140a21e220"}]}],"author":{"name":"Andrii Ostapenko","email":"andrii.ostapenko@att.com","date":"2020-06-23 17:01:57.000000000","tz":-300},"committer":{"name":"Andrii Ostapenko","email":"anost1986@gmail.com","date":"2020-06-23 17:19:39.000000000","tz":0},"subject":"Update constraints for eventlet to v0.21.0","message":"Update constraints for eventlet to v0.21.0\n\nEventlet v0.20.0 is affected by [0] on Python 3.6 causing\nneutron-rpc-server to fail with SSL enabled. The issue is\nfixed in v0.21.0\n\nThough issue affects Python 3.6 and above, there\u0027s a value\nto try keeping its version consistent among all Python\nversions.\n\n[0] https://github.com/eventlet/eventlet/issues/371\n\nCloses-Bug: #1884808\nChange-Id: I9c8e226166358c7d4e3b5cd57512ac08497aee02\nSigned-off-by: Andrii Ostapenko \u003candrii.ostapenko@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/391fc6ebc3acfdbefb84e0cf78791b5145aee798"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/391fc6ebc3acfdbefb84e0cf78791b5145aee798"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
