)]}'
{"id":"openstack%2Frequirements~737600","triplet_id":"openstack%2Frequirements~master~I84f8508e3a0add354297a4fc6dafa18da398af4c","project":"openstack/requirements","branch":"master","topic":"dragonflow","hashtags":[],"change_id":"I84f8508e3a0add354297a4fc6dafa18da398af4c","subject":"Drop crc16 from global requirements","status":"MERGED","created":"2020-06-23 18:41:40.000000000","updated":"2020-07-30 01:38:06.000000000","submitted":"2020-07-30 01:36:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737600-1596073003864-76a6aaaf","meta_rev_id":"d22fe441346edbd6a424f885d1ca8804c911d72a","_number":737600,"virtual_id_number":737600,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-30 01:36:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-07-29 11:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-07-29 11:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-07-29 15:11:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 02:59:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-29 15:11:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-30 01:36:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5369b4284d016cd92d2d1caf6d89a0c799a68395","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-23 18:41:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"660a27bf7c5ec93b377c6e50c5ec0b6123b4bb40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 19:02:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/563495946bbe466aabf5d2c6980e1c99 : SUCCESS in 16m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8605494a6d69ba9d61754811f54f5eab0f4539ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 00:47:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c35c405e342a4537942a4a37aee9c5cf : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f13e8f1c2f84067880b9c41c0a0b425 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb266e1be8664d4b9d948c648a272bae : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/593b5426e74649f5b7abada7996e7c03 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a7c15cd92d74d62a619dab4beaba7a8 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56230933b9de4523b9ff96f3ff5ecac2 : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/38ee96b01f6044ec89f00602777e3077 : SUCCESS in 8m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/58e5005bb3fc47a4b9d290dc966a0cc9 : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b35fc6a9c4364da2ae044bbda9b918df : SUCCESS in 9m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f03717fe927046b09230408cabbaf559 : SUCCESS in 5m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bf631ef2c4f34a4ca42ed283132df942 : SUCCESS in 9m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/15debedd77254ec5b0127c9a5d78c284 : SUCCESS in 7m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1244e655ced949ecbb89ff61fa41eb55 : SUCCESS in 14m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/77ebd8235fad4e369a755b980b566a53 : SUCCESS in 24m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/802772b4656d414494fe7ff0b0a9009d : SUCCESS in 14m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/99f4334717c34b5298cabe4607dbbfbb : SUCCESS in 17m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f3271673243949419ebba7f8e0c379b7 : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/810e4672d2d446d2ab37be672ddc9fd2 : SUCCESS in 7m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/23480b07b18842529f095c38e42bc577 : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbb87e76d26142cb89e13cf93904c3e5 : SUCCESS in 53m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c33ac446a294162b9e0653a2c50ec78 : SUCCESS in 1h 18m 39s","accounts_in_message":[],"_revision_number":1},{"id":"cd4660d4b8ed31e68744164244e959775ac8a098","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 02:59:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1860da244cd7f3b8c5bdc2a4c6144f8dbe23526d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-24 14:31:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"56abf15040e75d9cf605b103e7640aafaf979d5e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-29 11:48:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c7e821f4e3e5958479f039d1d56d61784f18c07f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 12:33:16.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"cbbeec1b555fc3cd692e1ba23775089b0270da7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 14:58:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be8a7e0461b84030af5c141e66ef97ce : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/177e9323b61046cbb7890c8d9ef8c587 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20ac2092595d4484825e07d336713e91 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/995a223b2196450a896526d35fa04aae : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4f54acfa24141fea1b0984852f17e50 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd7b977572f34250bddb651f86f5a57e : SUCCESS in 3m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cff911317e6b4fccbd2f480496452080 : SUCCESS in 8m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7c8b1eb8ce8a4665a4fbe8a550039d26 : SUCCESS in 10m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/de66d455610e4be19bc3627afcf5ca35 : SUCCESS in 10m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b7e0048b9c664fb29a6d7e4f09e05838 : SUCCESS in 6m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8f993184857343a8a4d61e0e0fa2b6fb : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af2ae7d69aca4d40ab393af909536116 : SUCCESS in 7m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a3698e34e3784fd7935bf5d49d306dc2 : SUCCESS in 8m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bd05eb91ddaa4691a5f5c8c10327382d : SUCCESS in 13m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/65a11326a9da4cd08d504d776a7e7aae : SUCCESS in 23m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/486a811c9f9741be95be2e85a33330d3 : SUCCESS in 14m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/655616ec7d664dbf984376c690d6cd9f : SUCCESS in 18m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2a04ef46d06943e9b1dbfbd246691b80 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b361b9545afe4785a7aa45611b4c8181 : SUCCESS in 7m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/64798aa2e885447ea95c4235a5073ec2 : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc32b4f42498401c81c3d59c9f2297f4 : SUCCESS in 1h 03m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc8e0ba26b3844ef8347f1de2f8e7a29 : SUCCESS in 1h 34m 59s","accounts_in_message":[],"_revision_number":2},{"id":"ec32d2b9820aa760c3307a84b33851328be51c02","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-29 15:11:03.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8ae379beec728b78956adb4beb9077f89c2cb1a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 15:12:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9910cdff4eb96fc6e30b7b9680b868c3d2501bc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 18:19:06.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf90cb49440d44f5b7f95e563277e9fc : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12a0ce930c4f49feb14c4d0f6ac6d1b2 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de604cd64bd74d12ada17108615f7d3a : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72cd41285aca451ebac78ad04058625c : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1d3298166e1430898d39c77a10ee960 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6c11c6c83dd43e5890c45115588810b : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/698624fdaf8244c5a6f61f4b5ce9407a : SUCCESS in 8m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/babfc104dbe74abbae9c38426f89ca68 : SUCCESS in 10m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5c49945e8b024a279950b24836e37d89 : SUCCESS in 8m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2415573c1bd14b9b8703911c7522b6ad : SUCCESS in 5m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1afaa7cccbd4464fa8e549e2fc0204af : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8718a6fd12c348cd835c243d28989640 : SUCCESS in 7m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fd2436acfdec475384da5e5fc3834c81 : SUCCESS in 7m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/917e35d25534485880c7c52a948804e5 : SUCCESS in 12m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ecd1bbf268d34ca59f9028624ac73b2e : SUCCESS in 31m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cdee514565b04ec0a89091e442827872 : SUCCESS in 13m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/00f59da3156b4410b31fbe597fdee048 : SUCCESS in 20m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a9d439ac4ee34691bfc8312db09e6b26 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/db8514ee0b744cdd910bb369c5a027b1 : SUCCESS in 9m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bddff40f71d14233914a73d348013c57 : FAILURE in 18m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/eee20b0461154ef4b0b47141a6124df8 : SUCCESS in 9m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d892daf5fc414eceb44ffd4b25ddde08 : SUCCESS in 54m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78fa5e698cb64cb0a1013b28c96a3a08 : SUCCESS in 1h 29m 49s","accounts_in_message":[],"_revision_number":2},{"id":"0c552a3354bca19a5d34223edc26f6351ab8170b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-29 18:23:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"38b4edcaceffe180e03812c70c313531ce839809","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 18:49:11.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"ee8b2a493661b2b6e857014a0f2042490d5228e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 00:11:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/852ad77be67b467f91b936329c786fad : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cba482738af4fac93e5623833ce7950 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c9af2f0778e4066b881a3b32940f4a7 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5526fe8a36e8481f9407bf6d5e7526e5 : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fff67da71dfe453a931e12deb7863d12 : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7dc383aad6414ebe84843f17046508bc : SUCCESS in 4m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/488686fec9ed4902bd64ab1bb585cbfb : SUCCESS in 8m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/57b31cb0cf5d4d99bc004000be39e515 : SUCCESS in 11m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e83bf4af694d4485b54ea63b6fa1f873 : SUCCESS in 8m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a399d783d2764803b196410f3f068a8b : SUCCESS in 6m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7dfdf43d38b045388bb0955bd97fc07c : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/15505caec70648d1a0677505df7cfd2d : SUCCESS in 6m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2d274aff56d34af38ed33bcfb06e4053 : SUCCESS in 7m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/93749bdf5a8b4ba79dec0a74bc2b2a02 : SUCCESS in 17m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fd2c1abbbbaa4534a2fecc0d5e4dea58 : SUCCESS in 28m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aa54de5bc8604ed4be8cf98ca898b5ba : SUCCESS in 12m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/07ca3b1399564f0b8c1837652636f4d9 : SUCCESS in 12m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5501fccde8bf4eb380a3b8a8880bdb04 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86374f49dc13471c9db6587c97a38139 : SUCCESS in 6m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e79c835d36a242e3b5cadcfcdbf7c075 : SUCCESS in 18m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce627bd3c6c447799855273aaf5360fd : SUCCESS in 50m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/659dbca821c743ddbb6725d3fa04405c : SUCCESS in 1h 07m 01s","accounts_in_message":[],"_revision_number":2},{"id":"bab9723ce412dc17592b4a22d97f3add8782d85f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 00:12:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"324c4c910387c9cefb28ca2a92e706097c6ae9f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 01:36:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3a485ebd3694e6883ff9fedf1d0accd : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/805694bce3c34c099b4b3ad3e81eb127 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0898665b768c456c8f65bd45a69cfb03 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ae8990cc4304dd29ef858a3b4a69003 : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83da4432052d4d9fabb08457d6d8e34c : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2753620d1ab9465b8d14674b77752916 : SUCCESS in 3m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/feff923f79a0437782108e0a979d32c9 : SUCCESS in 7m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/30ce19fc6fe34129b344ec66f4267da5 : SUCCESS in 11m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/70caa68e3fdd4a8bb5452de6e4c3fb6c : SUCCESS in 12m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7035a703ceed420785488ebc5422bd1d : SUCCESS in 5m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8e007706a2da4385befb13414bdb8e7e : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af01683fff7043ac817d4e6eebceafe7 : SUCCESS in 7m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2ba5d4a9c27047a9920f60d66110e597 : SUCCESS in 7m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b71d228801cc453d92967d1ce285f175 : SUCCESS in 22m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/77ea89639dd9429aae0b603def11f96f : SUCCESS in 15m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/507086d74eae45b180a632ec0c283ded : SUCCESS in 17m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e876ec13c354ea3a5908d5eedc18811 : SUCCESS in 32m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/df1465c118a84569b2a7f14119444765 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/302ce490925441fd8982bd3e2a00a6ea : SUCCESS in 6m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8c6328242f924d2585fcf0991625fe26 : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/017eba8f21274275b1a8cbe27d49f2e4 : SUCCESS in 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e51bddf3c44d408b93a26a239fea577c : SUCCESS in 57m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/27039fb81dc543c3b9dc9916f565f874 : SUCCESS in 1h 22m 28s","accounts_in_message":[],"_revision_number":2},{"id":"309d05b62430dfda99b449ac9f463f5456cd33eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 01:36:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7d7ae2662986b603e025568dd284cafd1a5681b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 01:38:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac29e83f84e4b6e9e0fa37f0dbf78a6 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"24a17033d6d410467c9c8924bb34bdea5cb53b82","revisions":{"737fef7e957eb2822490b542cbbf0808c9c51e29":{"kind":"REWORK","_number":1,"created":"2020-06-23 18:41:40.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/00/737600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/737600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/737600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/737600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/737600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/737600/1"}}},"commit":{"parents":[{"commit":"5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc","subject":"Switch node tests to inherit nodejs-run-test-browser.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-23 18:40:03.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-23 18:40:03.000000000","tz":-300},"subject":"Drop crc16 from global requirements","message":"Drop crc16 from global requirements\n\nThis package was only used by the openstack/dragonflow repo, which has\nnow been retired.\n\nDepends-on: https://review.opendev.org/737564\n\nChange-Id: I84f8508e3a0add354297a4fc6dafa18da398af4c\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/737fef7e957eb2822490b542cbbf0808c9c51e29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/737fef7e957eb2822490b542cbbf0808c9c51e29"}]},"branch":"refs/heads/master"},"24a17033d6d410467c9c8924bb34bdea5cb53b82":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-29 11:48:32.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/00/737600/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/737600/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/737600/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/737600/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/737600/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/737600/2"}}},"commit":{"parents":[{"commit":"8e268ff74e13e154db7568a560818f71b1ea8514","subject":"Merge \"Remove skydive-client from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e268ff74e13e154db7568a560818f71b1ea8514"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-23 18:40:03.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-29 11:48:32.000000000","tz":0},"subject":"Drop crc16 from global requirements","message":"Drop crc16 from global requirements\n\nThis package was only used by the openstack/dragonflow repo, which has\nnow been retired.\n\nDepends-on: https://review.opendev.org/737564\n\nChange-Id: I84f8508e3a0add354297a4fc6dafa18da398af4c\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24a17033d6d410467c9c8924bb34bdea5cb53b82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24a17033d6d410467c9c8924bb34bdea5cb53b82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
