)]}'
{"id":"openstack%2Frequirements~737620","triplet_id":"openstack%2Frequirements~master~Ic9d23c9ba37b5ef68b315db3c14ed5ffe2211b3b","project":"openstack/requirements","branch":"master","topic":"dragonflow","hashtags":[],"change_id":"Ic9d23c9ba37b5ef68b315db3c14ed5ffe2211b3b","subject":"Remove hiredis from global requirements","status":"MERGED","created":"2020-06-23 20:18:35.000000000","updated":"2020-07-29 19:36:24.000000000","submitted":"2020-07-29 19:34:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737620-1596051288259-be75a880","meta_rev_id":"db43248829125949e7260eb2a79be4e16ec2bf44","_number":737620,"virtual_id_number":737620,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-29 19:34:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-29 18:22:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-07-29 11:50:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-29 18:22:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 11:38:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-29 18:22:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-29 19:34:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6513bec832fa9553c65ce6dcde8765e081b6485f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-23 20:18:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d8f15cc4ed539038d0c6a6dc5e8c4dc18ccb9a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 20:36:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f2a13f34d0d449e3b57871a7fd52ba7e : SUCCESS in 12m 56s","accounts_in_message":[],"_revision_number":1},{"id":"9158bb31caf59212479f066089a0d48947d3e4eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 01:14:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5ee000e65b84e34a18317152192ba86 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/939b1c830f2746f5bd1ca86dd3b68353 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a960424e15274f948f86247617673d6d : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/deb751491c38405295f61d92b43195e3 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a10f8286af3049eb8441ed2153480c0a : SUCCESS in 5m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a314d40b5d244093b7ce9714cd9d12e2 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5b5d66c2196345d7b61616db469c4ea0 : SUCCESS in 7m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8861470fadf64a8ea78d08d0f09154a7 : SUCCESS in 11m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f134e692019e4f62bb1724b3dc157a1e : SUCCESS in 11m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2a6e474c8fad456692676e5d1e3ba698 : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3bf26b62fd5345dba39de3de574d5340 : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b65966f5f4f04ead9aab154862275a06 : SUCCESS in 8m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/854fc94ad40d4c19a9b4c7c39b458a3b : SUCCESS in 13m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0b9c81461e094f1ba4fd09528c987050 : SUCCESS in 27m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4bb63a2a927f47c4993cc21912498660 : SUCCESS in 17m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/385ede5a01d044528e493aa0701d3f91 : SUCCESS in 21m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d2376021e4a248d7adc3f991d5a3ee42 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/306a16b005474d78ba6a620d53cdfbb6 : SUCCESS in 6m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/37af8f8a41d64b66a97a5047ea4e82a8 : SUCCESS in 12m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5a945fd174041b79ddb0fe5992afc17 : FAILURE in 1h 06m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/718b5912956042c59c34569c7e3a5d53 : FAILURE in 1h 12m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4e2d96eb6bf01fc277ae4a061280a5d89b3d5be7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 13:08:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1199345ffae378acc7944b0d17e7ba14676f977d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 13:27:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1b174aa335174e298791710defdb1929 : SUCCESS in 13m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8f653da739e417200ba0a3effe7ced522da8bfb1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-24 14:35:44.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM https://github.com/search?q\u003dorg%3Aopenstack+hiredis\u0026type\u003dCode","accounts_in_message":[],"_revision_number":1},{"id":"a4ec7310fa8efd56235e5b4278cbef532fcfd473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 19:03:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c41b5e8eab54d499dfacdd82575cf3b : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16fec8b15c8b4266bd7c92aa9cbfffe3 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89f4d1a5c1064f02bf63186e567bd6d1 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40df0e1e4ef74375b884b9c617ff3a9a : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66b8eb5ed1e541cd9974e4f4281c3645 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/60bc83ca3ff749cd97d933929fa310c1 : SUCCESS in 5m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e4d88813bc394b2ea8b1282f66b3e7d7 : SUCCESS in 11m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fa8313d479de4b76aa1a03fa99f8161d : SUCCESS in 13m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5e574ce93783416fad57ea7df40c1543 : SUCCESS in 9m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2b8f4d3867044cada5bc5d5b2f01334d : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bd8506157074497586d815a7cc752e22 : SUCCESS in 21m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3135033a711741e4a364b2c380ccd754 : SUCCESS in 8m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b3b36e743421410e9e9a52bd57d31f1d : SUCCESS in 22m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b2b7fd6934c04b8b98ab66867c5e75eb : SUCCESS in 34m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2b22d54e88794ac89e602110b1327f14 : SUCCESS in 15m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/497ade293d4f46d58c0e56256d8be5ac : SUCCESS in 18m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cf4dc94ab0054b7abdf94f428bfd5538 : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/01ef48d207cc44d6aee13e5b95b049a4 : SUCCESS in 8m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f74f4e70a48b4ad1ad93690ce203b5b2 : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d130cd438d8347a691b78a9f5f77ff38 : SUCCESS in 56m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34c6b6f9c90b460593f5a93be5d753fa : SUCCESS in 57m 23s","accounts_in_message":[],"_revision_number":1},{"id":"86d9ec240d45db9fdb136d9a87dda045c7325946","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 20:10:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eed9d1bae985f4fe3a7f14bbebb6287f0f14fd6b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-20 16:26:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0560225ed3f6c46b06db3d24b8cf60bb8e56bd99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 17:08:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4dbd628fe7cd48f99c5465c9b279688f : SUCCESS in 35m 40s","accounts_in_message":[],"_revision_number":2},{"id":"11d135f445447255cc3f1460e44277b27df78832","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 18:02:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f6afd334ece4f9ea0ddf4cb806c9bd4 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8732ed991553464b9ae15cc3f68b5582 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dff47dd7f30444d1a647006b9bd8edb3 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6cde470eaff486bbc8bb9e121e843d8 : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b4bfb6c2d864b3fb6962cebd9dc4c74 : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d8bad1c3f92b4c248496ff36fb55c981 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/54aa86d98dd74172bba2d11cc23870e6 : SUCCESS in 8m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/107112e952da4c8eb02417a96c37b44b : SUCCESS in 10m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e4e39e9dc5a746088c546c69b94912c4 : SUCCESS in 10m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/136abbffea3342468a2a62e2f0f0eb4b : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/268cc7ce99954cdab889f5f3d5b3a769 : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d1092a82c9ab401b86960b12e05dc4b9 : SUCCESS in 7m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/65e8df73c5f6474ea1a4ffad7b66bf04 : SUCCESS in 10m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e96144326d27447e83cbcabc621898b7 : SUCCESS in 12m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0d8ec74c8e9b478981bfc77a8ca71470 : SUCCESS in 30m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c9dca432b63f4235a365777375da7f8e : SUCCESS in 14m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b43440337f3b47979b34b4a84d98c001 : SUCCESS in 20m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d3a863a8a3bc4a30ba86013708bc93e8 : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/393fc3b8c1bd44868d4a8922f2bee8fe : SUCCESS in 9m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/46bfa6f2aeed4288bca395ff792e041f : SUCCESS in 15m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/315c7abd77b54f65ad1ac76aa924cadb : SUCCESS in 56m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1accc2847747401985bdc49f4cb22ed2 : SUCCESS in 1h 29m 13s","accounts_in_message":[],"_revision_number":2},{"id":"001f6bbabbc59511e906dd865cddd1d3c681f318","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-22 11:48:02.000000000","message":"Patch Set 2:\n\nhiredis seems doesn\u0027t exist or removed here, I only seen standard requirements update, maybe the commit message should be reworded or these changes abandonned, thoughts?","accounts_in_message":[],"_revision_number":2},{"id":"0e6a23d811aed69a24090b78f4a0c986780fc161","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-22 11:52:57.000000000","message":"Patch Set 2:\n\n\u003e hiredis seems doesn\u0027t exist or removed here, I only seen standard\n \u003e requirements update, maybe the commit message should be reworded or\n \u003e these changes abandonned, thoughts?\n\nNot sure what you mean. Can you elaborate?","accounts_in_message":[],"_revision_number":2},{"id":"2a4f2093495872aaa7675af8cb60276ef2761d78","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 11:38:28.000000000","message":"Patch Set 2: Code-Review+2\n\n\u003e \u003e hiredis seems doesn\u0027t exist or removed here, I only seen standard\n \u003e \u003e requirements update, maybe the commit message should be reworded\n \u003e or\n \u003e \u003e these changes abandonned, thoughts?\n \u003e \n \u003e Not sure what you mean. Can you elaborate?\n\nSorry, don\u0027t take care of my previous comment, I re-checked this patch and it looks like ok, I surely got my feet in the carpet with gerrit during my previous review, LGTM now.","accounts_in_message":[],"_revision_number":2},{"id":"21fee5b7b1177312b415e7a1e9cae9f582dc9863","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-29 11:50:22.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"84442158b69c92f129cd4b66d8702bf397c6f798","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 12:52:11.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"4e01a64a5086110e062c4ef3e521ebd69af986c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 16:37:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/491c8c4345884557bd8796cd0c590a42 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64740efc854d454d9cf39fc0aaec7af7 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd18a925cbb942d3a2f7f440bd79de9e : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa49fcc4946e4629bd6e4a31f1ee9565 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9ca65808633449f91ac05887a5e5ee9 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/feb896ebbd444c45a7c5dfaab4c71719 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/52133b23a32a450f91570deae93e3666 : SUCCESS in 7m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8262381fee1b4fe6be023bc5c4f83476 : SUCCESS in 14m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/401fced0fed84839bc95a815a253102d : SUCCESS in 8m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aa9b89ade6584911afc16b29fe6e19fd : SUCCESS in 7m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cee31b2eb0ef4766afa6ba48ac958c7a : SUCCESS in 7m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9ab6e9be904d45258abf656555c7f469 : SUCCESS in 6m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/276917fc0eba433e8b085730d80453ca : SUCCESS in 7m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/906ce18337f64552850dff79b9e40bbc : SUCCESS in 33m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1d43ffadcd57459ca19186cfa84e101e : SUCCESS in 28m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ef1cd6fbbaa942788ea8b6f69e1fbfa3 : SUCCESS in 13m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b47baac5ffcb45e0b0b41e5ff5a82883 : SUCCESS in 17m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0f741ead42a04116b6a8368a18265218 : SUCCESS in 6m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/663a61d62c144c53a98d103bcec08a0c : SUCCESS in 7m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/01dd8a7cc58b4ab6aa000f9868f1b0f0 : SUCCESS in 8m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30ad4d84a00c4fae97cca61889d88cee : SUCCESS in 1h 03m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/252a94a3c99740808964f88bef997f03 : SUCCESS in 1h 26m 49s","accounts_in_message":[],"_revision_number":3},{"id":"aad4a8027eee9e1cd633a304725a6228d3800ea0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-29 18:22:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dd01091b4629adad0f5a6584c28d66b217d3d037","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 18:23:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0598f2bbd6a644f68811eb72fc1c9c4d8e60e68b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 19:34:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3149fd192e94783979641685027a3ff : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ba8f4fe47aa4e11a82b847c510d8e44 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54343a9834de45b6b04d83deeb814b49 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2d9f8ec80d54ba498bd02707bda3097 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b464717926a4bbc831292c4e076c69f : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8fafe9effd734c5aa5c048f66c633f09 : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a749a913e64f4a938221c1a61268aa02 : SUCCESS in 8m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ce664baba2194dd898b4e736328e5e77 : SUCCESS in 12m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3a4e24c08f5543569969cb2c8d8e3315 : SUCCESS in 9m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e78fc5e29ab1451a90ae7da57e5327ea : SUCCESS in 6m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aebebd091eea471dbba7c810697d5528 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2bdfdbd10ed5444d960a30efaeb82cce : SUCCESS in 8m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fd1628d4cee147f084bfd08b2b2596ea : SUCCESS in 7m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e0cd1083a4554f78844e97825b082643 : SUCCESS in 12m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/936344f60dfc4afc8de632f030f8c6ad : SUCCESS in 29m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ff4197e30a5043ec9425b9c6214e7a03 : SUCCESS in 13m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/17f66f939ad2404bb6ec10c8274c753d : SUCCESS in 16m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3fc7046253e844ce819f8e2d802c994b : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05cee0f52f9740d0a8ef611327ca3201 : SUCCESS in 7m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/acc1373a82d94d5ca86467b2bb5005a3 : SUCCESS in 10m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9660a84a62974ddb8bad35b69cd8ba3e : SUCCESS in 10m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4aa0fb01fc04012b3d0914c4eef95f7 : SUCCESS in 1h 06m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f514974320f44c8598b8c56e708d0354 : SUCCESS in 1h 03m 44s","accounts_in_message":[],"_revision_number":3},{"id":"f499c90482bf971e900113ca45c725df4cbec846","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 19:34:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6ba944954b775ecd647e137dafe486cc9421e360","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 19:36:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e545fa20fd2459da7e90ffc0c3bbc7d : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"77e37d59bfb04af26dfc4066bc55da84bafdd741","revisions":{"3a6d9421a029d36e7183e6e63f9871c7db0b8d1a":{"kind":"REWORK","_number":1,"created":"2020-06-23 20:18:35.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/20/737620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/737620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/737620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/737620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/737620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/737620/1"}}},"commit":{"parents":[{"commit":"5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc","subject":"Switch node tests to inherit nodejs-run-test-browser.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-23 20:17:48.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-23 20:17:48.000000000","tz":-300},"subject":"Remove hiredis from global requirements","message":"Remove hiredis from global requirements\n\nThis package was only used in openstack/dragonflow. Now that that\nproject is being retired we can remove this from g-r.\n\nDepends-on: https://review.opendev.org/737564\n\nChange-Id: Ic9d23c9ba37b5ef68b315db3c14ed5ffe2211b3b\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a6d9421a029d36e7183e6e63f9871c7db0b8d1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3a6d9421a029d36e7183e6e63f9871c7db0b8d1a"}]},"branch":"refs/heads/master"},"4781108bff0e0ded0ee074fcf8798434c58f76e1":{"kind":"REWORK","_number":2,"created":"2020-07-20 16:26:45.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/20/737620/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/737620/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/737620/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/737620/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/737620/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/737620/2"}}},"commit":{"parents":[{"commit":"237770babb13510eb5aad4813667a0265205b620","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237770babb13510eb5aad4813667a0265205b620"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-23 20:17:48.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-20 16:26:39.000000000","tz":-300},"subject":"Remove hiredis from global requirements","message":"Remove hiredis from global requirements\n\nThis package was only used in openstack/dragonflow. Now that that\nproject is being retired we can remove this from g-r.\n\nDepends-on: https://review.opendev.org/737564\n\nChange-Id: Ic9d23c9ba37b5ef68b315db3c14ed5ffe2211b3b\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4781108bff0e0ded0ee074fcf8798434c58f76e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4781108bff0e0ded0ee074fcf8798434c58f76e1"}]},"branch":"refs/heads/master"},"77e37d59bfb04af26dfc4066bc55da84bafdd741":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-29 11:50:22.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/20/737620/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/737620/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/737620/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/737620/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/737620/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/737620/3"}}},"commit":{"parents":[{"commit":"8e268ff74e13e154db7568a560818f71b1ea8514","subject":"Merge \"Remove skydive-client from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e268ff74e13e154db7568a560818f71b1ea8514"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-23 20:17:48.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-29 11:50:22.000000000","tz":0},"subject":"Remove hiredis from global requirements","message":"Remove hiredis from global requirements\n\nThis package was only used in openstack/dragonflow. Now that that\nproject is being retired we can remove this from g-r.\n\nDepends-on: https://review.opendev.org/737564\n\nChange-Id: Ic9d23c9ba37b5ef68b315db3c14ed5ffe2211b3b\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77e37d59bfb04af26dfc4066bc55da84bafdd741"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77e37d59bfb04af26dfc4066bc55da84bafdd741"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
