)]}'
{"id":"openstack%2Frequirements~737640","triplet_id":"openstack%2Frequirements~master~I0cec05207de1e9eb8bd11fc2eb582099d08ce991","project":"openstack/requirements","branch":"master","topic":"notifier","hashtags":[],"change_id":"I0cec05207de1e9eb8bd11fc2eb582099d08ce991","subject":"Remove notifier from global requirements","status":"MERGED","created":"2020-06-23 21:37:40.000000000","updated":"2020-06-24 04:34:15.000000000","submitted":"2020-06-24 04:32:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737640-1592973162606-47516c0e","meta_rev_id":"a1e4bc27dbb5b536279c4593a307feaf00f9520a","_number":737640,"virtual_id_number":737640,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-24 04:32:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-24 02:59:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-06-24 02:59:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 02:59:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-24 04:32:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b44d688b4b555e26bd3da38665e73f4d2b7834dc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-23 21:37:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"169940232434426f49e6fa2387f440bed49ec388","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 21:54:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4fff9dbc3c194b439699e5220947afce : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":1},{"id":"663567ef35b050ba57d3e7c86be0c89827af62fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 01:50:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4e9f7d40bf8420c8d4caf2373a4b5c4 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01704f0ad43b4ec3b7090677b7457dae : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/634bd85069f34714830f7996418c2667 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c352e63e257b4923b4fa4b7387e9af3e : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32c2a2af30f84ba097b4f6f9d1b2d422 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b0ab7a9af41e4024b7cc4f23545e213c : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2e7a98a9efa84c91a28430e4d5194b98 : SUCCESS in 8m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b89cc3cc251544c39f16dd96c93664f1 : SUCCESS in 9m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4f006e4a22354039830a1c0fdf10cc1a : SUCCESS in 9m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4278ce766a05437a8a19094c1f84caca : SUCCESS in 4m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e9964f016b61461ea70db15985d18b02 : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6ac450faaca4397a86660e98e19f929 : SUCCESS in 6m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7bb046ad8360407195ae8f701bb29376 : SUCCESS in 16m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2c17beaa5cc54c8ea7703dca43037a33 : SUCCESS in 31m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8d5d47820ec8468f939ca39fcfa2fd00 : SUCCESS in 14m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ac5d5ae4887b4f649be651673d73e051 : SUCCESS in 19m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/45feb3f51e0a45148df83accd1152e8d : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b9225cd2b6244c591667acd5c087f24 : SUCCESS in 7m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/03805e6d182248ae80dddc282fb27a2a : SUCCESS in 8m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87ffbbe5dbb94bedad367235e5ca840a : SUCCESS in 40m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d5db77395d946d2a8b6337643954daa : SUCCESS in 1h 22m 39s","accounts_in_message":[],"_revision_number":1},{"id":"893f41482cb4b360ed3b2bb0c8929fb88d73c9c4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 02:59:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9c5a57d82d5be167395f244686bba57b8fad3e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 02:59:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7904b28bdd959bd826b2d2f812f98671d91516f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 04:32:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d904390a8b0c85b01515401d2edbd7237a3b7b4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 04:32:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/443a6e2af3d44f0d847bedd1566a77c5 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b1bd80794674eceb8809c93f841df62 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/841217d31bde4ca5af221779c4acd0bb : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b52f1a2ded1418b991d06c86950b961 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c030573a41924adca05978dcef1033ba : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d86a2f89d6074cf99686c6578e042bc9 : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d10ce5c810bd412bb6b5a230c2ff870d : SUCCESS in 8m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0c662542309c41de8ddd2e5d11364c7b : SUCCESS in 9m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8f9d93978e914dfeada910e48a19ccb1 : SUCCESS in 12m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7e4ff5acff9f41dc85cf46ac4ff10fba : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3234e05078f340c8b239a61d0cbe1ad8 : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5468762f70c3497fa133c1c9b411c59d : SUCCESS in 7m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bdb0e09c0ff943a9abb10d85a50196c8 : SUCCESS in 14m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/439ff62595b748b58baafcd8bc45c148 : SUCCESS in 25m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/67e9149236ce4b5c8e14d81b594a043f : SUCCESS in 13m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7aabcc0452cd4555b260c7d31eae1444 : SUCCESS in 19m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b82197a41b9a4430be2cd14fc77a225a : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/adb08a3500fc4cf0b9cdbe7cab7c3698 : SUCCESS in 6m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8456fb9f294a412baebe8093de8365b6 : SUCCESS in 9m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2c6b827e50f849a1a2fd65cc25955a74 : SUCCESS in 10m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b734f174238d45e397aa586fc5bf6401 : SUCCESS in 50m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6aafc87da13d440889dfdb92319f7845 : SUCCESS in 1h 27m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3529530816a755e6f2b117e78d4ce7b860a572e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 04:34:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d26cca70fa9646ccaa3fba2db9fbdc2c : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b4c60526f27d1973f5bcf8ae92fe9aac1d8d761","revisions":{"1b4c60526f27d1973f5bcf8ae92fe9aac1d8d761":{"kind":"REWORK","_number":1,"created":"2020-06-23 21:37:40.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/40/737640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/737640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/737640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/737640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/737640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/737640/1"}}},"commit":{"parents":[{"commit":"0559be261ee86bdef6a97c802f53a263057b844a","subject":"Update constraint for neutron-lib to new release 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0559be261ee86bdef6a97c802f53a263057b844a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-23 21:36:57.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-23 21:36:57.000000000","tz":-300},"subject":"Remove notifier from global requirements","message":"Remove notifier from global requirements\n\nThis package isn\u0027t actually used by any openstack projects. Let\u0027s clean\nit up.\n\nChange-Id: I0cec05207de1e9eb8bd11fc2eb582099d08ce991\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b4c60526f27d1973f5bcf8ae92fe9aac1d8d761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b4c60526f27d1973f5bcf8ae92fe9aac1d8d761"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
