)]}'
{"id":"openstack%2Frequirements~737641","triplet_id":"openstack%2Frequirements~master~I93cc0e6490d38b547f9e707750d3ced1d9e8774c","project":"openstack/requirements","branch":"master","topic":"pathlib2","hashtags":[],"change_id":"I93cc0e6490d38b547f9e707750d3ced1d9e8774c","subject":"Remove pathlib2 from global requirements","status":"MERGED","created":"2020-06-23 21:42:04.000000000","updated":"2020-06-24 04:32:44.000000000","submitted":"2020-06-24 04:32:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737641-1592973164128-cf4bca08","meta_rev_id":"27421409d533070d091699593f2804c440b852f7","_number":737641,"virtual_id_number":737641,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-24 04:32:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-24 02:59:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-06-24 02:59:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 02:59:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-24 04:32:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3163ec42ff280cffc9630e980e58506261c62967","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-23 21:42:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6658d84149794d0f229293f8213ba2e741e24efe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 22:01:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/755068e6afaf4f42b9b3ad4b064cbd64 : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"23786e20d506d2d907c3ff1e4351727e0317cd73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 02:12:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3da95dbb67b47adae4ae523f854318b : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92155f7e99494d2aa88b3e71d0b61ca8 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fbfba923af84cefa645b202c4f4be82 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63c724b275ee4f90a79cf9625c680e35 : SUCCESS in 5m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3285980b27cb4bd0a4d99a468ff6324d : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4f04b903049141c8b163220d1eb21f05 : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/00e93793f6284acabb338f93c1ddc726 : SUCCESS in 8m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/40f50d0b47ef468bb768a4fa54353672 : SUCCESS in 9m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/922c59ef16f7478f8a7fdf3b666803f4 : SUCCESS in 9m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3f2bfd7e11a34fc6b38dc69011bfcc7f : SUCCESS in 5m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fec1898ef25745ff947c4febba935b60 : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b9ff9d109c334c9fb390e34c4b534565 : SUCCESS in 8m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/79808906261348d2a071a66fe029e925 : SUCCESS in 12m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/42a8b10e915f4f428349219b0b0e0c61 : SUCCESS in 26m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ee2168ca2d594419884b53bd47688b20 : SUCCESS in 13m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9cf2dd4891304165bf195ddf7f241ceb : SUCCESS in 16m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f49df493d0a7460fa0c5978ed4fa0418 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/116d77957b0143c29d82b656d5014ad9 : SUCCESS in 9m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e409c8f1396246d18ec11817c92673eb : SUCCESS in 12m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c288a85baedd47fa8ee7b0464bd9ac15 : SUCCESS in 1h 00m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec5ee9e8a0374ee8851f59f461a4a6c0 : SUCCESS in 1h 20m 28s","accounts_in_message":[],"_revision_number":1},{"id":"bfd371f2fab8f454d84d6e7cb88fe82c4abbdd7a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 02:59:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"880a780d8336a5f0295c00a45844056c3bddf6cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 02:59:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a1a46fd4fa762e427de7c6b278d339178cb24e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 04:32:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/287d4ef320fd4daf9a26d6cfb5720304 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/688bb879b038421dbb54f5b988bcf734 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80a914b13d7a44288456523aa22ca3af : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/902b2837c55c408aa2803095869b5ca7 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4826e8bb3b943009a7dde04cd80df02 : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17588e5923ad4666bd0af25153eac5c6 : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9c529e6a83c2405fa873a1bbb71db5a2 : SUCCESS in 9m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/187948055f3f4fce8d3004390c91f315 : SUCCESS in 9m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dfa8dae1ab9c4a118513062e89bae734 : SUCCESS in 12m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5fec0bb289604682a998c3b3e795256a : SUCCESS in 5m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4b6927d22ee843d1993f5e302e2407b5 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f6b0e48c6354836a1e6d48410e870cf : SUCCESS in 7m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/786b7814186f45879f181be591b1feda : SUCCESS in 19m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2b80282dab8149c490357e62ff630771 : SUCCESS in 24m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c463d8db0fe54391b88ed930e968eba0 : SUCCESS in 12m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/08f96e307b9040d19ff7059b12ad5912 : SUCCESS in 16m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0f2abc9c669546e98ec8df89cee33683 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/efab4bff011044c2a4edd54f3393ec7d : SUCCESS in 6m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fa666655d6b2489c94ef6e22668eca6e : SUCCESS in 9m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4532caf0564f4f3a8f41406db6371d07 : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97a1a532c8ba4e45ba2275862819f8fd : SUCCESS in 52m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/933bba907cc14ffd890cb8c84ce31c75 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":1},{"id":"7495a786db39612dc079614885f7315e67634123","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 04:32:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03820f41c474106352b406a4e5e7fbe66506c845","revisions":{"03820f41c474106352b406a4e5e7fbe66506c845":{"kind":"REWORK","_number":1,"created":"2020-06-23 21:42:04.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/41/737641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/737641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/737641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/737641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/737641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/737641/1"}}},"commit":{"parents":[{"commit":"0559be261ee86bdef6a97c802f53a263057b844a","subject":"Update constraint for neutron-lib to new release 2.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0559be261ee86bdef6a97c802f53a263057b844a"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-23 21:41:26.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-23 21:41:26.000000000","tz":-300},"subject":"Remove pathlib2 from global requirements","message":"Remove pathlib2 from global requirements\n\nThis is only referenced in some files in upstream-institute.\n\nChange-Id: I93cc0e6490d38b547f9e707750d3ced1d9e8774c\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03820f41c474106352b406a4e5e7fbe66506c845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03820f41c474106352b406a4e5e7fbe66506c845"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
