)]}'
{"id":"openstack%2Frequirements~737873","triplet_id":"openstack%2Frequirements~stable%2Fussuri~Iffa3f7631bec0a5f7b0c701df87305725cae5708","project":"openstack/requirements","branch":"stable/ussuri","topic":"cross-horizon-npm-stable/ussuri","hashtags":[],"change_id":"Iffa3f7631bec0a5f7b0c701df87305725cae5708","subject":"Switch node tests to inherit nodejs-run-test-browser.","status":"MERGED","created":"2020-06-24 20:24:04.000000000","updated":"2020-06-25 08:55:12.000000000","submitted":"2020-06-25 08:55:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737873-1593075312931-dd5307e8","meta_rev_id":"3e9782489a5727b4e930f4dc635c5f5191bef09f","_number":737873,"virtual_id_number":737873,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 08:55:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-24 23:44:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-06-25 00:50:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 00:50:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-25 08:55:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12309363da375ef992adf3832b478e23a9fc3fbc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 20:24:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c6a404e1830ec66ffb011643095078a73a22781a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 23:25:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9398dacabe3b4e628affd08f2ec403bf : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df16f7cd273a47678661608748bc4982 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23d918aa33544afd90fb6f61dee4caa8 : SUCCESS in 6m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/088efc2a2e8549828f9b26f8bf61abc8 : SUCCESS in 5m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1440b7082dda4778aa885a19d8fe8d7e : SUCCESS in 8m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/44376293182a42e8a1ddc4098d5e4b46 : SUCCESS in 6m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/917f390a71e24be7a2198a09a893a779 : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c3d1635fbaf94cbc80aaab6d958d3789 : SUCCESS in 8m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0e88325ece89477d98136e27dd47667e : SUCCESS in 15m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6bb2cd3200374dbca86735d6ca12a216 : SUCCESS in 26m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8ab3652dec9f42f2907652c0a6a96cdd : SUCCESS in 14m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8a39344a90f245b1ac39066b7c789531 : SUCCESS in 16m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f00e93ce1a8b454bbc0c8464a152c0eb : SUCCESS in 17m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9236e565e11343f4a1d144fd4b850d6c : SUCCESS in 7m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ca9f58b13fb14b24b1c16d1d2c842eb2 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5ae7354e7a3f476d8d733ae7bda8e77b : SUCCESS in 9m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0e0b7e6ff3da4cc09af188ce23e35a6d : SUCCESS in 10m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/10c67f6d14ff42f1a307b43161b66b02 : SUCCESS in 9m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1132221786e4e2ea30b1d4bcc106015 : SUCCESS in 52m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/efd670a6a8034820b8fde6c7643c64fd : SUCCESS in 1h 17m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a7ec2a5cf034c1b888c199b1d02ffd5 : SUCCESS in 1h 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/4088de6e7b8d4b20982395e2bf58960b : SUCCESS in 1h 09m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0d32d2b4d2c311e71cb10361c65dbe5594b27174","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-24 23:44:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ecb6988595414a3146399e1cc8c3d52c673aa790","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-25 00:50:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0c9c5941dc5696b6fb3cde2f80fa254e0fc0f74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:51:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f610eac21a4d39dbfc9ec78f83f2ab298809211","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:55:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/199c53635f644bb48336b494239129d8 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f94f1359294a47b68737b39c1b91c6ab : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/225c70ab1ea74494b5b10f2dcc1d01e3 : SUCCESS in 7m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1031b26124e544d48d19fdac07a610d7 : SUCCESS in 6m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/badcabf0c7cf4acba3d1f0a457fd49fb : SUCCESS in 16m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/20aec9b10bfa4435b04c5523263dcdfe : SUCCESS in 9m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fa38e84b262a4d4daacdda285ffd42ee : SUCCESS in 16m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b92d2012327a4109ab55cb95a68d4d3f : SUCCESS in 11m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ccf7d52eb6834d9b83724f8c2cce7f6e : SUCCESS in 18m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1905c65e195a44d89eb1dcb953a97244 : SUCCESS in 32m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/05d6614cf50f4a3b95819588489b949f : SUCCESS in 20m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4daa14e75e374dbfb4468c2d03e4e23e : SUCCESS in 8m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1c72ba4f066a4eafa2b997955c43b6d2 : SUCCESS in 9m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/69d50b7cde12446eba145581b6453576 : SUCCESS in 10m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b7581d1105664abca15e44ece9bb30b3 : SUCCESS in 12m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b4508173a7ab4843b228117a44d90804 : SUCCESS in 12m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13c1d5decfc04c689103efddc08e78ac : SUCCESS in 1h 03m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/29846f41259f42a58d3dd37851769cf3 : SUCCESS in 1h 21m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb342757ff7b4e10a6c4ea93d37c9993 : SUCCESS in 1h 21m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/65978df9fd2b4afd9adb9646ff2d0225 : SUCCESS in 1h 18m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8119e20f17e072470a033a748ffa284a60435769","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:55:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ece58b25ddb3754bba744e76ed17aaafae144030","revisions":{"ece58b25ddb3754bba744e76ed17aaafae144030":{"kind":"REWORK","_number":1,"created":"2020-06-24 20:24:04.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/73/737873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/737873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/737873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/737873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/737873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/737873/1"}}},"commit":{"parents":[{"commit":"d303fe350d74d2d40aa5a591ee9d372a529d10be","subject":"Ensure pip is installed locally","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d303fe350d74d2d40aa5a591ee9d372a529d10be"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-22 18:33:38.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-24 20:24:04.000000000","tz":0},"subject":"Switch node tests to inherit nodejs-run-test-browser.","message":"Switch node tests to inherit nodejs-run-test-browser.\n\nCurrent base of nodjs-npm-run-test is deprecated. We need to switch to\nthe newer base job to pick up new behavior. Horizon npm tests require\na browser so the job inherit nodejs-run-test-browser now.\n\nIn addition, tox is not anymore pre-installed on the OpenDev CI images.\nnpm jobs in horizon need tox, so we need to install it in pre-run phase.\n\nChange-Id: Iffa3f7631bec0a5f7b0c701df87305725cae5708\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nCo-Authored-By: Akihiro Motoki \u003camotoki@gmail.com\u003e\n(cherry picked from commit 5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ece58b25ddb3754bba744e76ed17aaafae144030"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ece58b25ddb3754bba744e76ed17aaafae144030"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
