)]}'
{"id":"openstack%2Frequirements~737875","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Iffa3f7631bec0a5f7b0c701df87305725cae5708","project":"openstack/requirements","branch":"stable/train","hashtags":[],"change_id":"Iffa3f7631bec0a5f7b0c701df87305725cae5708","subject":"Switch node tests to inherit nodejs-run-test-browser.","status":"MERGED","created":"2020-06-24 20:29:23.000000000","updated":"2020-06-25 08:26:05.000000000","submitted":"2020-06-25 08:26:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737875-1593073565004-1a3eceda","meta_rev_id":"0a05b42a8b7c5f19b5329bc92380ad3779a02e70","_number":737875,"virtual_id_number":737875,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 08:26:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-25 00:19:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-06-25 00:19:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 00:19:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-25 08:26:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6a3af159e5e45e3f0aa75f8f69cf6ecdbcbb6d9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-24 20:29:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55aa509ea07813526f47a1c6fa151f42cd0de8d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 23:56:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5963ab827f8a4baebf6758b1a1c6e847 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b928b6c4b1c34a66ab402fae1552d4bb : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c535662b64d43ab83abfc73cfc14ecb : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe280a8a44744b21b4c6978c96bca1a8 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d702d9dfc754c92b4e87f2ae60e08f0 : SUCCESS in 4m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b3687c3419fe47e8a194b51ca8fa99fb : SUCCESS in 13m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f352457b896a4ea0a477de4d0b49c54e : SUCCESS in 12m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/52091779367b4d2cb81839912a45ce76 : SUCCESS in 5m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/738f9050ea8046d28c9021cf6e2030d5 : SUCCESS in 5m 40s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9d1082956d28406e90648e52f2ff94e3 : SUCCESS in 8m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a03d1d55e2914f97b468cb3566e39636 : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2d006b134ebb47c8b4d99dd0117ac764 : SUCCESS in 11m 13s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5ddf1151d743465d8608690b401367f2 : SUCCESS in 13m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/88c713e7af534c42a257dfb10c69bbf8 : SUCCESS in 13m 22s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d289665d33e144fe9f58f0bf97912716 : SUCCESS in 23m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cdd1d611b5b7464ea3317c0b0e7df8c8 : SUCCESS in 22m 58s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/43b787a58e5a4497b1d59ae034da9c2a : SUCCESS in 13m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/225f6f2739e344528130f98bf9783fed : SUCCESS in 15m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fca2737f0da941559f80e7791b9dc697 : SUCCESS in 17m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/63629aedde7f489d8dee690b93443e9c : SUCCESS in 20m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/956f3afed0774d6d90edc85aee16df71 : SUCCESS in 8m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/025536cbc69844c8be3420e53bd72d43 : SUCCESS in 11m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/19de6944ca0041fe88212ffd78c2a2b6 : SUCCESS in 9m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f436211ffcd43eb85bda92ee891d0ec : SUCCESS in 1h 10m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e71ba7d6e7bc4bf3b71b95a355b3cea5 : SUCCESS in 1h 34m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/88ebfabcc8a34754914d0ce7d5522134 : SUCCESS in 56m 08s","accounts_in_message":[],"_revision_number":1},{"id":"828d156f8ef77e9bc1e72323d78452a6d2a5a9ea","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-25 00:19:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b8a53b4a77dcd41c37e46ef15a2dcaa5eda885a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 00:19:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc3de7cff34e1b0de2e00cb7f4a98af36edbcd55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:26:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dcd8baa7c714c34bf197675582235d7 : SUCCESS in 6m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8a9a63cf0df949aaa463ff3311c0d181 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36c46947802043e4bd467ba4e06ee66e : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1035c2c7e3c4f3094dce6532b34c5a9 : SUCCESS in 7m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/848f5cbd83584aa19c7ab5d6c5b54b07 : SUCCESS in 7m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/36971852061146acb9f3c1f64dbcd2a1 : SUCCESS in 12m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0f4ed472b73b4e80857ccd95747f22ab : SUCCESS in 11m 55s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/96f7c6b76265470c96105c607c1aeee5 : SUCCESS in 6m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c88dd71e28d5474abcfbef4332be5bf5 : SUCCESS in 8m 41s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/656cd6857701435b977b591944867e26 : SUCCESS in 12m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/db5afad7adff4b60930c7d264669542b : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ecb8e593c6ef4e42b8f950ef995c2c86 : SUCCESS in 8m 49s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e403bdad2b724e0b9c0530684e7d114c : SUCCESS in 20m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0233adfdcb9f4106bd02646121d7b8fb : SUCCESS in 14m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4bb1dd1c72804dccb1f565b5a331e984 : SUCCESS in 23m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9ac1bb1e8e39416db21f7fd62838700a : SUCCESS in 24m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7efb48d15a9245c3aec93e4d325f4f6b : SUCCESS in 19m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a8ff54d9a848415da601cb0425148c30 : SUCCESS in 23m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84d2cdf7cdc94c049382fe914d3ed51e : SUCCESS in 8m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7855d877d8da4f3cbcb91801a4b5f847 : SUCCESS in 12m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0108aa211c3447298ff516dfb678508d : SUCCESS in 10m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1767d9dc674c4490bbc31aa108c8faaa : SUCCESS in 1h 11m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2edf6147c5744c2c92042468f7e64ec7 : SUCCESS in 1h 31m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b823896a66d645158ad07ed3c583fd5e : SUCCESS in 1h 04m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0a05b42a8b7c5f19b5329bc92380ad3779a02e70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 08:26:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5996ae1bd1fa68b8985b49c4a07380d9b723dd4b","revisions":{"5996ae1bd1fa68b8985b49c4a07380d9b723dd4b":{"kind":"REWORK","_number":1,"created":"2020-06-24 20:29:23.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/75/737875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/737875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/737875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/737875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/737875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/737875/1"}}},"commit":{"parents":[{"commit":"d43e0a585719930516fd82504d84e3867512c7f6","subject":"Ensure pip is installed locally","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d43e0a585719930516fd82504d84e3867512c7f6"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-22 18:33:38.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-24 20:29:16.000000000","tz":-300},"subject":"Switch node tests to inherit nodejs-run-test-browser.","message":"Switch node tests to inherit nodejs-run-test-browser.\n\nCurrent base of nodjs-npm-run-test is deprecated. We need to switch to\nthe newer base job to pick up new behavior. Horizon npm tests require\na browser so the job inherit nodejs-run-test-browser now.\n\nIn addition, tox is not anymore pre-installed on the OpenDev CI images.\nnpm jobs in horizon need tox, so we need to install it in pre-run phase.\n\nChange-Id: Iffa3f7631bec0a5f7b0c701df87305725cae5708\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nCo-Authored-By: Akihiro Motoki \u003camotoki@gmail.com\u003e\n(cherry picked from commit 5b67b111ca3dd5b45caf1ef8f2ae8070487a00cc)\n(cherry picked from commit ece58b25ddb3754bba744e76ed17aaafae144030)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5996ae1bd1fa68b8985b49c4a07380d9b723dd4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5996ae1bd1fa68b8985b49c4a07380d9b723dd4b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
