)]}'
{"id":"openstack%2Frequirements~738045","triplet_id":"openstack%2Frequirements~master~I890982ffb106305a303821d9210a37f6d8914cda","project":"openstack/requirements","branch":"master","topic":"complete-retirement-networking-l2gw","hashtags":[],"change_id":"I890982ffb106305a303821d9210a37f6d8914cda","subject":"Remove the networking-l2gw deliverables from project list","status":"MERGED","created":"2020-06-25 14:35:22.000000000","updated":"2020-09-01 17:50:14.000000000","submitted":"2020-09-01 17:47:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738045-1598982459377-5b9a2423","meta_rev_id":"f6ffe206c3fd96afb3c41af52533c5c84a380d0d","_number":738045,"virtual_id_number":738045,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 17:47:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-01 16:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-08-31 12:19:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-08-31 08:35:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-09-01 16:01:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 02:09:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-26 15:19:10.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-08-31 08:35:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-31 12:19:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-09-01 16:01:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-01 17:47:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59edbfd269002c9d074b32bc111d1ab59ffe96e0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-25 14:35:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ae7fc7833cd25295e8dbda47548adab6578b19e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 14:35:56.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"111acb11307cc4e7a3a3b94af5d49dc31c50b454","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 14:36:38.000000000","message":"Patch Set 1:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"7909ccdb64693718994f5fd66934ac71e4d130b2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-25 18:03:34.000000000","message":"Patch Set 1:\n\nyou don\u0027t need the dependency, this can merge directly","accounts_in_message":[],"_revision_number":1},{"id":"c66609c5bc04959268ef77ecef6e7e1415152765","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-25 18:04:18.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f201dda9f94564ada38fabed805ca00cde3cb090","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-25 20:24:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5a0b8a5d52d9aa2ef3e0df019a23699d1bb38bb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 01:13:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2031b889f166413895eb348c239a2316 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53fa236bcfa748c3a9006b8edbe725e9 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cbde4c3222349efb8cefa6a1bddc661 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1a316056f124fe98538318f57edb3ae : SUCCESS in 6m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffba5663b5d147ee8c3bf930ea5cab0b : SUCCESS in 6m 07s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/5efc1b983885420db440a84a4e68d451 : FAILURE in 5m 44s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f983f89373aa448292004f7f5f81fd12 : SUCCESS in 1h 02m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17dd473b425b43ea991e8793ec6a3550 : SUCCESS in 1h 30m 17s","accounts_in_message":[],"_revision_number":3},{"id":"3697a98f26c281946d815b66e7b8d43906df8fe6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-26 02:09:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8563c609e4e0b5e11b52a110648896431d250162","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-26 02:09:27.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"cf1092a7b6faa9bdd676d254b918ac632042fdf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 02:09:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cd34093a24ecd0a0fb92b86d0fe0bd29ad7000c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 05:48:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ab220ef35034580be75f456b71241cb : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c7918dd18204ca6ac21a4e0e91cffcf : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a46df46013ba4eabb92aef4229c0adf1 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1600ab549e76453187740756687d3c6c : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e3f14e99c384f0c9e86ce6fdba45732 : SUCCESS in 4m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ef679a1752f48c6aa68d764e3fb307e : SUCCESS in 55m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32e75be3b9b3497dbb963da29a38d4bb : SUCCESS in 1h 23m 10s","accounts_in_message":[],"_revision_number":3},{"id":"b0b94c2cd85fbd5f49a35a4c0465bb18412f2dba","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-26 07:25:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"46a2371dc6dc4916ae8458f460dfa93b8a43bc4d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-26 15:08:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"94ee19be21936401121835f956335651e66aabb6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-26 15:19:10.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\ntake my -1 as WIP for now","accounts_in_message":[],"_revision_number":3},{"id":"a089cae6564051fbbbbf845bd2e3b7d57b928185","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-26 15:24:02.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f2d496c7d4f5b3564b47c4787710ee67c26b5264","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-11 14:04:04.000000000","message":"Patch Set 3: Code-Review+1\n\nbased on https://etherpad.opendev.org/p/tc-retirement-cleanup (~l35.)\nwe can move with this","accounts_in_message":[],"_revision_number":3},{"id":"e5edc2ba2f6be626261694be78f84c62d960d670","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-15 18:09:26.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nlet\u0027s revert the one line and then merge this...","accounts_in_message":[],"_revision_number":3},{"id":"94626029b5c666b9926d3993019d56affed698d3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-16 01:04:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6f8425be216a1ff62f554531fc2a42cb37351b46","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-27 22:37:55.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"992ec871c6d2ccbb45d2f89443801a4bc7d08501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 23:44:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45e3eed41c5142c58ab6a4fe8715b0e5 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9de9472252be41b48fb3304f063cfa30 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/347e90649eb6497ea72d0b8ccc11416e : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3840b3b5f45d4d9fa550a30c59faf0b8 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3b332a3535e1482385f0251a6e673dbc : FAILURE in 4m 22s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/1b4f9fa4826e4957953b7b4a3c2ac9c5 : FAILURE in 3m 58s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/744b5a7ef6ff4bdfaea57fdddc53e787 : SUCCESS in 1h 05m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/898a91958e314ba6ab2685686b6d1bfa : SUCCESS in 52m 18s","accounts_in_message":[],"_revision_number":4},{"id":"7073c417d4b6d6bb8d7bfdc96e1cf7127a2af25b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-28 06:52:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9543697be6a420d68cbc4c48c5e9280e54cd69d4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-28 18:46:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6dbfd226e1e5e18fbd9a12c1165cffd06524fbef","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-28 18:47:34.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"e1c949c87d6a323939d2baa9eb2219e0a5da2a8f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-28 18:50:20.000000000","message":"Patch Set 5: Code-Review+1\n\nWaiting for CI","accounts_in_message":[],"_revision_number":5},{"id":"4d85c198f45212953c91ae70684558802dc65559","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 21:21:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/262b0d8beed44c57add11bc9917ffde2 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd681708be184cc1bdf4bf0c7c3f1a29 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/064b122087e14233bd51256d34c4f3c0 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a576e5cbcd2149b887fa8bfe47ef0a6e : SUCCESS in 10m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15bad9674aeb4fe685f0cbc20c18c563 : SUCCESS in 3m 55s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/1f905d1d421643ed938e71725554af02 : FAILURE in 7m 20s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fce47cb4cb94add9f3a0e5702e120a4 : SUCCESS in 56m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c4431c9d68d4da5bc554686a375dd35 : SUCCESS in 1h 16m 09s","accounts_in_message":[],"_revision_number":5},{"id":"6a245106b3b4a996372aadca13798a349fd4c5f2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-31 08:35:31.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6c3d1b60923e632f6c30639edd13ef7b5c0f9eed","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-31 12:19:26.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"552e14b8f201c1644d3465ace95eef35a71ecbd6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-01 16:01:11.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4d32649d14877148806b237ee21f949b80691742","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 16:01:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bb891a223f1505f1a7d61c5ae017f63d7321823a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 17:47:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"943513a653c7c76c491cbea6f206a353e7d3fbf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 17:47:39.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/538682717c5c4e3883f7c719a2fb467a : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc9968896b9c40378858e66fcb70de59 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/763246cb999d4221b83d08b9f9fd8be1 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80a186aabe0a436c881be687dc56a00e : SUCCESS in 9m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/829e5e896e934d619c3a7912353df9e5 : SUCCESS in 7m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46903f75f242491f96e9c749f9419fb9 : SUCCESS in 49m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c233bee9e1e4449e82f819c27c8d001b : SUCCESS in 1h 37m 07s","accounts_in_message":[],"_revision_number":5},{"id":"a01c7ede6a2c062f89406366e8af772e9ffa7add","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 17:50:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dfa8b529f994243bb41b62b6b84bd84 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d6c683de8a0a025d6a770a5250fe4f9efbac19bd","revisions":{"7e0b9f9b8567b8b9cc2f7ff3253efdd861a81122":{"kind":"REWORK","_number":1,"created":"2020-06-25 14:35:22.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/738045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/738045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/738045/1"}}},"commit":{"parents":[{"commit":"feef36f5b642804c3e9bbd44254dd30f67bcf84d","subject":"Merge \"Add an ironic cross-job with py38\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/feef36f5b642804c3e9bbd44254dd30f67bcf84d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:32:40.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:35:21.000000000","tz":-300},"subject":"Remove the networking-l2gw deliverables from req","message":"Remove the networking-l2gw deliverables from req\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nfinishing the retirement process and no longer will be\nmaitained under openstack[1].\n\nDepends-On: https://review.opendev.org/#/c/738035/\nDepends-On: https://review.opendev.org/#/c/738038/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I890982ffb106305a303821d9210a37f6d8914cda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e0b9f9b8567b8b9cc2f7ff3253efdd861a81122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e0b9f9b8567b8b9cc2f7ff3253efdd861a81122"}]},"branch":"refs/heads/master"},"5551ffdea35804817a50f56f47dfb670a65eb525":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-25 18:04:18.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/45/738045/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/738045/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/738045/2"}}},"commit":{"parents":[{"commit":"feef36f5b642804c3e9bbd44254dd30f67bcf84d","subject":"Merge \"Add an ironic cross-job with py38\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/feef36f5b642804c3e9bbd44254dd30f67bcf84d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:32:40.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-06-25 18:04:08.000000000","tz":0},"subject":"Remove the networking-l2gw deliverables from req","message":"Remove the networking-l2gw deliverables from req\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nfinishing the retirement process and no longer will be\nmaitained under openstack[1].\n\nSee also:\nhttps://review.opendev.org/#/c/738035/\nhttps://review.opendev.org/#/c/738038/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I890982ffb106305a303821d9210a37f6d8914cda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5551ffdea35804817a50f56f47dfb670a65eb525"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5551ffdea35804817a50f56f47dfb670a65eb525"}]},"branch":"refs/heads/master"},"df245cc9178bde1df60ee82bf0e7680bfaeb2fb0":{"kind":"REWORK","_number":3,"created":"2020-06-25 20:24:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/738045/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/738045/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/738045/3"}}},"commit":{"parents":[{"commit":"feef36f5b642804c3e9bbd44254dd30f67bcf84d","subject":"Merge \"Add an ironic cross-job with py38\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/feef36f5b642804c3e9bbd44254dd30f67bcf84d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:32:40.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 20:24:17.000000000","tz":-300},"subject":"Remove the networking-l2gw deliverables from req","message":"Remove the networking-l2gw deliverables from req\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nfinishing the retirement process and no longer will be\nmaitained under openstack[1].\n\nSee also:\nhttps://review.opendev.org/#/c/738035/\nhttps://review.opendev.org/#/c/738038/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I890982ffb106305a303821d9210a37f6d8914cda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df245cc9178bde1df60ee82bf0e7680bfaeb2fb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df245cc9178bde1df60ee82bf0e7680bfaeb2fb0"}]},"branch":"refs/heads/master"},"56e025f5b52522fe481473ed713f21e731824de4":{"kind":"REWORK","_number":4,"created":"2020-08-27 22:37:55.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/738045/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/738045/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/738045/4"}}},"commit":{"parents":[{"commit":"feef36f5b642804c3e9bbd44254dd30f67bcf84d","subject":"Merge \"Add an ironic cross-job with py38\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/feef36f5b642804c3e9bbd44254dd30f67bcf84d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:32:40.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-27 22:37:53.000000000","tz":0},"subject":"Remove the networking-l2gw deliverables from project list","message":"Remove the networking-l2gw deliverables from project list\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nfinishing the retirement process and no longer will be\nmaitained under openstack[1].\n\nBoth repo are in x/naemspace now so keeping it in g-r.\n\nSee also:\nhttps://review.opendev.org/#/c/738035/\nhttps://review.opendev.org/#/c/738038/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I890982ffb106305a303821d9210a37f6d8914cda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56e025f5b52522fe481473ed713f21e731824de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56e025f5b52522fe481473ed713f21e731824de4"}]},"branch":"refs/heads/master"},"d6c683de8a0a025d6a770a5250fe4f9efbac19bd":{"kind":"REWORK","_number":5,"created":"2020-08-28 18:47:34.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/45/738045/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/738045/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/738045/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/738045/5"}}},"commit":{"parents":[{"commit":"feef36f5b642804c3e9bbd44254dd30f67bcf84d","subject":"Merge \"Add an ironic cross-job with py38\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/feef36f5b642804c3e9bbd44254dd30f67bcf84d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:32:40.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-28 18:47:32.000000000","tz":0},"subject":"Remove the networking-l2gw deliverables from project list","message":"Remove the networking-l2gw deliverables from project list\n\nnetworking-l2gw and networking-l2gw-tempest-plugin are\nfinishing the retirement process and no longer will be\nmaitained under openstack[1].\n\nBoth repo are in x/namespace now so keeping it in g-r.\n\nSee also:\nhttps://review.opendev.org/#/c/744110/\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nChange-Id: I890982ffb106305a303821d9210a37f6d8914cda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6c683de8a0a025d6a770a5250fe4f9efbac19bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6c683de8a0a025d6a770a5250fe4f9efbac19bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
