)]}'
{"id":"openstack%2Frequirements~738476","triplet_id":"openstack%2Frequirements~master~Ic475dc692971da7ad0c2dc4decb3a580620fdf98","project":"openstack/requirements","branch":"master","topic":"requests","hashtags":[],"change_id":"Ic475dc692971da7ad0c2dc4decb3a580620fdf98","subject":"Lower upper-constraints for requests","status":"MERGED","created":"2020-06-29 16:35:52.000000000","updated":"2020-06-30 00:08:04.000000000","submitted":"2020-06-30 00:06:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738476-1593475603416-a5e95a55","meta_rev_id":"1a99228ff1206da804ec47d1c40ac96c152fe245","_number":738476,"virtual_id_number":738476,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-30 00:06:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-29 21:22:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-29 21:11:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":1,"date":"2020-06-29 20:47:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-06-29 21:22:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 16:38:08.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-06-29 20:47:47.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-06-29 21:11:13.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2020-06-29 21:22:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-30 00:06:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dfd488766f24219ff3a174cbfdb9f8eabaf64d6e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-29 16:35:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94d16563f7a873e12c17ca3eb5b5bcaf47e85131","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-06-29 16:38:08.000000000","message":"Patch Set 1: Code-Review+1\n\nSee for example https://bugs.launchpad.net/cinder/+bug/1885526","accounts_in_message":[],"_revision_number":1},{"id":"b83250b329009fc38614b5b630dd72b3d4aea34e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 17:25:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6a4578d2a06547cc9d3e05a6b7c7ae72 : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d588510ea263d89b55d2fc4cbee3d5fc2f267f07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 18:17:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01a1b5fba283475bbefb5de3831dbc59 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae07c80c3f6d4b2cbaa60ceb5262573a : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7577b549a68d4eaabb8003f92b1dbaf6 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0e510130fb14822b22652e5e41df7b4 : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e09aac39acc4b5a9392a4139bc1b8b8 : SUCCESS in 4m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d7e08e8d0d62451a9dfcd5a017178874 : SUCCESS in 8m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/70cef2ff90d345d29f024d0f3262da07 : SUCCESS in 9m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/56d29deb71eb4217aee36f8cdb0ef838 : SUCCESS in 10m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/094df01fc00c47b2ad92119a930bd853 : SUCCESS in 11m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c9081d92e1c844da88cc137fab01f7d7 : SUCCESS in 6m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0b13cf5a9b99446ca3551d9b3915b05b : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2b5f3b6e1c1549d6ab59a1b1bbb9b72a : SUCCESS in 7m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/669c5b48116e4890a17a604946b901d5 : SUCCESS in 8m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0a91a09bef19498aa73d5f8c12e74bcb : SUCCESS in 13m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/136ee804e7cd427fa1ad55f1bcac00a6 : SUCCESS in 29m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d90b7da81c9148528ebc1c8d70065fbb : SUCCESS in 13m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2d47c32037ae4d4c917111d68f3dc9eb : SUCCESS in 19m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c3ac03b0dbd34ab985baa5188ee5e02e : SUCCESS in 6m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c5b84c09059a42f69fb5316b03029a1d : SUCCESS in 6m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/77eba3fe4aec42e28ee52226d013408a : SUCCESS in 10m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72011e28322e43c098e7b84dd48d6c00 : SUCCESS in 56m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01ad62bbf3fc464aac5d15fb05d6d00f : SUCCESS in 1h 35m 23s","accounts_in_message":[],"_revision_number":1},{"id":"cc33d83656bebdafc64eec53686df9b147009792","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-29 18:20:52.000000000","message":"Patch Set 1:\n\nplease block the bad version in global-requirements as well","accounts_in_message":[],"_revision_number":1},{"id":"437353bd5a0a38b5313bec228acacdfa37464f2a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-29 18:23:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"47a05d0bc7f0be363f5bd20f83bc091e76be7cd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 19:07:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ea2ff8b59c07477aa4f68235471a6cd2 : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":2},{"id":"10caf6627f56d944e7a04a184e873639eec80f61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 19:39:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ea51fc151e94c91a2e76cd18ead0e6c : SUCCESS in 8m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/817c0e0a7a944bf19adefaae5fdc9ee3 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37b9ea5405424081b35be7271b2d635d : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a256b0c5305403d91fae5a6acf6a523 : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a75c62b62be4ab8afc2a25eab9aa194 : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cda0060322734777bcf365f8b6039ff0 : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2d343a4325d343e3a80d0e33fdea6544 : SUCCESS in 8m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/41cf76ceafcf4c1c99d9c8f3654ddce4 : SUCCESS in 10m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/850f520def6d467ab636245b5694f213 : SUCCESS in 9m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/76f9799644ae476b9a5c18ecab6b2e60 : SUCCESS in 8m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7e834b44902b4cfd88e38d2fdffd93ca : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/69d6ed548c324a6f8dd2ce315999aca1 : SUCCESS in 12m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/327b5c24a6a14a0088915e267cd50c00 : SUCCESS in 7m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/17c3e4f6ed9b47daa332ef42485d5688 : SUCCESS in 48m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/334bf91614864a27b75e30c09b0489bf : SUCCESS in 28m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1910125d9aac4882998382175fdcbb1e : SUCCESS in 13m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b1a0e9b9a4be47f08ba329df9592a2fe : SUCCESS in 17m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3227f58c202c40c99a93b073393b2d7b : SUCCESS in 7m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2d2131e6ea7942658efa7ede88ce3166 : SUCCESS in 7m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cb360f7ccea64653868dec6ecdd9d2ec : SUCCESS in 24m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42f0edc5560c4512b5848a49455b8a68 : SUCCESS in 38m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c779109580ff425ba8acc51028e99108 : SUCCESS in 1h 14m 40s","accounts_in_message":[],"_revision_number":2},{"id":"2aa291c074d674ea1efc26fc08202fa265adba03","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-29 20:47:47.000000000","message":"Patch Set 2: Code-Review+1\n\nits blocking all focal-migration work - https://zuul.opendev.org/t/openstack/build/c2a24f4c137145709b40f13f950b91dc/log/controller/logs/screen-q-svc.txt","accounts_in_message":[],"_revision_number":2},{"id":"3bd86e654a876ed9ed4092b2cfe912887ed55178","author":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"date":"2020-06-29 21:11:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"702c12db0e03a31efd7df5c19a4aed3bb69a58d4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-29 21:22:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4416e993b2fcea7989aad699ab8578e6737e317d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 21:23:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b72b825214913e7ea66e218ed1405294755731a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-29 21:55:29.000000000","message":"Abandoned\n\nTransient failure in one of the gate jobs. Short circuiting the failure to get this rechecked since it is blocking some folks.","accounts_in_message":[],"_revision_number":2},{"id":"b9855e28f689d0fcbbb641b39a1ed991a71d4fd2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-29 21:55:38.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"d9ad4afad3bebc1903b956e7e014ae9e9f57277b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 22:33:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/08235057a33f419c95d754f29e09a457 : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":2},{"id":"483a8322eeee296e1a256c6e3542a97b6870c64a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 22:49:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e5f4f722b6b4f9f9185b4361c046903 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f83314b5ef5c40e9ab1d16c537c04e3e : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c3430070b914490b8260fd88023fc51 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7179d1378db4c21afe1c5f6cacf3775 : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/48f77965c6ef4b009915e34b2db22095 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9fdf1fb0cdc54134b7f5994606a1526b : SUCCESS in 4m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e6712d18b1ad4208a508bbffc249336c : SUCCESS in 9m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1520afabdcbe4048868c59cd4be30891 : SUCCESS in 10m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/26e4ca85fc4945429728b7b408922e92 : SUCCESS in 8m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/25d588c798c54874943182cab8caed83 : SUCCESS in 5m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0651fefa693a45bdb5522aba04a4b4c0 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fd6e25dd605941ddb69f9a45024ce405 : SUCCESS in 6m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/159f1933b9a04814b93542d9e7ab654a : SUCCESS in 7m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/affe729feb3d426baead8a64d5dd1f08 : SUCCESS in 14m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b674fddbcfcb4ef58fd40eca7457dabb : SUCCESS in 27m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8247f3c9ddc14a62a32b6393b05566b6 : SUCCESS in 13m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9f31b64c282f43a586f99545bc3904b3 : SUCCESS in 21m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/06a84e11d54143a4b1489069d92f26ce : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb87c6a420f24cbab6d8961506ac1882 : SUCCESS in 7m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f05a9411e0c84584b3c9659a97374414 : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37cc1cac21c14960826441daea776bd5 : SUCCESS in 38m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5681dcfbe7549f19c3a184d6cf28044 : SUCCESS in 52m 44s","accounts_in_message":[],"_revision_number":2},{"id":"9089786645927aa5252c352f12610af8bfb349eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 22:49:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dda360a95eee8e1ce88f3785c7bbe8e4ebe0a741","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 00:06:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0d089c127b6210c4e4ed2f98deece0bbfffbbb35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 00:06:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2088e0aaadf4917a28d6e78cdee4230 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac7766f8f0d149408832b254e7652092 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe9be1c38ff54fa4866f5f380c157483 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b4a942c3e024a36b6d910c3fb69a142 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f262d13d9d9f44ae95070c425d5ba73b : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/42cefe4fb69b432cbbc477b22a01701a : SUCCESS in 3m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3a0cc9e249b041d3a290b96cba40d49f : SUCCESS in 9m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/37245c9736f84230b9368ac23e0645e5 : SUCCESS in 13m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3746b611d3024757a3cbd8bd4d44a1e9 : SUCCESS in 11m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8a6d2c777e234315b8776c5d7a133ad7 : SUCCESS in 4m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0c5d9aa2840649b08d2da4fa08ef69b1 : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/75b06f1860514e9287b8afb14d881e0d : SUCCESS in 9m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/56cffc6dc1b84159a8a128dd45a57a7c : SUCCESS in 7m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d2f43dda5ab94d9780f84700150a3827 : SUCCESS in 12m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/75fbb3e503e14dadbdaff2a665bdac87 : SUCCESS in 31m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e41d64e2efc34163a822856970f2f24a : SUCCESS in 17m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/35079094a7a14490b3e38124899b23fe : SUCCESS in 23m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/474cf0c7d5f64d91b083513198f48160 : SUCCESS in 6m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32761737cef9489e8ceaf27b8e8984d7 : SUCCESS in 8m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/624f0b59da0245dca77bc76b25324b10 : SUCCESS in 10m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3cd9c625d1294caba8715d6483525e0a : SUCCESS in 10m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/371d0f2a1d934f159331ce1fe6dda199 : SUCCESS in 54m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5521426ca31e4fedabab387ae95941c7 : SUCCESS in 1h 14m 24s","accounts_in_message":[],"_revision_number":2},{"id":"a349e16201ff5a02e8f665ecea34825b560ac4e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 00:08:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cff7a27dcd044435815fbae8af4046c7 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"680f96b6359451287e124ff5e40f26badfe4d534","revisions":{"1e22dd23b370a07efcd56893baba3808e1d4e0c5":{"kind":"REWORK","_number":1,"created":"2020-06-29 16:35:52.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/76/738476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/738476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/738476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/738476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/738476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/738476/1"}}},"commit":{"parents":[{"commit":"ac55c89134c823f87fc66ca231772980667862ef","subject":"manual upper-constraints run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac55c89134c823f87fc66ca231772980667862ef"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-29 16:34:30.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-29 16:34:30.000000000","tz":-300},"subject":"Lower upper-constraints for requests","message":"Lower upper-constraints for requests\n\nRequests 2.24.0 has issues with eventlet related to monkey patching [0].\nUntil those issues are resolved, let\u0027s lower our constraint to versions\nwithout this problem to help unbreak several projects.\n\n[0] https://github.com/eventlet/eventlet/issues/526\n\nChange-Id: Ic475dc692971da7ad0c2dc4decb3a580620fdf98\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e22dd23b370a07efcd56893baba3808e1d4e0c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e22dd23b370a07efcd56893baba3808e1d4e0c5"}]},"branch":"refs/heads/master"},"680f96b6359451287e124ff5e40f26badfe4d534":{"kind":"REWORK","_number":2,"created":"2020-06-29 18:23:06.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/76/738476/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/738476/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/738476/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/738476/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/738476/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/738476/2"}}},"commit":{"parents":[{"commit":"ac55c89134c823f87fc66ca231772980667862ef","subject":"manual upper-constraints run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac55c89134c823f87fc66ca231772980667862ef"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-29 16:34:30.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-29 18:22:53.000000000","tz":-300},"subject":"Lower upper-constraints for requests","message":"Lower upper-constraints for requests\n\nRequests 2.24.0 has issues with eventlet related to monkey patching [0].\nUntil those issues are resolved, let\u0027s lower our constraint to versions\nwithout this problem to help unbreak several projects.\n\n[0] https://github.com/eventlet/eventlet/issues/526\n\nChange-Id: Ic475dc692971da7ad0c2dc4decb3a580620fdf98\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/680f96b6359451287e124ff5e40f26badfe4d534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/680f96b6359451287e124ff5e40f26badfe4d534"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
