)]}'
{"id":"openstack%2Frequirements~738518","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Iac5e49a58a46d9c33632ae0ca4096fc102fe7123","project":"openstack/requirements","branch":"stable/train","topic":"ensure-virtualenv","hashtags":[],"change_id":"Iac5e49a58a46d9c33632ae0ca4096fc102fe7123","subject":"Ensure virtualenv is installed locally","status":"MERGED","created":"2020-06-29 21:44:55.000000000","updated":"2020-06-30 21:25:46.000000000","submitted":"2020-06-30 21:25:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738518-1593552346465-c88092e0","meta_rev_id":"879f703c29a53c49b82325485242135cf204dbb2","_number":738518,"virtual_id_number":738518,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-06-30 01:28:14.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-30 21:25:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-30 11:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-30 16:09:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-30 16:09:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 22:42:05.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2020-06-30 11:32:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-30 16:09:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-30 21:25:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"097ba743bb3e0a432010c680feb867042fbf29e3","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-29 21:44:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"170b458397009d730cdc17128c33d01b8e6a06c0","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2020-06-29 22:42:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42a4bcdf83fe94ae076d4b889924c388750ce94f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 23:09:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb6ef0ef39d4f479edf49b82fd5e6e7 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4dfe9329be084c22a29dc3e6872b4ce0 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c440591491f848db8115b07f18aa7a7e : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b26e45e85d044789be97182e9bd646dd : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04469083470e4c92a66906785237b2e4 : SUCCESS in 4m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbd01fc0f75942e09011fc5813ba8cb7 : SUCCESS in 46m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9c128eeaa0044e5980ef65612e22e36b : SUCCESS in 1h 23m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/74c876296b474f3a88ce16c65f141fe9 : SUCCESS in 54m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d29e6e9a90a28adf94e26a8f5e98d0b2101f72ea","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-30 00:07:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"82ddc8c2c085e7d91ef317c2eb0197c7a997cbc5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-30 01:27:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5272333e3a462022c774d4c7b544f897d5dd7eb2","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-30 01:27:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"548559773cf6467321601b165799b8a40e9b3fa4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-30 01:28:14.000000000","message":"Patch Set 2:\n\nThere was some question in #openstack-requirements as to whether this needs to go into stable/train and then be backported to stable/stein, or whether it should skip train and go directly to stein.","accounts_in_message":[],"_revision_number":2},{"id":"06088a4031967bd7ba28fa1acca1c19582b61474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 02:54:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f37616fb3b0475ea46d106429bbda92 : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eadaf36122d644c5996e00010ecef4b2 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a709ea0032e4fa68f8a6c8841f93ed5 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47c9feab623a4b3b9f3456a2fbd9a667 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0468e1e4f09b41f5be1c14f96e52c7ea : SUCCESS in 3m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bdd34465229246c996d2d91db8f145fb : SUCCESS in 1h 11m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/72b56b9556484b5a842c7e069c5c9e3c : SUCCESS in 1h 22m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4ba22eb459184ddfb360a882ff9ed061 : SUCCESS in 1h 00m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1080ff96bfdbad8bb72a24d0c9559b7332dd1f87","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-30 11:32:48.000000000","message":"Patch Set 2: Code-Review+2\n\n\u003e There was some question in #openstack-requirements as to whether\n \u003e this needs to go into stable/train and then be backported to\n \u003e stable/stein, or whether it should skip train and go directly to\n \u003e stein.\n\nProbably doesn\u0027t hurt to have it in train, but looks like the main issue in question is not present there.\n\nSince this isn\u0027t a deliverable code change, we can backport to the other open stable branches without waiting for this one to merge.","accounts_in_message":[],"_revision_number":2},{"id":"96316cea52eaa89cc0911f1a5d4724649995603f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-30 15:03:29.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 05be2c1305f621acf8985aee7caeb0461fa35167","accounts_in_message":[],"_revision_number":2},{"id":"c721e26265fdb8f95523d0d9584d7cce873aa54b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-30 16:09:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f1d9f3020669006bc6198f7d31db4bb9942bed43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 16:10:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f6df4b9feb59385b22fd19d22200515afb87674","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 21:25:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a61e263e7b654f3c8d23c17b2d555d82 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b566eee439f54816afff18160300b566 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8afaf0e217a497c8eb5215dddcfd987 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad00a48a27154f1688cc562546de86cd : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/12b2ca2c08ae427a81ce6b1f2744eed2 : SUCCESS in 4m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cfab7579f53d49dfbb092662a5cea351 : SUCCESS in 1h 18m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/17a77c3113004418932799e91bfac6cd : SUCCESS in 1h 30m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/be82b8816d194cb1893a7659c20272bd : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":2},{"id":"387e497bc10c6778b967b37ff648be44863ebe25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 21:25:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"051549d82e354692de707d16a4bee361d69066c7","revisions":{"042e8a2b831177957b74624fc711f41a2899d44f":{"kind":"REWORK","_number":1,"created":"2020-06-29 21:44:55.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/18/738518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/738518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/738518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/738518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/738518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/738518/1"}}},"commit":{"parents":[{"commit":"f0e5e7367b032a13048f50f8087cf5ae4d3286f1","subject":"Merge \"update constraint for python-neutronclient to new release 6.14.1\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0e5e7367b032a13048f50f8087cf5ae4d3286f1"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-29 21:43:56.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-29 21:43:56.000000000","tz":-240},"subject":"Ensure virtualenv is installed locally","message":"Ensure virtualenv is installed locally\n\nThe stable branches that still support py2.7 need to use\nvirtualenv, so make sure it is available.\n\nChange-Id: Iac5e49a58a46d9c33632ae0ca4096fc102fe7123\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/042e8a2b831177957b74624fc711f41a2899d44f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/042e8a2b831177957b74624fc711f41a2899d44f"}]},"branch":"refs/heads/stable/train"},"051549d82e354692de707d16a4bee361d69066c7":{"kind":"REWORK","_number":2,"created":"2020-06-30 01:27:11.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/18/738518/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/738518/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/738518/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/738518/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/738518/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/738518/2"}}},"commit":{"parents":[{"commit":"f0e5e7367b032a13048f50f8087cf5ae4d3286f1","subject":"Merge \"update constraint for python-neutronclient to new release 6.14.1\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0e5e7367b032a13048f50f8087cf5ae4d3286f1"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-29 21:43:56.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-30 01:27:02.000000000","tz":-240},"subject":"Ensure virtualenv is installed locally","message":"Ensure virtualenv is installed locally\n\nThe stable branches that still support py2.7 need to use\nvirtualenv, so make sure it is available.\n\nChange-Id: Iac5e49a58a46d9c33632ae0ca4096fc102fe7123\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/051549d82e354692de707d16a4bee361d69066c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/051549d82e354692de707d16a4bee361d69066c7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
